)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~917132","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/infra_distro","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-03 21:00:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","subject":"Ensure find is installed on MariaDB containers","status":"MERGED","created":"2024-04-26 11:50:40.000000000","updated":"2024-05-03 21:02:00.000000000","submitted":"2024-05-03 21:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917132-osa/infra_distro","meta_rev_id":"2714151c498568ea9a1922f86d7f92ee4d4269d0","_number":917132,"virtual_id_number":917132,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 21:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2024-05-03 18:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-02 20:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2024-05-03 18:46:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2024-05-02 16:14:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-26 13:24:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 20:29:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-03 18:46:14.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"f0bcd0972dd7243379e7ad4d6f33169b6e1ebde1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-26 11:50:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94beacf96019fd1c3a0c4d1a6bf8370eda2c66fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 13:24:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e059adb34ae416b992f0646a6634491\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d596e82147e458c9898f92fd1a1cc04 : SUCCESS in 5m 34s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d6bd30302504e22b7410190039acd74 : SUCCESS in 18m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c616ffc4696a430080c7f8ca2af3b6a5 : SUCCESS in 1h 25m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2045f12c0b3b45819dc5e598cb91c266 : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d1d1b003a1f418a8efb172c6e269199 : FAILURE in 20m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/990867a46aba42319bba05558a9cfb9a : SUCCESS in 1h 00m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9f3eadc654804e00965d44f53bd647b1 : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c40cfec691c4fbe9575b88a63c9c562 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1afc369aaaf0480486e27a226624b834 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9dd10fcdd24f73ad7d6834e9ebc24b : SUCCESS in 1h 16m 14s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08e7f9f5f0a7448ab4dd64b90b3b9c88 : FAILURE in 19m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e58e5cb8d2d84868bbf8aaf272e53a6b : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":1},{"id":"eb4ed36327af6f00962563fee5da4f3f0905cdbe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-26 14:17:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f8d1d8aacc4db18f1c246163c0a0ce4d43fd861","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-26 14:24:07.000000000","message":"Topic set to osa/infra_distro","accounts_in_message":[],"_revision_number":1},{"id":"20e369e6c1c3d9316289d51035b072ae45d484ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:34:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af873db92dba4665ba0d12244eea193c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1b4bf24d454a4b906936f7d2d9c996 : SUCCESS in 4m 35s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/284701518c0344e29a8f8a6f8cba71be : SUCCESS in 13m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/118d31f0984e42979ee6e23dc1f025ef : SUCCESS in 1h 19m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1478ecfc3fd24472b3c191ef0f8707d9 : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/653cd94c6eab4c999ad862346222696f : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d62287da52534a76be54233c769a4832 : SUCCESS in 1h 08m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee297fc20e8f4563b1253027eff47051 : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f430dd0e4e842bc91e216982662ae2c : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/831b610ec82940a587b7577acc67db54 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f281e44506c04d449a117c76a02bed5f : SUCCESS in 1h 17m 28s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d8900b8b55748518bf77098e342ac83 : SUCCESS in 2h 06m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1e3381f687d443f9f676de529c04477 : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"388ab49b99c0f4c5dc2f06f3282851ec2578c99a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 16:14:26.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"562286776c0d3cea829e171ab12fcb76a8615b1c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-02 20:29:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1ac2150318b2f0dab45f22e180d5a11507c5256","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2024-05-03 18:46:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec4ac43e6daac2a274fbf8b856f93ecab415f3f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17b75a1b891185802b944598c88eb83e19c7d24f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:00:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa359b5d26da4423881ed25b6baf4838\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eced8f1186fb4ae1905ce623207175af : SUCCESS in 4m 13s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0db8ba3a9674da3a34414e9629f40b3 : SUCCESS in 13m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e8748959ef74d2e8343e333d5d1734d : SUCCESS in 1h 38m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d1a77876aed4f3c96499fdd2459f00c : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d2c600011a2e4e948a45b5ba9d1bb87d : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b286c928c7e74961959ca81d0ad444db : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c53e377fba43f79c78f7b60afd006b : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e79c64b35494d39918166f4d4d8f422 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/379214d944c14a30aa7186fd5b089c30 : SUCCESS in 1h 07m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b31fb85f01a740129be0708502984c04 : SUCCESS in 2h 03m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea4e784ea02e40e7b2c2b059f02c1679 : SUCCESS in 1h 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fbd5d720634b262e77b0ef385fd0979e2d5c9f75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:00:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2714151c498568ea9a1922f86d7f92ee4d4269d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 21:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85a3c9ad8d4465aa94b197fe8a343cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b3d870d58348a39a8a32117fb2856d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64b54f24583bdbadebf532fe5fcc9eadee7e7a33","revisions":{"64b54f24583bdbadebf532fe5fcc9eadee7e7a33":{"kind":"REWORK","_number":1,"created":"2024-04-26 11:50:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/917132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/32/917132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/917132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/917132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/917132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/917132/1"}}},"commit":{"parents":[{"commit":"046cd1364e26a3bd2cbb1062322d195872ee4296","subject":"Implement installation method selection for MariaDB role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/046cd1364e26a3bd2cbb1062322d195872ee4296"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-26 11:50:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-26 11:50:22.000000000","tz":120},"subject":"Ensure find is installed on MariaDB containers","message":"Ensure find is installed on MariaDB containers\n\nwsrep_sst_mariabackup script that syncs data for members joining cluster\nleverages `find` util heavily. While Ubuntu LXC image seems to contain\nit out of the box, EL does miss it. While we do not see any issues due\nto missing `find` on modern MariaDB, script still apparently relies on\nit\u0027s existance, though can handle cases when it\u0027s not present.\nStill let\u0027s ensure it\u0027s present, as it feels playing important role.\n\nChange-Id: Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/64b54f24583bdbadebf532fe5fcc9eadee7e7a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/64b54f24583bdbadebf532fe5fcc9eadee7e7a33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
