)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~918075","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2023.2~Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","project":"openstack/openstack-ansible-galera_server","branch":"stable/2023.2","topic":"osa/infra_distro-stable/2023.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-08 11:41:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","subject":"Ensure find is installed on MariaDB containers","status":"MERGED","created":"2024-05-04 07:08:18.000000000","updated":"2024-05-08 11:42:18.000000000","submitted":"2024-05-08 11:41:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918075-osa/infra_distro-stable/2023.2","cherry_pick_of_change":917132,"cherry_pick_of_patch_set":1,"meta_rev_id":"972a27a61e0b27ffe167e0c924fd6ea962d5f9e2","_number":918075,"virtual_id_number":918075,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-08 11:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-05-06 15:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-06 18:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-06 18:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-04 09:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 18:43:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"24a6e18b3595bb91372cbae26036fc0a5b39fea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-04 07:08:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d82b472da7e544055dccb7bf68a096d28763974a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 09:27:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfd6240616014e72b4bc90f013423312\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b92d1490c340708a163b0e385febe0 : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24cb428cfa7c42658af33c4ce31b22d9 : SUCCESS in 13m 05s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/89b3d75c3d2e44619ae55dbd46587a2f : SUCCESS in 1h 22m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/066483d9a300490a9c694c647d285edc : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a65cd33925245e1a838bbbb0c18a276 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/016e8185ec384508b6ce390ef42a30b8 : SUCCESS in 1h 02m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ab2bbe550194572891e0c8ad62fa70c : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daa6dc5ee18448d38f0909ec5dfc921c : SUCCESS in 1h 21m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b432b9234b6e4f688d0b08cbb3584cde : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85e2fe122ce54864810648b23c35a4f9 : SUCCESS in 1h 22m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4796f21993a348b4b489bb96cfdd3b11 : SUCCESS in 2h 08m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91729c5800f5467a8ed0d8bad1948939 : SUCCESS in 1h 51m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0dbcfc9b0370a0d15ba8a77100743635c56b0db1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-06 15:38:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"847e6b276293b4e2bccd5c9da031c05b0c0ec17d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-06 18:43:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"921ca1c8cd4c1cd9badb8bae3f0e77a8be0a252e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 18:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0de7e96e2b209737bc7886fd852ee01888103a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 21:04:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58444b013f1a4f5cbdd5c9ca2f85b2d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02a529a199940b5b5fcf458b8ca31c5 : SUCCESS in 4m 14s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47de224efe744f95a619ec757e81d4af : SUCCESS in 11m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1b3f35e469f4c55bf355f3a3dd53a80 : SUCCESS in 1h 33m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0501d56b7fe447ebd307f392b8dfec7 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/525cf022e8c04920a7c0d87d83c3655b : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d829ed27c67a49e989f93caf671947fc : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d49efcc5398f4c1a869f7f5fd6b5612a : SUCCESS in 1h 16m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35a1cb2218cc46ed8f98dddfcfdff448 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62e7ae1da7cd4015ae55da822f66e796 : SUCCESS in 1h 18m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52c367678c85496cb0bb72e15eee8b9c : SUCCESS in 2h 00m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f554e8f5d6b947929afd2da4e9d700d8 : FAILURE in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9a2b51d44cd68f59721cf9b30671450df2f09d57","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 07:01:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"926b3f4c71c5a5853fc8bc6dbbf3c7c100debef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:22:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ddafc83d77416a8892805850f41c27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533ac687d8f04825bb92dfb2c63b2e46 : SUCCESS in 5m 42s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36a970aa2438480a9bfdb3d59069d18d : SUCCESS in 17m 41s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/68739d37d33c400489d071aa964c144e : SUCCESS in 1h 29m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c00bef29fc624838821c9743b054fe70 : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/edd0f61e413742afa60e0511ea1dbc4f : SUCCESS in 1h 37m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/11d240c34a47439d9727aca2c0f3ad47 : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c236cd450be749f680949b45775900c5 : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb861f6d625a477d9c50c56290becdbe : SUCCESS in 1h 00m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/318e2c9584264dc3b0e5fe40ec40b5ad : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc20c62ed4d74458be9fcc9ca6c5c668 : SUCCESS in 1h 15m 00s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ecb5892ac7d4ea4ad8a4066698bfb0f : SUCCESS in 2h 02m 54s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c86bd7e51274e059bc335d18ed131ce : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":1},{"id":"21c5b616b9be1a5b647e47e7d623393b5b53391d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:24:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42aaebf23a5d0a681b0b4e6b02fda731df983415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 11:41:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57fdbd830ae44b6dab6c94a6e818cdec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4015029d4e41b289ffe40d3332abe5 : SUCCESS in 5m 25s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/165cd86e4cd746e8b5798e4669599939 : SUCCESS in 17m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6eb03fd1127448e3ba05a9b59a9715db : SUCCESS in 1h 38m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ce102df45f0481e86fdc533a63fccb7 : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0771ba1e92d640a882f279db1ecd3911 : SUCCESS in 1h 15m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dde2775fdbc240b490df86abb9f951d4 : SUCCESS in 1h 20m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46df9cbf5b9a41129dd533c571c49b23 : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3cf3cbcd67540f48e39f0e38202e006 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee469cd73122450a9fe0e45abe5232fa : SUCCESS in 1h 20m 34s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccac9f03b8b34de49c79233ffa0c9625 : SUCCESS in 2h 09m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a15d9b1d804ae88b519d40cfd4965f : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9a8da3be8e828fdff6040eb97b0d23503f02ba99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 11:41:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"972a27a61e0b27ffe167e0c924fd6ea962d5f9e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 11:42:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8263e5681c624d97addf3164b07a13dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d76527e78847bd9bcae19a1d197f56 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3a1857fb57b6413fff53a143a9e52cb72ed3274","revisions":{"f3a1857fb57b6413fff53a143a9e52cb72ed3274":{"kind":"REWORK","_number":1,"created":"2024-05-04 07:08:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/918075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/75/918075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/75/918075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/75/918075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/75/918075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/75/918075/1"}}},"commit":{"parents":[{"commit":"a5ee577b6d7d46447e5a13321a6fc02f902edd2b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a5ee577b6d7d46447e5a13321a6fc02f902edd2b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-26 11:50:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-04 07:08:18.000000000","tz":0},"subject":"Ensure find is installed on MariaDB containers","message":"Ensure find is installed on MariaDB containers\n\nwsrep_sst_mariabackup script that syncs data for members joining cluster\nleverages `find` util heavily. While Ubuntu LXC image seems to contain\nit out of the box, EL does miss it. While we do not see any issues due\nto missing `find` on modern MariaDB, script still apparently relies on\nit\u0027s existance, though can handle cases when it\u0027s not present.\nStill let\u0027s ensure it\u0027s present, as it feels playing important role.\n\nChange-Id: Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567\n(cherry picked from commit 64b54f24583bdbadebf532fe5fcc9eadee7e7a33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f3a1857fb57b6413fff53a143a9e52cb72ed3274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f3a1857fb57b6413fff53a143a9e52cb72ed3274"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
