)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~918076","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2023.1~Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","project":"openstack/openstack-ansible-galera_server","branch":"stable/2023.1","topic":"osa/infra_distro-stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-10 08:26:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567","subject":"Ensure find is installed on MariaDB containers","status":"MERGED","created":"2024-05-04 07:08:38.000000000","updated":"2024-05-10 12:33:04.000000000","submitted":"2024-05-10 12:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918076-osa/infra_distro-stable/2023.1","cherry_pick_of_change":917132,"cherry_pick_of_patch_set":1,"meta_rev_id":"031b275a06668f8e28c57a196fa1003d0ccb8174","_number":918076,"virtual_id_number":918076,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-10 12:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-05-06 15:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-08 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-08 13:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-04 09:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-08 13:44:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"cbf787d0363336f81ab3f913651f9b617cf0c61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-04 07:08:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a7128768d3bee3be0af18654f6067f51ad13d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 09:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f789629ef5c44f5cb7eadb500e757a4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d29fd75620b4df68f789e148495bec1 : SUCCESS in 4m 13s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78dce52dfefb4c849d9636b59a05c7f9 : SUCCESS in 12m 51s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfb5bfa03122442dbccc7c576c366c3c : SUCCESS in 1h 21m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9764460a9ed749c48056a02df780ddb4 : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e4b06e7586c4f8bb1f87bfe89c36325 : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b26ea6a617284ea0a22bb60158b75c39 : SUCCESS in 1h 05m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae5b497730e40da99361d1194a61c9e : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502c691b49964c55b4fa6866b0478772 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06dd473f6b014fc3995846d693138b7a : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51f2b80dd2b2420ca9cab9dcd925e2cb : SUCCESS in 1h 16m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a251ec45aa3447e394bb85d25b394b17 : SUCCESS in 1h 56m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36ef7d5620fc4b09964a06e64627b484 : SUCCESS in 1h 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"daf93db218f6e408ae99fc3c8f789fef84179614","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-06 15:38:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40662ff28fccb95d82b23769d6300758b10c54b8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 13:44:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37f8a00c3e8ca97028721efb674f088e8d67addb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:46:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8991c9d0df1a8ef879f9a4d3af1823b342f0323f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 15:37:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba5d6131b5a4a1fb6f617a0a879394d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0d5b282882486581616319e390f5ec : SUCCESS in 4m 50s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38460f959f85414793de8d3b77cebc42 : SUCCESS in 11m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e36487bcbc241ca8e73e575ee593598 : SUCCESS in 1h 30m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b93d41313824c24964c7bec59b31f8c : SUCCESS in 1h 40m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0042b213fe9f4ed3a90469d2fa13a96c : SUCCESS in 1h 05m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/289e797e2c324474a487feca4f35cf0c : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867d2171e3f14581ad329794a79513bb : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8a50fd8175e4877ae37d438b176bbba : POST_FAILURE in 48m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b271f66fb984d72b1e050da2ca58b58 : SUCCESS in 1h 46m 32s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd46a72a84ee4295b60cc69e34c97f7e : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"468b9cb4da8670c8ac7a179f07dd228e0ab6d8bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 08:26:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5b3e39100c31432b4f567df6ed557e6cfc23d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 10:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab314a33d46e4adb8af087bb80399d8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014ebaa99ee2478eacd1629b0dd3be98 : SUCCESS in 5m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b4621f4b6244912ad9029bbe521071f : SUCCESS in 17m 26s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/070baae64ec341c295f8fd8d4697bc67 : SUCCESS in 1h 21m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6d09c3731b24245b78e058077bc980a : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b3b6889d6de4bef98e3220fd25cb4ae : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfb90a7a5080466fb974a9a68ecfc409 : SUCCESS in 1h 21m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66408c5f567047539fc5b8956c806b1b : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80c16a017ebb467ba90550a0257afc4d : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3128b1374f6403c89f00dcd2cd565fd : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c5aad29909443d4b5443a803473e2ea : SUCCESS in 1h 21m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54901ca595194d0f830893d200f796a6 : SUCCESS in 1h 46m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dd5849f32634284abaa1a54555c5e6a : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1fc534b81c138e63ba0a49dd6fe3ebeae41eb5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 10:52:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa8a5536e43b33a590cc410fcc116e0869c2f4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 12:31:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8438b3f0755437b8fdde303d4a57d45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578e3b4191b74c20bfe0bd566e2f2b58 : SUCCESS in 4m 15s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4ca541b3df74636aa36e466ea0c8845 : SUCCESS in 13m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3678b917b77448bbde28cec825d9e37 : SUCCESS in 1h 34m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05f89957bfe1427f9ce0229c6cdca319 : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e47375ae3ccb47368fd712699f4ec59e : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/101390ce7ee040f48da97b1c5f18d43e : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f773eed31ef04ee3869a2f34510f87d7 : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e6981256d34fcea1853e1740f03161 : SUCCESS in 1h 02m 36s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2894f9e320748a49a6938acff2db40f : SUCCESS in 1h 24m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76531de55d494838850421bc9f2e627a : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2e5116aafa1de82438c5a1eee4bfa71cd0c71c2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 12:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"031b275a06668f8e28c57a196fa1003d0ccb8174","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 12:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb34c9055f2a4427bc8e5b60850b80d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2520c0c90ba541d2b47f08835b0dcb55 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff3216bed99fdfe528eb94fc0ce22d1fbb424d06","revisions":{"ff3216bed99fdfe528eb94fc0ce22d1fbb424d06":{"kind":"REWORK","_number":1,"created":"2024-05-04 07:08:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/918076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/76/918076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/76/918076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/76/918076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/76/918076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/76/918076/1"}}},"commit":{"parents":[{"commit":"265722b67ab24344c5f7de5d5781fa80c3d8234a","subject":"Fix ignored database directories configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/265722b67ab24344c5f7de5d5781fa80c3d8234a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-26 11:50:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-04 07:08:38.000000000","tz":0},"subject":"Ensure find is installed on MariaDB containers","message":"Ensure find is installed on MariaDB containers\n\nwsrep_sst_mariabackup script that syncs data for members joining cluster\nleverages `find` util heavily. While Ubuntu LXC image seems to contain\nit out of the box, EL does miss it. While we do not see any issues due\nto missing `find` on modern MariaDB, script still apparently relies on\nit\u0027s existance, though can handle cases when it\u0027s not present.\nStill let\u0027s ensure it\u0027s present, as it feels playing important role.\n\nChange-Id: Ia10a9ab589275f18ff2c92a21cd2ff9aeac07567\n(cherry picked from commit 64b54f24583bdbadebf532fe5fcc9eadee7e7a33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff3216bed99fdfe528eb94fc0ce22d1fbb424d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff3216bed99fdfe528eb94fc0ce22d1fbb424d06"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
