)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~922553","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2024.1~Ib2cd632f437c4771452c22f77bedd3012f37b86b","project":"openstack/openstack-ansible-galera_server","branch":"stable/2024.1","topic":"fix-cluster-state-condition-stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-07-05 07:24:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib2cd632f437c4771452c22f77bedd3012f37b86b","subject":"Fix condition in cluster state checks","status":"MERGED","created":"2024-06-22 07:31:05.000000000","updated":"2024-07-05 11:28:46.000000000","submitted":"2024-07-05 11:27:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922553-fix-cluster-state-condition-stable/2024.1","cherry_pick_of_change":922375,"cherry_pick_of_patch_set":3,"meta_rev_id":"9397afc977ec26eb3201ef33b34179ff081fdf05","_number":922553,"virtual_id_number":922553,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 11:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-06-25 07:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-22 10:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-06-25 07:23:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-22 07:31:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"state":"CC"},{"updated":"2024-06-22 09:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-22 10:05:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-06-25 07:23:36.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"50b70734066f45d498179525e6cd81208d8576fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-22 07:31:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f55ad5581d9a0d445bade1bb334c7f7107421593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-22 09:29:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d3b5c04fbe44e9a753ca3383807e64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbab272e5c43459cba1129ede5ab44a7 : SUCCESS in 4m 01s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0655bbd8d76e447881fb8fcea2d4819d : SUCCESS in 11m 07s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7152dae3f31a428a8c8df1b73f81c64e : SUCCESS in 1h 30m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a4cc887a0a04d20a63306d6a25774c2 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3237973d6db84c809f13c7ceb19dbb1d : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/af3fe13490ac4fc59fe4f16ebb4f8ea9 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ca1376a74ad94c08a18e6729e8ab0c20 : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eebf5f0a040049c89f07d3b225d100e1 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfdc6681bd6b4acf9240e21d528883f3 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d83011c4313c46b7893a79c82ae71f5f : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47218aa42b0340f7a6355d6fda59445e : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dfdc7fb14432420faf2e1ed49aeb8ede : SUCCESS in 55m 48s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f2a60713c374b37b611384ed8f8a4a0 : SUCCESS in 1h 11m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/43173cf424a84523822ecb40dea27ad5 : SUCCESS in 1h 52m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/490066aae200462f954f78cce1e2eac2 : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6dc6f6d8949bb04309416b39fdb05d4a796bd42a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-22 10:05:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a294174e8a395216c0ba31780b2054a6be92dd8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-06-25 07:23:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a74f922b308e88d545043e492b06aadb92db6d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:25:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf61b8b6a2d28324e2f3ae33c891cdb4de988d9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 09:07:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25f4012a8f0d43ac97760e5d13a997e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c9ec639eee46cebd77362e9f9f3231 : SUCCESS in 4m 10s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7145253b476b43568cc43b4c7273e915 : SUCCESS in 11m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/861e0a931f5f439587642d0bb748246d : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/883deba6bfc14ffab157a44a5fc7c302 : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e459f2b84fef4e2ead73af1dcf00f40b : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c873dcc677e3476e8c84ef099f6ec0b5 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dba5f653238448babc837ff05c1eeae2 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/002da1f6c57f46edbd8cce44438b0caf : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5023490e1ded4f3b9e756e5bcef7f6ca : SUCCESS in 1h 11m 22s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88dafc0635fa4da48278cf40ad5a7ddc : SUCCESS in 1h 11m 59s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/62212063fb924b31ae06bd36d68c48a1 : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/060fe29f7f79461092378d02138523bf : SUCCESS in 1h 03m 03s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd424d8e6e3747958e86827640d309d8 : FAILURE in 32m 16s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/422e4595c464445ba3934202646a6205 : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fbdf903cee203b13ffff2e4cb61ce7844fe2f171","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-26 13:31:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec4486a82a4393ce5f9629191b84581442849321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 15:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9179bbf9288143848e8cd4c3cbccdaa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c684467cac047fba26d3c2004e9449a : SUCCESS in 5m 34s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f721f9c1fc841ffa7b905d88257db1c : SUCCESS in 10m 28s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac6136842d66460199a3f404d900be51 : SUCCESS in 1h 27m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/14ea974dbcee46928c34d0fe5f61adc6 : SUCCESS in 1h 29m 37s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8bcf75f6ef2e472dbe266a873774c515 : SUCCESS in 1h 36m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5a571f4e71f84a1ca5bf4279ab7f0544 : SUCCESS in 1h 06m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f02b0d4064547ce9d3e0dbac593a222 : SUCCESS in 1h 24m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bd5c242fe0440c3b6448ecd6eef2faa : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/760b35785cb3407abb022e803c995219 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d93294fea934ebf907033adac57a136 : FAILURE in 1h 15m 06s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/530af1a402e24b689794cbb18beb3ef2 : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2997ac48b3ea4569845ad57b703c3f1d : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c5363ae6b44f959c3cdfce80274c5f : SUCCESS in 1h 06m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7938cb6e9183430e8e74f8f7de4ae04a : SUCCESS in 1h 52m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/920d823ab78642a982bf2056d5372673 : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b7c416702bd661f74ac9a80721ab2aaa1530dc36","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-26 19:36:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ed816181826f092f23373b7883c895cfa6ddf83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 21:38:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e670eb3a40a4b718bf4fb0f0d011748\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76d4c4a362a4a129b607cea12758d10 : SUCCESS in 4m 07s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bfd77c6e3554488b612b37853f09ee4 : SUCCESS in 13m 43s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe0ae60196cf4fb19b695df788cb500a : SUCCESS in 1h 24m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a76fa1b34b9d4bb990489d26fb0f8676 : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2ea4ae4d7a646bbbc18b490373e06e1 : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/555fafcfa3634b3999181ac2319ee5ac : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/35968af6521f46df9195f76339815ecc : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35dd911215f4474daae19c347bae037b : SUCCESS in 1h 09m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2335d50c3ce64f3e8ececdb6e209e3b1 : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/140c09b7c3734eb3b7fa82be2918f569 : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c35dae2c18e45a48385ae88fe57df32 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9ac9c88ac7264a4abf50719cc3fd5e0e : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f34ca9e265e4ede9b6cbd2dec5eb17e : SUCCESS in 1h 04m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0f52c9eec9b248e3ae98faf59c6f20b2 : SUCCESS in 1h 54m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1bfea6d18794b19b09b1b1600329f1b : FAILURE in 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"12b4a89e6f79f4783aeebc0903540d87fd447d05","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-27 14:17:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c73da544f8a3715d0cbdc14c617bcc5dc5d0f166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 17:23:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b21fc9984384e1abd7cc9ed1f17db90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c131ed25f02433f84f0a50652a703c2 : SUCCESS in 4m 46s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79e1a54f6f1844f6a529fabc0360948c : SUCCESS in 12m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/15d5f7c8c49d439fa5a102f3f5022d5f : SUCCESS in 1h 27m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c160d7c52d89469b90c48561270ec2f4 : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/16864fa441824d40913787016537a746 : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cca82db682d14ed4b1b74743634149d5 : FAILURE in 1h 13m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fc5b03c16bd8436da43a441a3fb834fe : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92f45d8c8d6341329a28066a87cbf8be : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6488f674e2a4ff997662c084c2f791a : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9509b55cbaf64ee1b3adf8478cb5a9ce : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67d1a3b1ad9f449db5bf9450502ee150 : SUCCESS in 1h 17m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ae3bcdd6cb34488f9c92b81aec58300a : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bb49760cb0d4cb4bba8184b874c7423 : SUCCESS in 59m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fafddcc3806c47f3ba587f32252639a6 : SUCCESS in 2h 14m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94bb9f3231b64265abcd570650124ff3 : SUCCESS in 1h 42m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f0022fe636b04bf4e03639d762c754c53d535dda","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-02 15:20:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"781fb46e8c7d9b2d3cc8d4f2fea2ed67217eeba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:56:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42ba7bee3737418f9c63a517d57a9ac0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53422525ed74eb78e3431ebd51c6548 : SUCCESS in 4m 22s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2645d080b0d345488d96e0d12f656eaf : SUCCESS in 12m 49s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d58d55b1420d44838f807f6d6f7df677 : SUCCESS in 1h 19m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73ff5803153e4288801956259531f260 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20835a58d4d6447a9996e27784bfe526 : SUCCESS in 1h 33m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6ec280e019404a5083f8750bedd0a5bc : SUCCESS in 1h 15m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f3ae78a7b2ab4cbbb89493b2707aae3e : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72422e48bca94bca8efb36824a8b7550 : FAILURE in 1h 18m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efbe645789474ba1a14119af3e860f2c : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47034d64a78946b2be679d7d218cbb8f : SUCCESS in 1h 37m 49s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/035aeee7977442aeb0f789da5e1e9e5d : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d6a7e3b3f734cd19b784a2563924c86 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a05768cef6b43cca3e377adf48ed0f2 : SUCCESS in 1h 16m 58s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8a37527101a4385be22456e21598078 : SUCCESS in 2h 03m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2533463cbba345d090a088f6029421fe : SUCCESS in 1h 49m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f01090f9b590e22bb74c3cf4c4dee64feb45f8e5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-03 07:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4e4fbf04a63df8ae4f69815a0ef1dafde845cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:28:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f282d371665947669879fa1b7b586224\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0da21bca17e4d9baacc31d0a0d3b030 : SUCCESS in 6m 16s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bfc5c7bba3b4604ba780bf7cbc6b53b : SUCCESS in 11m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ea52408e9d64dd5afd6dd548931398c : SUCCESS in 1h 37m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ab53830b99a44b8bb9661fc0bf3d555 : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d08ab5b99dd7403a9310751a532aa15b : FAILURE in 35m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/389a35897e4d4cfd8b6dbacf2127ea9a : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/31030a29be1148129fa7aedb784aa2cd : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a5dbc289c414de2b75896b6a1373315 : SUCCESS in 1h 08m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb149a0e25d64a0cac839dacfabb0550 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be3fc2351a84414394e350cf2421ff96 : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9d5ec78bc8b412fb543c961d357e861 : SUCCESS in 1h 01m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a1b9d4f5a5c54b67873c951a980aaea1 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f58f60bd2cf84cb98104be7315349032 : SUCCESS in 1h 06m 53s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ac6634e02ef4a799d43318cb7711974 : SUCCESS in 2h 08m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd910a9ce9c4e7d8d889064e687a881 : SUCCESS in 1h 54m 22s","accounts_in_message":[],"_revision_number":1},{"id":"754cb3315659cf4f8426a38105e3bfdb0a5adaf6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-05 07:24:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8da81e70683149bbe99127e5e396f60121c731a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 09:26:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3202c1d6737b4048b6457d4818497363\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1d4b6c5b76431a996468dc7370c6b6 : SUCCESS in 3m 49s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a83a110cd846429d932457b328baa82f : SUCCESS in 10m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec5e8e36c33a4508bcfa064cdafd8ea7 : SUCCESS in 1h 18m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/796aac164a484184bdccaab78a0cf412 : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/265836e211ab409e92b684982e22464e : SUCCESS in 1h 32m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e511d93edba4371be0274daf3143926 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dec70aec902b4c5e913f4a0d2735fad6 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c58e75cd238148508ba831d7c48b2d42 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ac31526b5b1440ba8b132e1ea2e170a : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c2f91f0b5a34d99ba2e3c5c447ac055 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/732e9479bebd4f5a8cc3d8228be3c438 : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a768bd8180f443580d81f115a06621c : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/186d5e1361be4dd3a68d0100aa647b01 : SUCCESS in 1h 08m 43s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2e6dc2037a6418484baf4be147a4801 : SUCCESS in 1h 55m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9279d9b91b54831bd026b17815b9771 : SUCCESS in 1h 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c0a936ea3e5cf2ba9c79f3db4cb9c3b7434aa418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 09:28:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d54e6faee5e09efc16c1e47cae02f42fbe35543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 11:27:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa2ea2090217466fb95a6a1a851c9767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78fbe925beb94da9a09b328e00fef20b : SUCCESS in 4m 03s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28279e86fa2e48b9b14fe4b54419c8a3 : SUCCESS in 10m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4a3752d8b78e44c494a8b6a55b07fd87 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/94da605ffbb948da9a9b040c0e486488 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e7bb61e61eb34c70bd0535435e8a87c1 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/eb94f0c2420542b8a7a8a9beab1ed9a2 : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dbdf056ed0a45a593c5644dcf2013d3 : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f67596fe273b4213820c774d1525bd0a : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c82a0137e874d6998d532369872ba68 : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/566e23534ab4434f857bd386235c134e : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f65f3b7cb6f4e09896fcc3056b3a7ba : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b76461d3159241fd8031ad8ccfea21b1 : SUCCESS in 1h 06m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3974a12531724005ac4c7ccc54c90b0b : SUCCESS in 1h 58m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33af6c423be04025976afba0684822ec : SUCCESS in 1h 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7e57002e2018872393fb8e598ab212c20d4a79fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 11:27:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9397afc977ec26eb3201ef33b34179ff081fdf05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 11:28:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8658f2fb869b498e878ac15a5f6ce803\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c81bb1fc504f959c8f89d813ed9f98 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"360fd75e037e2d91b2ee58555f5fed456908f0ed","revisions":{"360fd75e037e2d91b2ee58555f5fed456908f0ed":{"kind":"REWORK","_number":1,"created":"2024-06-22 07:31:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/922553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/53/922553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/53/922553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/53/922553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/53/922553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/53/922553/1"}}},"commit":{"parents":[{"commit":"7c248a441cbd21744f94ff0a88a41bba1412cde0","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c248a441cbd21744f94ff0a88a41bba1412cde0"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2024-06-20 09:16:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-06-22 07:31:05.000000000","tz":0},"subject":"Fix condition in cluster state checks","message":"Fix condition in cluster state checks\n\nConditions in Ansible \"when\" clauses are combined with \"and\". But the\ncluster check should fail if only one of these conditions is true.\n\nAdditionally the second condition on the node status check was inversed.\nIt checked for a healthy node.\n\nChange-Id: Ib2cd632f437c4771452c22f77bedd3012f37b86b\n(cherry picked from commit da6abc6d0af2d8f082652d4040894f2ef8958250)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/360fd75e037e2d91b2ee58555f5fed456908f0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/360fd75e037e2d91b2ee58555f5fed456908f0ed"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
