)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~922819","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I77cc32f49864f7caaad52150cc6684a488caa128","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-07-03 11:58:18.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-07-03 07:19:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I77cc32f49864f7caaad52150cc6684a488caa128","subject":"Remove xinetd clean-up tasks","status":"MERGED","created":"2024-06-26 13:30:55.000000000","updated":"2024-07-03 11:59:18.000000000","submitted":"2024-07-03 11:58:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922819","meta_rev_id":"8365774a58b3cd474a6c259a3d13c987eeefc11a","_number":922819,"virtual_id_number":922819,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 11:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-07-02 15:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-02 15:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-07-02 15:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 15:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-26 19:35:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-06-27 10:03:08.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"651e12bb2e04faaacbab13d773729bf14a2af932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-26 13:30:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73bc1c37ddaca363d3c7d8189026dc4e8d75136b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 15:38:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c4d8e6fc49449fa9fc315b1efa63f89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d1c135392f4d40a539aef00b1542fc : SUCCESS in 5m 16s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2d8dfb178ec44379cece356148f28c3 : SUCCESS in 11m 57s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b65c193a347049c3933ae2b7b6a45399 : SUCCESS in 1h 25m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3efcf170b06f4ab6ad881e090c33753f : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2c62e8cbd6648c29c57464720f760d6 : SUCCESS in 1h 28m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c61b83a4ceb49768305dffe9900f765 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5916ec8035b74c74835ed815cb8b53aa : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47f34d176ed346dc915c95ade53b5f39 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ee8ee80f9f40129215dc2db8056525 : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17e86cd3bcd74cf4a751b11f45531fba : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d33333a07d7945a6bf54e57a11a8515d : SUCCESS in 1h 00m 30s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b9ef114405c4048ad5ab2127075f403 : SUCCESS in 1h 05m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e16ee65ffc7c41c581b29285c76d5ab5 : SUCCESS in 1h 46m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1da6e0d32ee4bd8bbf4d375157d3f17 : SUCCESS in 1h 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7cdc1b7f0f3f7d1dab54b9171ffe9d716220b3e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-26 19:35:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4563075d8ad293422ab706dc8556538aacc13585","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-06-27 10:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53fc1e318d6c001fdff14421a863f76b76e9fdfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 10:05:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f67bd888c5822e61ba82317d1e20bb6053436b1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 12:13:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d418bfa5a644499994dfdaf4ecfe2adc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c3ee64d1734d878928985f43b7f02f : SUCCESS in 3m 51s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06fa9f253be34944bf310819073f21bf : SUCCESS in 9m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a21521d73c5d4a67b785d36c73d5b8db : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11bada5c59074f4aae29e403e724fca3 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/41cb5d861e704a5990c13591bebae9c6 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd5f3d4e09294560b1a8a2030c8a364e : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5329742ebd14c5c882253eca5ef5860 : FAILURE in 38m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4dab2f7c44c447cbf54ed6203633352 : SUCCESS in 1h 07m 16s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8307c7ea846443438768933b7150fcaa : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/34b99dd84b164afba4e77615af37f1de : SUCCESS in 1h 11m 17s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce8cfdfce3d24c8687e2d80f9f2bcfc9 : SUCCESS in 1h 05m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ba36fcbb0074f59a1d6e7a2fda06fea : SUCCESS in 2h 02m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dea3a6658cfa45ebb34089344ed53e68 : SUCCESS in 1h 40m 04s","accounts_in_message":[],"_revision_number":1},{"id":"07b5e93aa6ea101224759c3ff7f28b59d7e75fdd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-27 14:19:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c78e4e86256ce6a39f199e33f6d18deee7a098ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b6a85023c1472d83bbbc7a5b94d28e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05d44bd808e4b17be3d6d265a6818aa : SUCCESS in 4m 03s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0455e6649074437a5f93e2f054b1dc6 : SUCCESS in 12m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9b0e86fda59457b81f9e9ad67102593 : SUCCESS in 1h 23m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac4d91200e884a6a8c765b0c843254e4 : SUCCESS in 1h 28m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9348769b24b642658c0144686a114194 : SUCCESS in 1h 34m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6181ddcca5c8417d9d69b3aca2943e53 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65a26b4c094d4cb7a166092fd97d2152 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ada7dd1b8a1f4e08b6fc37fd77debd9e : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96bc6f130edc4f3393dc54b3153ee7f9 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c354977612e42dbb5d320c558bfc2c7 : SUCCESS in 1h 03m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7a632b9e4cc34f0d927ac5a1a033a506 : SUCCESS in 1h 06m 32s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05132f5b07d647339b9721994a38fee2 : SUCCESS in 1h 17m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7b2ada3817948219f4a754434edbb06 : SUCCESS in 1h 50m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54dce1a618de420a9b361f1aab67deb7 : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3ee94234443d78265a8c7d5191019b48a846f4b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1067fc4b89b41e39a2cab1260749dc2ac3c93bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 18:54:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0ef3ebb7e34b09a6d687f4cd179eb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a037aa68ca64601ac9d806a86efd89c : SUCCESS in 4m 20s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb4f77ace39c44da8752a97ba9ad515d : SUCCESS in 12m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c03bec6d8c634975abae0385c9a2f90f : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9f6617d7d5846a6a395ed0de4f842d4 : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/69c33346129a49379af3006124651909 : FAILURE in 1h 19m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9763690933614647a63a5df1f59356a3 : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba796c8c659b40618f1d5b65bccb2002 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c41827f5943845b8a22c3ae8598774d1 : FAILURE in 46m 56s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c2b8b10805bc4ae6bf756423da09c9c1 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e45e3fb8f0d44531a0e3c1d26f25235b : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867e785d4e7041a29abce81d749d3386 : SUCCESS in 1h 14m 22s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5da28c1e3344098b4e3063e19ef9869 : SUCCESS in 2h 06m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd5afcf2aedc44ecbf1f13cddddcc75e : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"535d54f687597c7c1b76a45451b28c6c29d97956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-02 15:27:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"907651d37c6d31a7a716ddcd588c8405c475eee8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-02 15:27:31.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b187de1060f17e0631822bd648f5bdf89b605ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 18:41:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b969414599ee4fafbdfaf94d47852bf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7baf501a954b8a8b1dc0ade6d2a26c : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/917234efeb4b461db9eaea7ac83aaf49 : SUCCESS in 13m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb4a40ae61b2469da25d88f4d2a181a8 : FAILURE in 1h 54m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a3b2727c1344a81877d9ab90dd1c4e8 : SUCCESS in 1h 29m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afd713df62da4e839e349a6a9437ff71 : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/de446a6ec48249c293ae5a9422bd9637 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7484fedbc1a741af934c77ec0d37d144 : FAILURE in 1h 19m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92062bd9da3f4c32bf1ab38a9e6c7d77 : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fae7db532834630bec8162c13cb7b65 : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d180403ec28d467eb8c16691389ab73f : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b7fbe25ba2b4148a46908afee270c2b : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06e30daacda84eba9b5991041555abb1 : SUCCESS in 1h 06m 51s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d22be426ced40529796532bd8e0ff63 : FAILURE in 41m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8493729fa56f4fe383cbfb8fd1d30a49 : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c7bf04e70e7ae82ba7d7486b0237be2a40e74d0c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-03 07:19:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"705b1fa4bd8d9d290fa4aa2f7cd51c47b0d8cd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:30:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357de3bcdb63489894302a84508f2c29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e5452ab9f24a85a3492b275bf3add4 : SUCCESS in 3m 50s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b70982766ae4ccfa21b0f8d22ac3482 : SUCCESS in 10m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a6d4e4d0f0145d1baf0fb422e738574 : SUCCESS in 1h 19m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/77c364fbc110428581ddf2ffd8213416 : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41859298a35a4a4993f2079a526905c7 : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/01039e806ec94de1867843e8ae710e75 : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6b6fe70c8504d15945411a6253a3ddf : SUCCESS in 1h 37m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46dd9be73629482fbc1efcb768eb5de4 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e3924fe2b904a8ab03bfe6a25d53a46 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cded81bc27044b929da3d5a7ca7bf970 : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/49bd159c8b144313b2633e80a4f9b93c : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dad90ce33c594b6a9b5846421c5c94b2 : SUCCESS in 1h 07m 23s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cb942c06e9c147248394b441c6b7532b : SUCCESS in 2h 04m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/646f0cd788234942b6b97cccfbc3be89 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6b7bebf753ab16c74d4fca67943fbcb81be85349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 09:32:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22245e99dbc1dda96caf40de4dbebf14b35ea0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:58:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a85a6f72ca4b76af7a199b6320a459\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b3f4d1a720410d94de8940d94074cc : SUCCESS in 4m 13s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d9fada1b5004299b9c455692f5bbe95 : SUCCESS in 11m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c41f7d9dfec2486a9edd332b69b394b0 : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84a27a1e9e3141c4a4945b031a3e6e61 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d080ebe6b254b7fa5687d90ee136add : SUCCESS in 1h 26m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a262b557c8d849a8bc1a2f5cc676c9d9 : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cbe3f3b79e0486db6abbb097c2dc4f0 : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef3bc5f7ac574732bfec104b42570658 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/861edfcef5444043a0095c581edf497f : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/29af85bedc064b5eb909900eb4e36152 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96c7894eaffc4ceea170ace8428ca645 : SUCCESS in 1h 05m 42s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f5b0f81f4b64662b232068c0a472253 : SUCCESS in 2h 03m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46b723a739ce4731aa846c1994fcd5a4 : SUCCESS in 1h 38m 10s","accounts_in_message":[],"_revision_number":2},{"id":"756f8afa4a51bfae605ba96c3186fe513dc47504","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:58:18.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"8365774a58b3cd474a6c259a3d13c987eeefc11a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:59:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d7a96cc9db4b44adb6b3eee61209f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b07ddbb312d49aabbe04fefea8c0809 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b57dd9b8378aa351321a8ed4943ee4186479d1d","revisions":{"b663cb17d0f556d2723dd3efda0c3a07fbae5827":{"kind":"REWORK","_number":1,"created":"2024-06-26 13:30:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/922819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/19/922819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/1"}}},"commit":{"parents":[{"commit":"87a3d651cdaf90ef972e0063f4ada06329773206","subject":"Merge \"Add distro infra jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/87a3d651cdaf90ef972e0063f4ada06329773206"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-26 13:30:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-26 13:30:46.000000000","tz":120},"subject":"Remove xinetd clean-up tasks","message":"Remove xinetd clean-up tasks\n\nChange-Id: I77cc32f49864f7caaad52150cc6684a488caa128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b663cb17d0f556d2723dd3efda0c3a07fbae5827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b663cb17d0f556d2723dd3efda0c3a07fbae5827"}]},"branch":"refs/heads/master"},"7b57dd9b8378aa351321a8ed4943ee4186479d1d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-02 15:27:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/922819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/19/922819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/19/922819/2"}}},"commit":{"parents":[{"commit":"da6abc6d0af2d8f082652d4040894f2ef8958250","subject":"Fix condition in cluster state checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/da6abc6d0af2d8f082652d4040894f2ef8958250"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-26 13:30:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-02 15:27:27.000000000","tz":0},"subject":"Remove xinetd clean-up tasks","message":"Remove xinetd clean-up tasks\n\nChange-Id: I77cc32f49864f7caaad52150cc6684a488caa128\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b57dd9b8378aa351321a8ed4943ee4186479d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b57dd9b8378aa351321a8ed4943ee4186479d1d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
