)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~924355","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/noble","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-07-24 08:02:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-07-24 09:49:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c","subject":"Remove installation of libaio1","status":"MERGED","created":"2024-07-17 18:16:01.000000000","updated":"2024-07-24 09:50:46.000000000","submitted":"2024-07-24 09:49:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924355-osa/noble","meta_rev_id":"817d5b62fa04bd52897d0cd91cbe7c8e8e01c530","_number":924355,"virtual_id_number":924355,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 09:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-07-23 13:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-07-22 11:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-07-24 08:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 20:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 11:05:03.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-07-23 13:01:29.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"ab047aec88143e36a4b01c7aa0343014339e2111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 18:16:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d6841bdada7cc6abd8d84181563f5f6e2b37e8f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 18:16:20.000000000","message":"Topic set to osa/noble","accounts_in_message":[],"_revision_number":1},{"id":"a79c2504256bb46950164012741d9e68c7008c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 18:30:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f597b8e72db597a66c8d6d1165a2c797ca51e2ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 20:27:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/842d401c6908464f928269a0593e49aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7b68438dda4b51b915c63341373aaa : SUCCESS in 4m 58s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdd72b5b7b244521a5305c85931f2874 : SUCCESS in 12m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9532a1ca964149298df0078fceb29bc8 : FAILURE in 1h 07m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1d31bbdb56848b3b0b8bc527456f849 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/803644e0f50f4df2bd2938925161e82b : FAILURE in 1h 09m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/46a2f6b3cb8c439890d6d25b386a7c87 : FAILURE in 58m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cf127a0254e4588b996a38e60de3a22 : SUCCESS in 1h 06m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd9531b9ccd4e829f823e5f4431c6a0 : FAILURE in 58m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2e33c1a55694287a8d974e26517c33b : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2d445c6012b49f9b6318d8d09308889 : FAILURE in 51m 42s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d474c31febbc4474ba18dc64a811e841 : FAILURE in 45m 07s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/130903c4ea5241a587b82175d5ffe3a0 : FAILURE in 45m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/917347e582a14d64af07caad2a181ebb : SUCCESS in 1h 54m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/943f833275c34fd5be58d9eead8b9029 : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6fd74b7ce6667c94cbce3a38ae539e0d87bcb671","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-17 20:40:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43515beaef5048d1aa850347c84e9195d754214e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:55:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb726010c01b43d69950ac9819bd2973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484c8347bc29411e8de49a7d1fcb2f25 : SUCCESS in 4m 53s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6824bf8c7b38443bb4c324559ec3d6c8 : SUCCESS in 12m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d705d8bc690f4c98a6b654c5f2c83bd6 : FAILURE in 1h 09m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5219949f1b7448488bffef46a79ff315 : SUCCESS in 1h 32m 48s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e40ec2af9073408ca3409b78e1c69828 : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/82692ceb3f174fb7b2ab9f4b930ec177 : FAILURE in 1h 03m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa8a2d7951ca439480f862abb8164797 : FAILURE in 58m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dc9416583a346769dddeac3e8ee499f : FAILURE in 58m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4917ee28c8045f5b813d39b45608314 : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e329f6c0ec7347cfa53afc94d53bae36 : FAILURE in 50m 26s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4d9512f0885148f5ae7c974d9f1fd213 : FAILURE in 47m 54s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/044a4ccfa342411b840d406baec8ccaa : FAILURE in 55m 29s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/872b95043549446aaaa1e14580b62331 : SUCCESS in 2h 00m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd24f6037464755bd22746a40f54387 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":2},{"id":"71d6105f1e4e00a8f56368efe87392de1d681669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-07-18 07:47:27.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a777c922ab9b860c440981d7dfcfa5d8123ad6b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 09:56:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/009a7447b0da40a3b4f44cac2862c60d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cdc413b7df4d52956825bfdc9c6991 : SUCCESS in 4m 16s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bdbe5ba73bc4cbe816296ca7041c401 : SUCCESS in 12m 50s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/43e2e6211545493ba9e145a94ce25300 : FAILURE in 1h 01m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4049cd9203a04299b6a2238ab4ab9846 : FAILURE in 1h 04m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/024cc72ae42e44a4abb0ee731daf1431 : FAILURE in 1h 11m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b995262c45e8468cb8ae74ec561640a1 : FAILURE in 56m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/711dea4db2e4414eb92735a6cfb2307c : FAILURE in 56m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b22eeb1937a4610b02fdd646f6661b1 : FAILURE in 54m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/496e122249aa46d799c776b9cc6f58a3 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4989b4f050e34d9d8cf4ef07cc69a824 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d1c162e54784d038a6a7453d15394a0 : FAILURE in 53m 45s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2075cd4cd725465688aa9553054d2998 : FAILURE in 48m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04a9055eb77548649572bbcc42c650de : SUCCESS in 1h 53m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e9935458c6422193ba061fbcd1eb35 : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5b68645de22c633326e016ddfbaa0b60dffd7633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-19 12:23:20.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dadc7d62420953724860ff97e0098b48df7417aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:37:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2a1d217d654d189f993cd62d5feaaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97c9d055a9d4e4c954d5bfcf2c2ba84 : SUCCESS in 4m 19s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/363c0fe7e3e14670977b25ae743d325d : SUCCESS in 12m 08s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e508c5d19f94ab48349e087aacc678b : SUCCESS in 1h 23m 40s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ac9ed04e20f45038cf29ae7ec378079 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c00d0bcf3043411d85f5e251be1e00ad : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d562402564cf4768aa490efe46dcecfb : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b398fe02265e4b2e8ff455fd04d14333 : SUCCESS in 1h 12m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/701a9ceb492c431693399e1999560e24 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/389cf0ef095b4b77ac1b34da4a5afc24 : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f4d768ab82c4f388a15eb92dfff6e60 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/316f1d5013fb4cdbbc2254be94b7b802 : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f48e3c94dec14d2aac8707afb5200245 : SUCCESS in 1h 08m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37ad9bbe3d134e2289ce6aa74d78c59a : SUCCESS in 2h 04m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68118f1e3b9242dbaa54f941585b07c1 : SUCCESS in 1h 34m 23s","accounts_in_message":[],"_revision_number":4},{"id":"02f6b628dc3fb3154e493d02c12eefabbf081fdc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-22 11:05:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bf4743c6a4840707701aa920166eb635bf55a3fe","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-07-23 13:01:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3d6b02cb62a6cdeeb0b3c0df9d875453b4a69fb8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-24 08:02:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7636d8d2adbd11be295c43ac9fe334a689e81c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 08:04:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"36550a2e2c1edaf72374b5bf1fa769f1e105223f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:49:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e21723ffe2234a53a9c5d22f8294ffc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1ab47e85704611ac42c2e3d8f330c1 : SUCCESS in 5m 42s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2dc8e7dabc24d15a99b4fe3ac9d42a2 : SUCCESS in 15m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57a6a91bf4f14896970ada1addc2c5e1 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a08b730b16ea460c836850f7b1184120 : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1c5f800d274348b4a47a61a5dd6fd603 : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bab416f8ead34c7abd370bf664e46f3f : SUCCESS in 1h 07m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/028d2bb3ef644f4eab2f76a835b09c1f : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65bca08dba9a4a208c9426017a87bc0b : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/66d38f82635a46f8a06f039b0aea5971 : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/34103e83f4a748e09b6b456b31ad56f8 : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc049d43169844d8b376d261e29b84ff : SUCCESS in 1h 07m 22s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c244f0b73e6b4ff186ad8c1fa67143a1 : SUCCESS in 1h 36m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f5857dacf0405f8477b78a41f06e9b : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":4},{"id":"122ed1843b27c7f7743d5f581e918e7eca65795d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:49:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"817d5b62fa04bd52897d0cd91cbe7c8e8e01c530","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:50:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3a33ba6bc943deb5122e9b250fec33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f919f36bf3486ca58d4255e71a5371 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"516c1c255cdff9ba6fd4cd75f2d432a329f66a32","revisions":{"64d69c7d60a8404b4761602b49fe23906e332604":{"kind":"REWORK","_number":1,"created":"2024-07-17 18:16:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/924355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/924355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/1"}}},"commit":{"parents":[{"commit":"c91daf74fe7a374231ce28ca3b3252eeb527c43a","subject":"Use mirror.mariadb.org to install packages from","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c91daf74fe7a374231ce28ca3b3252eeb527c43a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:15:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:15:31.000000000","tz":60},"subject":"Remove installation of libaio1","message":"Remove installation of libaio1\n\nThis package does not exist in Ubuntu Noble.\n\nChange-Id: I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/64d69c7d60a8404b4761602b49fe23906e332604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/64d69c7d60a8404b4761602b49fe23906e332604"}]},"branch":"refs/heads/master"},"03924c6572bf7c2ae329df5af8cbd01ee645112f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-17 18:30:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/924355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/924355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/2"}}},"commit":{"parents":[{"commit":"7b57dd9b8378aa351321a8ed4943ee4186479d1d","subject":"Remove xinetd clean-up tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b57dd9b8378aa351321a8ed4943ee4186479d1d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:15:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:29:49.000000000","tz":60},"subject":"Remove installation of libaio1","message":"Remove installation of libaio1\n\nThis package does not exist in Ubuntu Noble.\n\nChange-Id: I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/03924c6572bf7c2ae329df5af8cbd01ee645112f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/03924c6572bf7c2ae329df5af8cbd01ee645112f"}]},"branch":"refs/heads/master"},"1db46bb270e4dab7e397b205297a5a883d18aeca":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-18 07:47:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/55/924355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/924355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/3"}}},"commit":{"parents":[{"commit":"7b57dd9b8378aa351321a8ed4943ee4186479d1d","subject":"Remove xinetd clean-up tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b57dd9b8378aa351321a8ed4943ee4186479d1d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:15:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-18 07:47:27.000000000","tz":0},"subject":"Remove installation of libaio1","message":"Remove installation of libaio1\n\nThis package does not exist in Ubuntu Noble.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/924384\nChange-Id: I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1db46bb270e4dab7e397b205297a5a883d18aeca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1db46bb270e4dab7e397b205297a5a883d18aeca"}]},"branch":"refs/heads/master","description":"Edit commit message"},"516c1c255cdff9ba6fd4cd75f2d432a329f66a32":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-07-19 12:23:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/924355/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/924355/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/924355/4"}}},"commit":{"parents":[{"commit":"7b57dd9b8378aa351321a8ed4943ee4186479d1d","subject":"Remove xinetd clean-up tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b57dd9b8378aa351321a8ed4943ee4186479d1d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2024-07-17 18:15:05.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-07-19 12:23:20.000000000","tz":0},"subject":"Remove installation of libaio1","message":"Remove installation of libaio1\n\nThis package does not exist in Ubuntu Noble.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/924384\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_keystone/+/924474\nChange-Id: I37ffedcb9c76ebaf1b2cfc27a2ae90990c39cc7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/516c1c255cdff9ba6fd4cd75f2d432a329f66a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/516c1c255cdff9ba6fd4cd75f2d432a329f66a32"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
