)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~935665","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I7ed89d7e8a6338564d86548a797b9f40ba43b0df","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-26 17:18:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7ed89d7e8a6338564d86548a797b9f40ba43b0df","subject":"Use rhel packages for all EL-based distros","status":"MERGED","created":"2024-11-19 15:41:35.000000000","updated":"2024-11-26 17:19:53.000000000","submitted":"2024-11-26 17:18:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935665","meta_rev_id":"aa36c17264308d2aa148e1aec9667ed36b2e3431","_number":935665,"virtual_id_number":935665,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-26 17:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-11-26 14:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-11-26 15:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-11-26 15:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 17:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 14:09:43.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-11-26 15:10:37.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"4628d919056b0ead1392a6da6fa37530b91fbea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-19 15:41:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf6ed15901c9aae9b762432e968c9dac8121408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:36:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dceb9f619993492e8834b7bead48ac60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66b5b08574b4ed491b92b9a058fa3e0 : SUCCESS in 5m 10s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/40edf61a1f524fdc9d66d457f4c66546 : SUCCESS in 1h 24m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5738885d2d94968bdbbb18ecebd592b : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2909b3890bf846eaa74cc757b5e3e50e : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ded596c4d57e4d159eeb4f13b5bf06ad : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7951cca93c5b4f719593626ca43bacff : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/803f3dfd2d7d40e680e370a75ec0b699 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85a17a307520456885df3e21ace3d510 : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2cdf941d4ab4a0ca94904063c9e6aac : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5326d9203634a47af3f1d06cf412178 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1adf475e361b49b1b9255983dc1d94e1 : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5eac4b880f7b453cbc0449ba2eb93b85 : SUCCESS in 1h 03m 09s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/98931853944a46b28e4b1494d42d0102 : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dbc4456f4af460faa4736c3bf6dac78 : SUCCESS in 56m 27s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fa360a9baf2409eb934650caf1d5147 : FAILURE in 1h 35m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/971f91a6c6de40a6b0053d9b4ccbff22 : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8e83d00b94ee949f5f3b41cbd6c152aea0318b98","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-19 18:13:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fac3f84a2138963f0210346bc3fbe8d37e90377e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 20:22:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd901647ee9144f1bb0ac32cfabb21bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9d3b3931b64da592d9a0147298d6fc : SUCCESS in 4m 08s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3c35ec043a147a78816eaeb697b4f0b : SUCCESS in 1h 19m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/89335d559ae141009107b0adbda0bcdd : SUCCESS in 1h 31m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d92447273c4546c4ad487c65b99910c1 : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1d153658a5ea49848cca96e08f799767 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8106ab3275b49ce88a963dee9bd8a90 : FAILURE in 1h 03m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7cf7947199f4817afd36930dc519c3f : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d242074363dd4f12b53933cbae7aeab4 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/637258b83c31491883e248e5fee8d6b3 : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e419690ef9e446838598946ba3ebc300 : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c0b30b8323d4337ba00cd22d9aadbae : SUCCESS in 1h 24m 41s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccfd653f63144159ae3be779efbe81fb : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6390d759931a43b69e858e7a0e04c470 : SUCCESS in 1h 00m 32s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3af6c5839134b289d573c33da594f87 : SUCCESS in 1h 11m 04s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e1ee16fe23a467b875aeae2e0ff0ebf : SUCCESS in 1h 56m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36c281a2a9be459c8606a14b25133747 : SUCCESS in 1h 47m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bf171eee7d07672c0d5b93ce3a77a570854af71f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-26 11:15:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c76bf999a6fd4578511ecb054dde4f85f64fbd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 13:06:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89787a28acdb4e968b25b066cb3233f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb02ed2afd9942dab71979ea6e3c4aee : SUCCESS in 4m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/23a8b264f8eb4023a2828a15f6146baf : SUCCESS in 1h 18m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f14249b8ae04460e890e526f293625e1 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bf082031d5940a4838abfa17b52dd17 : SUCCESS in 1h 29m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fd3348f572134d35a3d064a14d62cf7f : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79bb6fdb45ff490280382a51d8f08198 : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/613dd2f2ad864c639ef0ecd2af03a620 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1f9d978c14f44baa819062740a0d51b : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d95fb07c9eb249d0923f3b402dac01e6 : SUCCESS in 1h 12m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8a264c8556d4d0fa73ff66fabf39b5c : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5085bd881d8745ffb7abced7f78c148e : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/534c8e1d3c1c4283a9a6585fe527f6f2 : SUCCESS in 1h 01m 08s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b293f31faba74003a5efcdc8ae3bbe7b : SUCCESS in 1h 00m 15s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cd6207de65a4df1a26545bb40ddd39e : SUCCESS in 59m 36s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a4a47c69646471ab76d6fc601025d7c : SUCCESS in 1h 45m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b463cabb0a3410f8dcd53d6b9bc2c72 : SUCCESS in 1h 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1c9dcf00df68a166e0d06dde797d2fdab28e3b39","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-11-26 14:09:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf8db827041c090903bc7aa182e4d4cb6ec1ad56","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2024-11-26 15:10:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b060fa0afddb74298494a2f292112910bd487b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 15:13:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73b69a4bc8dffafb48f8eeba8e78cbceca70b125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 17:18:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9348a14a25a14337ac3ae59b93a6dd30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0d14afc6484adcafe75f9fd9b01099 : SUCCESS in 3m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fe0fd5d99d34a219d0adb37d55305bc : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/efe7c5aeb9004370a9e9efde5da74918 : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a6f4f1d601dc49cf9dddb3fa791b5d13 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5b41363b8434e369813b28111015eeb : SUCCESS in 1h 12m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac6bc34673dc4cc084fb5bc91dddb650 : SUCCESS in 1h 10m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af2637aba4d44e8db6d41627befbbe64 : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d833d29fd74e42bb57c2796f9231c5 : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e89fd51bb124e8581a7c430f0beaf06 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e10ce848678844f2b7f946ac82a21184 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50deea28bb504cc58c86fad3aeea8cea : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/112f774995a64be99d17e9cd872a81c6 : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07ed4e9dcc4a40ea9ddcd1490935b8f9 : SUCCESS in 1h 06m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3457374ec084429b3e6a815bae9aa8a : SUCCESS in 2h 05m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad60cb54ba8142e49a90abb9322b6fb7 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d565fa03916b00373fc12f0944be63105ebd88a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 17:18:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa36c17264308d2aa148e1aec9667ed36b2e3431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 17:19:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e161d7fafc6044e9aa8d62f4b1e09d2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab2a8d269dd4255aa0cfdccaab9c067 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f773a8fb23a015969e886934649754a81d561601","revisions":{"f773a8fb23a015969e886934649754a81d561601":{"kind":"REWORK","_number":1,"created":"2024-11-19 15:41:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/935665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/65/935665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/65/935665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/65/935665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/65/935665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/65/935665/1"}}},"commit":{"parents":[{"commit":"ef1bb11cb3e9e10639f99a09cf69dfcca6d0c5d6","subject":"Update mariadb to 11.4.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ef1bb11cb3e9e10639f99a09cf69dfcca6d0c5d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-11-19 15:41:24.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-11-19 15:41:24.000000000","tz":60},"subject":"Use rhel packages for all EL-based distros","message":"Use rhel packages for all EL-based distros\n\nWith migration to mirror.mariadb.org repo path for CentOS was incorrect\nas logic for CentOS/Rocky is quite different in terms of resulting\npath in repo.\n\nWith that RHEL might be a universal one which would satisfy both parties\nwithout need for more complex logic in role.\n\nChange-Id: I7ed89d7e8a6338564d86548a797b9f40ba43b0df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f773a8fb23a015969e886934649754a81d561601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f773a8fb23a015969e886934649754a81d561601"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
