)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~939798","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I65a79d82816665dc903c259eab11c8f02361e51d","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-04 22:43:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65a79d82816665dc903c259eab11c8f02361e51d","subject":"Use module_hotfixes for yum_repository","status":"MERGED","created":"2025-01-22 13:48:15.000000000","updated":"2025-02-04 22:43:59.000000000","submitted":"2025-02-04 22:43:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939798","meta_rev_id":"61497f523be061a526d15c10e9ce2b2c24f8e144","_number":939798,"virtual_id_number":939798,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-04 22:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-01-31 08:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-23 22:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-01-31 08:37:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-22 15:53:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-23 22:12:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-01-31 08:37:02.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"35df5729e38b95efb136d50e01dd6f090693906f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-22 13:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd212e8c3de3ea96782a45756bbb26cff41fdc2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:53:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/362bb564dcf04a24aeb6471d8559c7a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caac36a6919940469808ca412f953f84 : SUCCESS in 5m 27s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd41ca9be2154274b8c18f99cf1d2af2 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/92feee47ec5d4e638cc091602a591ecc : SUCCESS in 1h 25m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9cf9538adc1436695ce2f6c01916381 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25e60961810043f1bda85d6d1dcd03ad : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fe5ed814f60043ed9ed63d7ca5a4d4b4 : FAILURE in 1h 16m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f53408b5fdd40bc9eff44efb673159a : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7972a8830b0142a08dd20d61eb5a683d : SUCCESS in 1h 10m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ab3ec2ec7340fbbd61a2c092afb5cd : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f679b22c5227491cbfdf9e7b8f1454e1 : SUCCESS in 1h 19m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaa63b9074244b3ab79e010267113ec1 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b483ac44e24cb181eee202c61e7388 : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e16b7d55a50d4e168f68d656f1cb0d36 : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7299c6242c014d4d9ab72cb53c68b3df : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e47d843604fc43fe97f68840284257f9 : SUCCESS in 1h 03m 15s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6985549bcc7940feaff00f2c0f23a06c : SUCCESS in 1h 50m 47s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e96e5930be78415790eef9308c9c4433 : SUCCESS in 1h 57m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810c135052fb417483ffc6fc1a99c05b : SUCCESS in 1h 25m 58s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a60178d06864ba1a0158234c28909c4 : SUCCESS in 1h 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c0117482fe6f348799c0d50dba13d7f45bd47a76","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-23 22:12:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1af3c76e89d20a04bfc5d2b10a76ed6bd139b8f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-23 22:14:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f56abbb6ac3d077c55460bdf62669952b0762c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 00:41:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1afc6d54d94c400885e98e9b5ec9177d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd96417ff7b4d8792d713629e03b8a6 : SUCCESS in 4m 20s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d13bd1326b53433d80c0a6a6c26c26ae : SUCCESS in 8m 01s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/286b1c74f27e4bcbafb865e7e08691c1 : SUCCESS in 1h 15m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d78878e20214de3be7f7bd7cd60914a : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a4352dc41ca4a2b9526a818c8e1b5db : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0127600d7cd2437cbc8f2c978f867fca : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/880097fc63124232b4aa27287b4828b8 : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad9c223c09e41dda916c5e26442c5b2 : SUCCESS in 1h 50m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4efe29f358e643bf82247c7f430e7bd0 : SUCCESS in 1h 50m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b57fd6f9724cf0a6a819318adb3773 : SUCCESS in 1h 36m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d08419e4ae33418b82153e279131b314 : SUCCESS in 1h 50m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b390c37067334e0ba5d329fbabfe318b : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f37f2c20670447c9baf6b97ce24d4366 : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/44cca1c4964945db9c9879e5993f5352 : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4a46afe87cd4946aa4ce88aa5de19b9 : SUCCESS in 1h 34m 16s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b71adc6c2b554f2b87843020cdd39afd : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ca99702763c41cfadcc611c4ed33519 : SUCCESS in 1h 05m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132aa070b33549dfac8a23d2cd80de37 : SUCCESS in 1h 54m 59s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7847826d4372401b8ec65e9f9aae629f : SUCCESS in 2h 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c9c75214307fb5775ba33335258939c436ef91ac","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-01-31 08:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"967c4cd3ff4f659b714d20c10d28d07112461734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 08:39:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"007008ddf0fabe400534bfc1095cd724e73c5d8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 10:42:28.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/939740 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"4b5a0ca91ebb28701677b9ce74604b5e94f2224a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-04 18:33:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21803cf0db7876037339765b941b1d6ced1e247e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:36:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faca77171a44453cac6810de0231f173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9efd302ae0d409aa7a5e19e23da6241 : SUCCESS in 4m 16s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad0aff94ba2043a39c1b4c3bf7cab172 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/072c92446ef642dea6b4eb597df1f5d7 : SUCCESS in 1h 20m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59ba1761916c4fa3bc8a87da94975dc7 : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c42c50363e2c4159b6e07a393eee2f47 : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b3d0c1338ef448cdb104734ac047024b : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85dde78c007542bebe9acbd843f32680 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1c91b1e9fc0429488941918b0028028 : SUCCESS in 1h 21m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68670ccbc15e49e7922d25da5aed09ea : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e74652ccb8646c8a69d039b638cd1a1 : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c3d8ffecabf4f8d8b884a986e511306 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93c6f87967824efc81e4c214e814e154 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff6d4b68fc424b48b29393e0679070a0 : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/67d55a951be44411b6f3138e45167488 : SUCCESS in 1h 01m 57s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/087b61fb786f4247be0faa3d4f53e615 : SUCCESS in 1h 10m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6e3e9802f7041fb9f1328c8e444624b : SUCCESS in 1h 51m 24s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04db09b82111432dbc43673ec6ac5ede : SUCCESS in 1h 51m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1976c55fcea4362899f58c55ecb6567 : SUCCESS in 1h 48m 16s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a64448abcbe4497a81b4fc313c94cfe : SUCCESS in 1h 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b82aadd5dfa8600d52e97de78eca8600a99fde56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:38:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f378df8583cee6b9f25d880deab9f1766d9aa60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 22:42:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161133e1d571456ba8f6d829514a2daf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373a4e20782e4f339a9e6acbbad6fc9b : SUCCESS in 3m 56s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f453a5048c423d9fc2e7811bb912e3 : SUCCESS in 7m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c83a71f9d3c846d5afc7f62c531f1dd7 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b5b8ca3a2fe04582a07541a0803133a2 : SUCCESS in 1h 07m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0853ef04471c4ae4884240c2d3628678 : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc038dd8a9942eabfe6e0fa70821adc : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/751bf42fda7d48dd89392c3160f13de1 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/283eb33d071b44cda98f8ce815bfda54 : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39c2ae495ef4924b131e6a562b84d7b : SUCCESS in 45m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b996f53078a44c68e81b22781d9a651 : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/600b66989a894017a4bbbbbbef212f5b : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ffe92dd8892741cdb8993010d26dcf0d : SUCCESS in 42m 13s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/758481d33ed4468eba3db29ea85c33f0 : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eddfdc9c2f6445829809f2aac0bf51b6 : SUCCESS in 1h 08m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81eb4d4c8bc04cc29332539f8983c401 : SUCCESS in 1h 01m 23s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2388d501191f4aa3b8e931316c0285cb : SUCCESS in 1h 58m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82b2ba77364c45dcb4db92beb419e838 : SUCCESS in 1h 33m 54s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b52f3e8933514741a3454f745a92f0ed : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"845a4c32a9d025e11713daec12d762ef805a885d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 22:43:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61497f523be061a526d15c10e9ce2b2c24f8e144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 22:43:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605a203e544a4a95999894664375fc4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92568a7d9cf549bb8d13da8313958cf2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e31cc13254684ae21710145224bd6b84e3ecb1fc","revisions":{"e31cc13254684ae21710145224bd6b84e3ecb1fc":{"kind":"REWORK","_number":1,"created":"2025-01-22 13:48:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/98/939798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/98/939798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/939798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/939798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/939798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/939798/1"}}},"commit":{"parents":[{"commit":"25eba0bab68041d923b60648688b024177975ae9","subject":"Extend example playbook to contain valid values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/25eba0bab68041d923b60648688b024177975ae9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-01-22 13:48:04.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-01-22 13:48:04.000000000","tz":60},"subject":"Use module_hotfixes for yum_repository","message":"Use module_hotfixes for yum_repository\n\nSince ansible-core 2.11 it\u0027s possible to use module_hotfixes argument of\nyum_repository instead of standalone task. This improves role\nidempotency and simplifies logic.\n\nChange-Id: I65a79d82816665dc903c259eab11c8f02361e51d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e31cc13254684ae21710145224bd6b84e3ecb1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e31cc13254684ae21710145224bd6b84e3ecb1fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
