)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~948711","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ic8ae457f734dd9c73ef40d59c9c5d441af552d47","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-10 11:45:17.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-05-06 19:33:38.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ic8ae457f734dd9c73ef40d59c9c5d441af552d47","subject":"Bump MariaDB to 11.4.5","status":"MERGED","created":"2025-05-02 16:03:58.000000000","updated":"2025-05-10 11:46:03.000000000","submitted":"2025-05-10 11:45:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948711","meta_rev_id":"48dfcc783ead76951d48aa3962274869c0732e58","_number":948711,"virtual_id_number":948711,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-10 11:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-05-07 10:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-02 19:49:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-06 19:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2025-05-07 10:17:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-02 18:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-02 19:49:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-02 19:49:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-05-07 10:17:52.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"1d4473ff9adb2093badefa08296f5de39a35fb33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-02 16:03:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6bb39faec5be2b0df1e2cda00af67d5d1cc117c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 18:00:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfba7504fff54921b45825c3394f524f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1578b99672e4b9cace599ae61336488 : SUCCESS in 5m 36s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f899a779e16b47a28f09010339537aa6 : SUCCESS in 6m 16s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/24f7cbf90f784685ba68d790b3341cb6 : SUCCESS in 10m 42s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/40231b51d0504567a741d4e93cdefd0e : SUCCESS in 11m 22s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5665d2679da54be89a6ffbe475770e3b : SUCCESS in 19m 17s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/464de38b0b334c22a4cbf7c2809440a2 : SUCCESS in 13m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c354a0da913541e9a88d89a2c3cd7c49 : SUCCESS in 11m 13s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9393cd66e4a408ba8dc7b24cddc9638 : SUCCESS in 1h 13m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7aefaa60ee98421cb91e898d73cbf8b8 : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f1fadb0545a445d9420638e35620f01 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f571e7981f147d8a0d048bea1490028 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c987f678b9d470f90e1ad09e32b8714 : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f4c865be664177af03a45c4525dbb3 : SUCCESS in 1h 13m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/738535af8c084eeea2f32c101e51ba5d : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba99ca468a994a99bd0b21c267577436 : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6f7380e92094b5a9dca99573cada578 : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b0eb15ec500430db47e13c4a1e2d89b : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5fc5e1bb8da4efb934895d142fdb28e : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/be178b10152244098bf894301a19bb56 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e776e7bfabd844aea19459451f6413b6 : SUCCESS in 1h 24m 33s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2bc004fa8f724abe80061ced39922c5d : SUCCESS in 1h 25m 54s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fa63c9b70c74377867a38f40554362e : SUCCESS in 1h 49m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/305fcacf6db84c3189e27e4d7a76d9a3 : SUCCESS in 1h 45m 36s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/896391c5d31e431abc2845d07ebed599 : SUCCESS in 1h 42m 01s","accounts_in_message":[],"_revision_number":1},{"id":"65c918e698073f048151ca7417c9dc35a29ad462","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 19:49:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"930dfe4e09abdad9e453d33ccf65861ccff56a9e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-06 19:33:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d192fb82653ea6c83123002c3f5ed12eaf531a03","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-07 10:17:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73c1abaaf8be05a6b0372e4f35c40bb396262e07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 10:19:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e3f170a765f8975c45ff50e26b82041c9359191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:03:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e63b3e122665489b8b2e1403b7bda8f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4018ecccef4e12ae7354a7bacbf7da : SUCCESS in 4m 46s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e286820f088c48d69cdfe2ab32ad24f4 : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a8ff10628b847659743bc7550c62536 : SUCCESS in 10m 13s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/619aae69d0884007b05788b071f6f013 : SUCCESS in 12m 09s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48449287cb084ba797bedbbc5f6c7d73 : SUCCESS in 11m 10s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e6ae48d6614af7ace4aa66cf3a4f61 : SUCCESS in 12m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd652fc8f024cfaa55cdfa4ab33cb59 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b628d50b3ca94e1783460ee32569f522 : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d67a4db1937e41a6b985759c8202b484 : SUCCESS in 1h 25m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9977083ddfe9464d8c82bfa86bedd00c : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a91673abfd6450085521d3ce1a13dbf : SUCCESS in 1h 23m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbebc4e6eb004a07b9802e22cac611b2 : SUCCESS in 2h 02m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d8c3373bcef4a50a3388231d71ee2b1 : SUCCESS in 2h 42m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aaa3a2cb23f4748bd4418359549677c : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c55682c3c5704250baae08accd2d4ef5 : SUCCESS in 1h 48m 57s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ca12e29705e4562baf9a824201ffc56 : FAILURE in 56m 51s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/920900c9b50b4588b3fc0bf6e0659982 : SUCCESS in 56m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1c9c4bcdb6904c3fa2c31ccbff2196f0 : SUCCESS in 36m 58s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0060984e931d40ffbb439ab723114086 : SUCCESS in 37m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c1f4359dd5d448594b9de68032f3d3b : SUCCESS in 54m 38s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39e700a5cc2c41b48fa9ee323a375b5a : SUCCESS in 56m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98674515d3f44b2e9442a54fbb983686 : FAILURE in 16m 45s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2c7c5f2b03c4e6b9286c56968e42f4e : FAILURE in 1h 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"295802d6c77abbd4ae083f0c6c512c306e0bf820","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-07 13:41:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50fbe15d0ce30980d60aa9598da649daa5b67832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 15:43:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b7cbc2b52f64580b4585c1632b85bd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f666cc1996408aae2ed072eab49d3a : SUCCESS in 6m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e247e3afa09742f7a5c0fe43840a93bd : SUCCESS in 9m 03s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa1aa390fcf3450e80b2d630823a280f : SUCCESS in 12m 29s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/dccc0759771e4da0a4cf525ee2551214 : FAILURE in 5m 12s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1610371832f74253a8f47e90e16f10de : SUCCESS in 9m 25s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/714bbbe3f21f412ab33e7f8ea4443e2e : FAILURE in 5m 30s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e24cc2aff4f847a9a494decbf2b86ba8 : SUCCESS in 9m 54s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9507c83cf2a748cc8fa296d560a4df44 : SUCCESS in 1h 14m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64ebf5e754e2468aa780244d5fd9c055 : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea32b7563d704b42ac1fa69bb87ef539 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7339c6d694a747fb80ed3752c4a1cb8d : SUCCESS in 39m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdb1d3f828f141cc93181957b3a1c8e2 : FAILURE in 56m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc53cdc161a6491080526a7b2bf03229 : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6de0144351f94a92840a6487ebab7751 : SUCCESS in 1h 42m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35a2c71b0ec346d39b3609e1d5028c40 : SUCCESS in 1h 44m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6a8d3d6f05a45cf8f866b05858ef5db : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e4bc369e63f4b26af510b9e9ebe5f66 : SUCCESS in 1h 57m 17s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/366e814134a643eab63a602c7d094989 : SUCCESS in 1h 11m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7bde4c40b40a4aa69ee104e09fa6efd8 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf87b0fe2aea415fb04fbbd2725f81a7 : SUCCESS in 1h 15m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4aff358d560a44c08aa45868649db069 : SUCCESS in 1h 51m 07s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5ee7c2edeb94d3f9095f1cb96e358f7 : SUCCESS in 1h 34m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3703f8c0be9f4b018ba17faf16f985aa : FAILURE in 58m 40s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c27db177b2541f7b2c17ebb5ac4652a : SUCCESS in 1h 46m 17s","accounts_in_message":[],"_revision_number":1},{"id":"db5d06c3212f1a1f4d1a265af0e12301495e3d37","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-07 21:35:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27e4517025192470b38760085cfb955d55d056d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 23:11:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ad8f8ec59814533b649987d343f8a5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912001e4d6c1474d8d59ce274f7bcce7 : SUCCESS in 6m 29s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3790a1f81c344b0dac54b9f3bb5b9690 : SUCCESS in 7m 20s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac9694613e043a7afd9138ac768a659 : SUCCESS in 9m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c74661c053db4c22bf08f38809548d11 : SUCCESS in 11m 15s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee3d0d431c1a475abe541086389ece56 : SUCCESS in 10m 20s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b29643f8bdf0432684f75b80d068429d : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9363fb429c2a42eb92efe709242e8ed3 : SUCCESS in 10m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/77134547cfab49388d22521cdb05769a : SUCCESS in 1h 16m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f8357101df54117a2e741b05a756ebd : SUCCESS in 1h 22m 05s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb27e011dda649f8a7ab4ad7efd89afc : FAILURE in 1h 21m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/10fe36d54d1f429b985decbd6ecd6161 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/681b86c354bc4a75ac3a58062bca5ea2 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9facbd2862b46dc8824af71431d558c : SUCCESS in 45m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2245c1cda1f41aca88aeba8859a24d4 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/569845b2ae304bd5b75ef769b1d8e87a : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ace3cac370459caec75758d8315eda : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f51085d262ef4cad81fd1aab5ac57761 : SUCCESS in 1h 17m 35s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6619599bcd8e4da7aca1dd5aaa9d45c5 : SUCCESS in 43m 06s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/08cb822b7e9a4e92a400c356d450731d : SUCCESS in 57m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aa1eb341d6449b78bdbfc8e093c329d : SUCCESS in 1h 07m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/402e96dcaf3147629f779e1de5dd138e : SUCCESS in 1h 32m 23s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2baf14f050ca40c888d90aafede1d5b3 : SUCCESS in 1h 23m 36s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccf0bda7be32493aa320ea151bcfee71 : FAILURE in 22m 43s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae68fc8f599e4a1391d73686b46177c9 : SUCCESS in 1h 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6207488cbf88ffb8d26578bdd201946b53f0c578","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-08 07:26:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"791452448afa04560646f86e260b7fa8775c26f6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-08 10:57:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bb44b37e68b9b05ea892b3ab72f632a529ae2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 12:55:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96116cc1ce7e433fba62f3596e75248f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bb5ce09078451781053c4ca3fd0c9b : SUCCESS in 3m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f579129e6d24573a3d733905313f256 : SUCCESS in 7m 00s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/111c5c32308c4b378d07bb83a2294ae7 : SUCCESS in 9m 06s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ef60bf2e434419a87c6fd3c4c38be8b : SUCCESS in 10m 21s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc717d00e6f64d43b97291e898bec24c : SUCCESS in 10m 55s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5eb77703ecf45178eba9af7f2caecaa : SUCCESS in 11m 00s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9dc287f982a4a70acac6fc5d3fee453 : SUCCESS in 10m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ef39703f9d547ef873538b0ff2ec8c5 : SUCCESS in 1h 05m 01s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a03aa0b9f2a64e7dbc209a8c083d5edb : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78f3964250994828b4787f2ccb68f725 : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/46984f9861b3419d8851cceb472d82cd : SUCCESS in 1h 19m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef4774f2ce3c4b6b9264621e511c795b : SUCCESS in 57m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/884847aecd1b486bbeede551e38564a6 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/035715dca326441886a0f38a2450555f : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82043210d7dc4966a76207543d5ef03f : SUCCESS in 1h 07m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/206fb8219b864145b349f15d13ca8415 : SUCCESS in 1h 18m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d2c53b2a23417ab887ba05a1b0f88b : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4d0c034c49b4ae0b46c8effaa7ba79c : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2cc4271e94cd42f39112fbc7c9632750 : SUCCESS in 1h 26m 09s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8fcc4e910204116bdab25efb117149b : SUCCESS in 1h 00m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fc59a1d51d4416ca31459eaa25420e6 : SUCCESS in 1h 46m 28s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5b86ca8c63e4cd1aedc67f10679cbf6 : SUCCESS in 1h 51m 00s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/600414f8066e4508a0961a97bed57c97 : FAILURE in 16m 51s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8ee0ed1abb40fab11c7e366e165b4e : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c65240c41038ef5b9df3a8df544c897c52beeca3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-10 07:37:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0dfa1504799067680907212fbb2ebc7cbeac730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 09:43:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d41aa4af36b4044a54a61c6acac0e9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d1bea418a34e86af6b4a2a9a95d234 : SUCCESS in 5m 39s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e18fd889942044e49e973d5f05cfb406 : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/54e8ce536b3e4a4e9c70751518250671 : SUCCESS in 9m 10s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/83899a51889e485d85d38c8a51e3cd43 : SUCCESS in 9m 31s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c3610c3f3764f9aacd3aaf4c46521c3 : SUCCESS in 9m 13s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/687a47463d694a8ebd9ed99f1748dfa5 : SUCCESS in 7m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/987a998ec7df401c84f431286a14fb37 : SUCCESS in 9m 49s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/067c2d5930604b17859d5cf0d90752dd : SUCCESS in 41m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7138c8265f84a9fb594d92c583d6753 : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7deb5845fbe44152aec18f73cb39294d : SUCCESS in 42m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6b357f8738d34a7eaa3df56d5379acec : SUCCESS in 40m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f24211f6616547f3b1516e2cbd4d137c : SUCCESS in 38m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0b335354689461fb105dc1940af385c : SUCCESS in 38m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd0c69473bc84153a4226bb0c968f539 : SUCCESS in 38m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/716f351f6946401cbc84f7845d934734 : SUCCESS in 40m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe53cfd21b4d4626902addc7d8ff9d74 : SUCCESS in 42m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ced69af2e72b46ac90e7d267c06461e4 : SUCCESS in 43m 14s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c233ad63a7d94c6ca4d4adc1783bd0b0 : SUCCESS in 38m 17s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6892073e9dd842e8b15381ad764c0442 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/000e50cf3e3f405db2a5f26fe216b7bc : SUCCESS in 42m 12s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f1cdd455af1445ca4c21c85ea0fc192 : SUCCESS in 1h 58m 02s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9d55c1419c6448593ca0795acd40bb0 : SUCCESS in 1h 06m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abab6cbb5c994e7f96c6b466813a0a99 : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acd1a22107144abbae6e2c83333c3952 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b47105dcbee270c1b1f6c280b6b296a7b45c0023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 09:45:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b310ee1f82157804ea6eae73405614e228257370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 11:45:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb77644cd0f4556b07a53699c2ff292\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c769deae00f842398dd3879ee75724db : SUCCESS in 2m 57s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb37eeae6e994824b773067303b72019 : SUCCESS in 3m 47s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/d245a0eefbae46ab84fb25ba02f3c109 : SUCCESS in 6m 06s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e90fa2f62fb544ac844827a5d1c83290 : SUCCESS in 10m 06s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d82376dbc6b54a6fbf1f025a61cd91d8 : SUCCESS in 9m 56s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ff58cd10c0c44ac91a34adbfb589e76 : SUCCESS in 10m 50s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35c4537a5f3b4c71ad9f2be1f798cbd9 : SUCCESS in 10m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ab7f529a524248b4957b6a12ca22eadd : SUCCESS in 58m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b433f8b80d0c4346a4ed75088aba7cc4 : SUCCESS in 1h 02m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4917f89b0d024f5dbba65f51e55d3543 : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7367befc37b44b9832dd5a6f683afc8 : SUCCESS in 55m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a5d0c2f2a1b4ecfbcb7f516a2e75b77 : SUCCESS in 1h 07m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e473e2d07f3045d4a77fbea425a7aef4 : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68f16b3d38dc4eff97671b6c70874652 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d567ffe701a14c30a7e210e335ccab5c : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1d01c14222c4a31beea71997df7bd2a : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1de28633a3e4269885a216fdbccfc00 : SUCCESS in 1h 16m 31s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c031e795c4442ae971adcae1f4f855c : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ad524f7dfdf4884a6a7baeeb2c7e89c : SUCCESS in 54m 48s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/472ec065e52d46e8988891272f211690 : SUCCESS in 1h 52m 01s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddbcaf06c4b74a429a49d31041f79348 : SUCCESS in 1h 21m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1506f51c0ad7422ab604084fe658ea48 : SUCCESS in 1h 19m 06s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1e5019042d04a2497d4c426876595ac : SUCCESS in 1h 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d32f972b32e23cf3021b6616ed29a2c86903dcc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 11:45:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"48dfcc783ead76951d48aa3962274869c0732e58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 11:46:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5522d19c7b4c45489c1d34fc9de21a55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/788a0fbeb338418e9c1a7475222af0eb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5129ebbc5fd6e579f0405f4dcb2a2bd1af555743","revisions":{"5129ebbc5fd6e579f0405f4dcb2a2bd1af555743":{"kind":"REWORK","_number":1,"created":"2025-05-02 16:03:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/948711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/11/948711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/948711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/948711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/948711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/948711/1"}}},"commit":{"parents":[{"commit":"4fc44596845d8ff47a1bfca86469fc0580e15b1b","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4fc44596845d8ff47a1bfca86469fc0580e15b1b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-02 16:03:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-02 16:03:51.000000000","tz":120},"subject":"Bump MariaDB to 11.4.5","message":"Bump MariaDB to 11.4.5\n\nChange-Id: Ic8ae457f734dd9c73ef40d59c9c5d441af552d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5129ebbc5fd6e579f0405f4dcb2a2bd1af555743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5129ebbc5fd6e579f0405f4dcb2a2bd1af555743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
