)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~949023","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2024.2~Ic8ae457f734dd9c73ef40d59c9c5d441af552d47","project":"openstack/openstack-ansible-galera_server","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-08 09:56:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8ae457f734dd9c73ef40d59c9c5d441af552d47","subject":"Bump MariaDB to 11.4.5","status":"MERGED","created":"2025-05-07 12:12:00.000000000","updated":"2025-05-08 09:56:49.000000000","submitted":"2025-05-08 09:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949023","cherry_pick_of_change":948711,"cherry_pick_of_patch_set":1,"meta_rev_id":"e7b4ba73ba66b4a160b2c9b77c1a62906fa77436","_number":949023,"virtual_id_number":949023,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 09:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-05-07 12:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-08 08:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-08 08:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 12:36:35.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-05-07 14:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 21:35:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2025-05-08 08:11:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a21c21a32d90ffda38b40c47a77c194244efe01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-07 12:12:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aca04113019b002596568b93271df2bc297dcf37","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-07 12:36:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"474d7c97537ce800a7ea7703f97e964febd28178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fb6bb935abe4b5180d20dda910662cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da87bfde564447e6a3c7e3642009b9ee : SUCCESS in 3m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/01e74616c0ce4669937b9a2aa60c6079 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f32d8457f9d461bbc4d2dc1e9cc7392 : SUCCESS in 1h 12m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8ae440d25604283bb7fc5360e4c48b5 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef5da3230e3d4bfd85467b326e6b414b : SUCCESS in 1h 44m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2b9e978c8c44c45b758d1e0e9b700c5 : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b063ef63b1b4c21ac59c89938918b18 : SUCCESS in 1h 36m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c97d6b2100cb4e51bc9acd8f013b8180 : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/678c4e42cb2d4ba3a0d78e86c1b9116a : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/335806f0ad1c4dbf8f8fe2bd21371fea : SUCCESS in 56m 02s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b7b44286833465db906fe7cb3abf01c : SUCCESS in 1h 31m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42e69622601e4d229422c0e96401b08e : FAILURE in 1h 22m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d0527510f612d024ad83c2461422ef638aa00aba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-07 21:35:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60d28a80f33ee3d0aecefbd0b6bf08a0c1850f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 23:03:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef9280ca7ff4e2f96e909508a3a68d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f5b95205c94260a45af57cd2aae539 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/06499967acd1437db4bb1fb715e92938 : SUCCESS in 43m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdfe1dde8949444480fe132acf8a74a5 : SUCCESS in 55m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/014c411059244090984a0e0a89bde0fb : SUCCESS in 1h 04m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4277b675929749dea911bf2d03e0217c : SUCCESS in 42m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76f5ab4f07174c7598d38009e3df5868 : SUCCESS in 47m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64340ed8284a41c6ba9acf18937e95c9 : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/783a779b6f904e7086711a32dc798960 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/610ae57d39e5457c85e0df7660f61c2d : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7536623461a34edea90dfd9fee37259f : SUCCESS in 37m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf50981274204e6eb7cb4a8570d20fc8 : SUCCESS in 1h 07m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c118ec7d58f41e1bb86afd974f6ee94 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ae673960ea12ecb6df0398e821016f401cefc4d9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-08 08:11:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e6acd50b12db096bdce96e4a87188d72f376999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:12:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3b4681682bb17dfb6bf63b3264b520416fb24e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3376ed474b1048c190d4c42563b3af0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5476688107420faa5ea36f5bfd2072 : SUCCESS in 5m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/02d2d847399943469940e244673a6dc9 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71617c9ad4ee466d89f4bda3a41a5f74 : SUCCESS in 37m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7299cc2c5a72404cad4764fa69e84acb : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea7b7191a72a4a45a241613107e08d79 : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de61bf00dd15441aade1096f4708ec0b : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85417d35cc3647d6913cd4d135e1ecec : SUCCESS in 1h 18m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72751af2399149c78790572f3500b089 : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8bdf283ef61a4358b4f4efb8c3733085 : SUCCESS in 1h 08m 20s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4c2324dfa7214ce08f0aba62f354bab2 : SUCCESS in 59m 44s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e136709253cf4cda9ef7b55304c6b52a : SUCCESS in 45m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17b87d99c0394f2d873ea69527ccfe24 : SUCCESS in 1h 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"877290555b5998c0c4eb1a0ce52c5cd4a6767cf3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:56:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e7b4ba73ba66b4a160b2c9b77c1a62906fa77436","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:56:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03da90e28de840078dcf0091f1a5935a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63343640c7ac4664b6a97988be0459a9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"520f548e03b72c6668468aa539e0e57c79a0a952","revisions":{"520f548e03b72c6668468aa539e0e57c79a0a952":{"kind":"REWORK","_number":1,"created":"2025-05-07 12:12:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/23/949023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/23/949023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/23/949023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/23/949023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/23/949023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/23/949023/1"}}},"commit":{"parents":[{"commit":"2f2912a30812c0fa46a380c0a57ac09989e8758e","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2f2912a30812c0fa46a380c0a57ac09989e8758e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-02 16:03:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-07 12:12:00.000000000","tz":0},"subject":"Bump MariaDB to 11.4.5","message":"Bump MariaDB to 11.4.5\n\nChange-Id: Ic8ae457f734dd9c73ef40d59c9c5d441af552d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/520f548e03b72c6668468aa539e0e57c79a0a952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/520f548e03b72c6668468aa539e0e57c79a0a952"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
