)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~949424","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-21 13:00:07.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-21 10:12:10.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-20 16:28:50.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2025-05-11 20:33:36.000000000","updated":"2026-03-21 13:01:04.000000000","submitted":"2026-03-21 13:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949424-osa_hashi_vault","meta_rev_id":"6c08a4a5d190d9e94e603f3321a59b6fdae0f3e4","_number":949424,"virtual_id_number":949424,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-21 13:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-21 10:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-20 16:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-21 10:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-11 22:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 21:47:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-12 08:58:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9ec2bbe3c1145a1ab8cb67bc876a7c4eb3e18629","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-11 20:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4917538d62118a8d9b1867056c4d2a3be97389b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 22:06:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78fb5af69dce441da7b4435bf1cbeeaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e449f09fec1644fb9c41dc717f9dec2a : SUCCESS in 5m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ffc52b24d1e4d44b88a8f35b9326c43 : SUCCESS in 6m 22s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc958e52273428bb80b0904430af7f1 : SUCCESS in 10m 09s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/60692f0d25cc49adb0d10c16c958e4f8 : SUCCESS in 5m 48s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d83f7a097d35465c80ac61bbd7b13719 : SUCCESS in 6m 56s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69603ca05bce4f8baa2f782f3e60d03f : SUCCESS in 12m 30s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0afcddab5e644188f3a00d70d74cca3 : SUCCESS in 10m 07s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3cbe02a14204ebe87e578f8fabaac84 : FAILURE in 38m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be17f11c16724a62b8857f769f8f374d : FAILURE in 36m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eca817c883b24b898567b265617cd3ba : FAILURE in 43m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e67ec3d596e43c3b169a740d0c78def : FAILURE in 30m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56be5882b25a48ef877b241377dac8d6 : FAILURE in 29m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f4e2f43b5da4948acac15076ae22711 : FAILURE in 31m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77f6d7ab9193408ea7ff165a6df392fd : FAILURE in 30m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fd18701a4a14de8af7a06ccb11f83ba : FAILURE in 33m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75273341f5ce431a959d9642c945bd18 : FAILURE in 31m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c112cb50244843469095d7afc89842cd : FAILURE in 31m 39s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2dd00b44a9846bbb53d824961253af7 : SUCCESS in 55m 42s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cbd33498766e41bcb6ff3164dfc91778 : FAILURE in 31m 17s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62007769a0b24a119a3c5749b0868281 : FAILURE in 32m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1fd19ed838fd4d6eb9a85a8b9953c8e2 : FAILURE in 1h 09m 56s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/66771a6b9e594ba18ee7c622fc514d83 : FAILURE in 1h 23m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37bdb7bed97f4068b8e3a40e83cb439b : FAILURE in 1h 00m 07s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82fbf91fd2d84220b78be7e22847c099 : FAILURE in 1h 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ba4706cc37af00252b26deac854bbdbfef8088f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-12 16:20:55.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"37fc4341f09189f9aaffd243669b16887b2bc1d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:23:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221171a8d04448a295db66aa53ffe105\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b7d7272d744cdab9f1766c2addc85d : SUCCESS in 7m 00s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e153d5c4f7da4cba9d613da202c35446 : SUCCESS in 6m 20s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0fdf768c818419eb5277a056dc42399 : SUCCESS in 7m 44s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a6d9b8b7b20e4291b280e7bb3a3fac84 : SUCCESS in 7m 59s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bade8a3c83dd4afb9543be4421967304 : SUCCESS in 9m 55s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/392346d0beb94ea683871fd1c3a77dea : SUCCESS in 10m 27s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bb597389b454916a1c7a6c238c6f95a : SUCCESS in 11m 30s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d054d8d230534a109eb5d5017dcb621d : SUCCESS in 1h 23m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a86b204b8904f649d12a6bf9604ee2c : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0507259442b4ff08068e0d1a41fd493 : SUCCESS in 44m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d260eb460db442c5801e6153815d865f : SUCCESS in 1h 13m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0aa524038e14f8d85218e77396b7ff9 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c51954c3614e62b2c430d8efdc4b7a : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bcf459f5fb44204a3cd8be8a2e086d0 : SUCCESS in 1h 08m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d6afee14b304dd6bf16f58c5dbdd968 : SUCCESS in 46m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afcf6513c1384f8fb2373b63603d5796 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8babbab5738146cda5e6a8d882ad6f21 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddb0c89288834d6b8925947c459d9865 : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/475c92216f0e40a59e934d7c04b2f350 : SUCCESS in 1h 36m 48s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a72acea511e469cbbefa9a1231fdc47 : SUCCESS in 1h 13m 24s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2b500985e9e40b4b84743fc5f672eea : SUCCESS in 1h 57m 35s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/185d7fb370ee46ceba4595ffb14b5883 : SUCCESS in 2h 38m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f798318a86a4cf99e898618b9a4f559 : SUCCESS in 1h 41m 54s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5bc2ddadf5840f3bb1b9f9c45031cde : SUCCESS in 1h 39m 25s","accounts_in_message":[],"_revision_number":2},{"id":"acfac2626f7a01b692f84dce4c375bbe1cbdcd7f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-23 10:31:41.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"8327c78649ef534f329f2e4e29e18e22e9f3576a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-07 19:44:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9268dc226fa324cd89f8294b5a1ef1859204c815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:44:30.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/ansible-role-pki for 948881,24","accounts_in_message":[],"_revision_number":3},{"id":"12138c0fa3a882e18924c3daa32e51ecf2961665","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-09 13:54:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"498ebc7ea1c7ee8f969dd4d0ee6cd17a6e440976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 14:01:37.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/ansible-role-pki/+/948881 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"086653ca8b3d03b000bc7541b0d2b9450205b3e4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-09 19:27:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"45df02883c3effb0ea88e94a343c01158af2c541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 21:43:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae54255eda645169be8cf6a7133fc81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5e0a568eba4063a1c4b3ffa10f76c1 : SUCCESS in 6m 24s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bde15a02d16c45b3ae099ebbc7151f70 : SUCCESS in 6m 34s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2913fd669de543359d17f4dbf3443e96 : SUCCESS in 11m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2cceffa46d344fd8ae9e1cc6a3acef9 : SUCCESS in 10m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d4e2b466f48d4c4790c6296c1767af65 : SUCCESS in 8m 48s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/754906c7e6a94309a45d6dc36ae137bc : SUCCESS in 9m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3f75b79493d24af5ab82aac7ba74592b : SUCCESS in 1h 03m 42s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e227f32cb18841cdb3afc840b3420f32 : SUCCESS in 41m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50e2452a5a224bbda68932c71ef0734f : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b06e23dab35469f84c4bb9985174641 : SUCCESS in 1h 03m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb7c7b0a1022498094e0ceb8575da5ce : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5114c4d58f9145538a3bc2ddd89ed753 : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f9d394fa54e456894f58bc9cbcf0fd5 : SUCCESS in 41m 56s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/801a526df4014aa39579dc86ce427411 : FAILURE in 6m 07s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ecb771f651144e884f48643719a230b : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/182b9021f277489590d5bf12a005abbc : FAILURE in 11m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69835423ba3e457ca6e8087ab5bd4f0a : SUCCESS in 1h 15m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/383aa5512f2b4224a861586d93dee086 : SUCCESS in 1h 53m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a7ba37e88e4ee2d0c8f76a926339f2e042bb064e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-10 17:54:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":4},{"id":"e926d86ee10757557cef35663fa99739d2843456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 19:47:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c0c1071da94860a010c6422f736abb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686b1a164b064d39adc2fca1278e72d7 : SUCCESS in 6m 16s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87394827192d4f2aa64f94c08c7e2967 : SUCCESS in 14m 02s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a020e9970954ba0b96333eceaf03c7a : SUCCESS in 8m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/48913475b3a342c391eaa1d03ef7ab3d : SUCCESS in 12m 11s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a09c429f5534e83bac989c640ab9399 : SUCCESS in 13m 20s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb02eaac61b041ba849b308c7d6349ea : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8414fe7f7d9f4020b48f50854e0992fa : SUCCESS in 9m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/40c323bfb69b489a9d97c62a7cee4698 : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d76dec1b960456eb3b7ef112390abe1 : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/70353038cd4341f9bad8f8d1da12e1a5 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/83758fb41ae04691b23c824c47781360 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b30db685e258496fa6b5d89dace64436 : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9195f9b0322546c69c2286c14a27c252 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b94cba65af8e41eeae1a0422907e6023 : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c9da1b21e554f29a66ffc4577fb6c93 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/fea3fa6cdda44e239db0b085e482b410 : FAILURE in 8m 11s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8cc77249195a43dd948887954377d8a5 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5f70f06e546d41c481ec099ea90f8e24 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/69fb81ec54834325860a36de804b0cb6 : FAILURE in 13m 51s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f736838381cd42d5a5240bac44b49dde : SUCCESS in 1h 15m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca139c3371e242b58d029a8de0704454 : SUCCESS in 1h 47m 00s","accounts_in_message":[],"_revision_number":4},{"id":"539e9a0420eb541c86c3425b81ac323f9a55de61","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-10 21:47:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2058d88e6bcb05fb5877a72f9863d858eb005e54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 08:58:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d30c323b1ca31565c58b215ae086f14da6420c68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:00:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d1e69effe8dc8de476ec98e033baf1887c34668e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 09:02:28.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"4b33eb88b8d99a02ea29bfc4453b4d8c57f5f5e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:02:32.000000000","message":"Patch Set 4: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":4},{"id":"e0234cb07c0ca881028978e331e015416c86c40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-13 11:29:58.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":5},{"id":"ba8c0c6d45db175426cb63550ded7df5b3834965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 13:39:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd413a3892ae4043af0ff994154eeb46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee6399cd8f14c4b9c5a6ed771600324 : SUCCESS in 7m 15s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/788cb108511646b8990054e6d5863d9f : SUCCESS in 7m 45s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a2ff27245cf641b3bd2aeaf1f5be4442 : SUCCESS in 10m 55s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/4e6718314d8b420b8d7a03bde00e9be2 : SUCCESS in 13m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0005ac27fc843cdad4285895f62fd6b : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/0ded2ee0b28c4625a050faa324f37eda : SUCCESS in 9m 55s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/580d5f5925db4d74b3971ad7e8721cfe : FAILURE in 4m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4af62b5c09af446babbb5fb97dcb7776 : FAILURE in 1h 00m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dedfc17dc2854f9b8b95424939cb1a7a : FAILURE in 1h 07m 07s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/20316c4e93e34b14a8f6c03fb01bbee4 : FAILURE in 1h 11m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2daafff86e5e44878ff4a1cd9a43a833 : FAILURE in 1h 20m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bd94c7e3cac549d9938702ec8f73f837 : FAILURE in 53m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58ab564728454764bf197dec6f528e73 : FAILURE in 48m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a84a35a06ba47d0a024cc34f74f8ae7 : FAILURE in 1h 32m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4069f07b4ae04e94bad30be16a8e6853 : FAILURE in 50m 19s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6431cdd0ba514ccea6d269b980fb6bc3 : FAILURE in 9m 39s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e521fb9c8ee14f818268e95f6ec7dfad : FAILURE in 1h 06m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0fb756fe01d940eb9017591f0820861c : FAILURE in 1h 07m 18s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/72426c538406439e86123cc5735127fe : FAILURE in 13m 22s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/453ab53078664fa79b5af6ad46e46ad4 : FAILURE in 43m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46d88837451a41ff9aca3bf98e0a253c : FAILURE in 1h 59m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5239e49bd00941aac0a6121e52ef67e512aaed71","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-17 17:18:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"990085eb26e70d193acdc7188f7ae4d2dffb0f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 19:46:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae049651374248b5ace0f0acaae2c695\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c63e9a8017441469f9d4361c5dcedb6 : SUCCESS in 8m 04s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3beef51badea4b149dafebd75316a2fc : SUCCESS in 13m 00s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/48e72716c4a84de6a1e356221fa53911 : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/738c9b8d4c5a4d59bfbfbfc372a60e12 : SUCCESS in 7m 48s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7228f85af1484bb096fab4c9ad2795d5 : SUCCESS in 7m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/31281c647db64244949c435aca1ec300 : SUCCESS in 9m 44s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/08a08dfd3cb54e1d80660ecae54bcb8b : FAILURE in 5m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a43db2ebcb1449b0b7841cac4b6f2452 : SUCCESS in 56m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d13618d0499425497ac382ddbed94b4 : SUCCESS in 51m 34s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/52893a6d95c94a8780612d7371e265ab : SUCCESS in 46m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/67d064f0284041b986fa5e0c2b36973e : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/463f60b0b9cf4a3fb8a6644903897366 : SUCCESS in 52m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/277767a78a704ab1a39cbeda86886396 : SUCCESS in 1h 53m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a8329ecd3dc436a866ea5bba5acc8fe : SUCCESS in 53m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/648ea49abe764bbbbd650760c06bc948 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/12b4cd1a531e471dab733c94d0185eaa : FAILURE in 10m 21s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d71d442021e49cc883f53816782aeb9 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6275ded2af93448cb8f1cdbeacc1c85b : SUCCESS in 1h 04m 31s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d04cff895ba448b88a97ced37d5abddf : FAILURE in 10m 28s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e597e2ad74f4ca6858a4e14e3b23d4e : SUCCESS in 1h 36m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29c58afd9d3840c5a68ce8d9dbbdda6e : SUCCESS in 2h 04m 42s","accounts_in_message":[],"_revision_number":5},{"id":"fa72a3a5b9e89f5d744611045b295095ff5914c7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-20 16:28:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f7f95f08a2dd3e3a7b3d8e155c397eef0fd90277","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-21 10:12:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"45982b45125e0475705d338070a1e673dee0ea1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 10:14:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3a41934d9e2687f3bfae4b72108f5f0343141ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 13:00:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1531d25cc7a9465389c19e7cc24e95cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d841907e55fb4db2a1b2ee58c3e4923e : SUCCESS in 5m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c34c608f3d2c4193af9f9b2704b0992f : SUCCESS in 7m 58s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/75d54787b67f451f8371f536edd65d1a : SUCCESS in 6m 59s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f048000f84bf467191f6d01823d028b8 : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d18e3c77059494980ac3370f71d6517 : SUCCESS in 13m 36s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc0db212156b4e5aaeeb0508ec114d4d : SUCCESS in 9m 53s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e6eb1a6fc0d4d55a32ef80b496180b8 : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a665a00a2a304f1599e438e746e11c5f : SUCCESS in 1h 19m 01s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c282340a97d74673a1653746e6525217 : SUCCESS in 45m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60e87856177a4f919b7b6c2911063f36 : SUCCESS in 49m 32s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cf4b6a071c8e497f97b79f236519c357 : SUCCESS in 49m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bec232474d564d4193fb950870ae5b5f : SUCCESS in 45m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/620392c64ef54faf9db7aa68ee9befb9 : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25c8688139d94addabaa6dd987257b33 : SUCCESS in 2h 38m 37s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/218e1038b2bd42ae92dd4fe20f46ebd8 : SUCCESS in 56m 26s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e9d1b2c9f8a24346a469b31647c2f153 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/919a8505647b460e90024097691be9ad : SUCCESS in 1h 15m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d20c4b2c264c538b4f8a9e968ed8d0 : SUCCESS in 1h 57m 00s","accounts_in_message":[],"_revision_number":5},{"id":"31b47b1259d376fc5a7d69337825e3e7bc0d8c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 13:00:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6c08a4a5d190d9e94e603f3321a59b6fdae0f3e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 13:01:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b14f024937d343e1922b924085688d8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0583e0f17e76425e927160cf18fe03f0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"00256b6cc14738b236610b1203c259bbf753eead","revisions":{"b730e0c015f3e2173019a193e5e2c35aac9eaa65":{"kind":"REWORK","_number":1,"created":"2025-05-11 20:33:36.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/24/949424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/949424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/1"}}},"commit":{"parents":[{"commit":"099b4a2d2cdac8d5f390e98234735b08b94104dc","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/099b4a2d2cdac8d5f390e98234735b08b94104dc"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-07 17:17:50.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-11 20:33:34.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nChange-Id: I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b730e0c015f3e2173019a193e5e2c35aac9eaa65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b730e0c015f3e2173019a193e5e2c35aac9eaa65"}]},"branch":"refs/heads/master"},"c174b848c29312bcd533df595e027757da87ff34":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-12 16:20:55.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/24/949424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/949424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/2"}}},"commit":{"parents":[{"commit":"099b4a2d2cdac8d5f390e98234735b08b94104dc","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/099b4a2d2cdac8d5f390e98234735b08b94104dc"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-07 17:17:50.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-12 16:20:55.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c174b848c29312bcd533df595e027757da87ff34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c174b848c29312bcd533df595e027757da87ff34"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1410a26147c4babda071b93fd9ba979310029a8c":{"kind":"REWORK","_number":3,"created":"2025-10-07 19:44:10.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/24/949424/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/949424/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/3"}}},"commit":{"parents":[{"commit":"099b4a2d2cdac8d5f390e98234735b08b94104dc","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/099b4a2d2cdac8d5f390e98234735b08b94104dc"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-07 17:17:50.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-10-07 19:43:49.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1410a26147c4babda071b93fd9ba979310029a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1410a26147c4babda071b93fd9ba979310029a8c"}]},"branch":"refs/heads/master"},"a1ba4bed19892ab8be656768689b5bd9e5350b14":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-10 17:54:07.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/949424/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/949424/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/4"}}},"commit":{"parents":[{"commit":"db6536b01c0e1d857ff7124916cba75f452bd69b","subject":"Merge \"Upgrade MariaDB to 11.8.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/db6536b01c0e1d857ff7124916cba75f452bd69b"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-07 17:17:50.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-10 17:54:07.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a1ba4bed19892ab8be656768689b5bd9e5350b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a1ba4bed19892ab8be656768689b5bd9e5350b14"}]},"branch":"refs/heads/master","description":"Rebase"},"00256b6cc14738b236610b1203c259bbf753eead":{"kind":"REWORK","_number":5,"created":"2026-03-13 11:29:58.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/24/949424/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/949424/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/949424/5"}}},"commit":{"parents":[{"commit":"db6536b01c0e1d857ff7124916cba75f452bd69b","subject":"Merge \"Upgrade MariaDB to 11.8.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/db6536b01c0e1d857ff7124916cba75f452bd69b"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-07 17:17:50.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-13 11:29:58.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/948881\nChange-Id: I73a5b0beb6083ef8d4d6cc3039f231964e15d6ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00256b6cc14738b236610b1203c259bbf753eead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00256b6cc14738b236610b1203c259bbf753eead"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
