)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~951315","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia094e35c35a9f02d777ca443653bf308c53a31bf","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/el10","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-04 20:48:49.000000000","reason":"Change was submitted"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2025-06-04 20:48:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia094e35c35a9f02d777ca443653bf308c53a31bf","subject":"Add support for EL10","status":"MERGED","created":"2025-05-29 15:44:08.000000000","updated":"2025-06-04 20:49:38.000000000","submitted":"2025-06-04 20:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951315-osa/el10","meta_rev_id":"73d488da4db0b3bb8fae3434ff7b0098c9278b71","_number":951315,"virtual_id_number":951315,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-04 20:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2025-06-03 10:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-04 18:31:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-06-03 10:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-06-03 23:02:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 17:28:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 13:46:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-06-02 14:24:00.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2025-06-04 18:31:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"264ee6faaaa4c4eb212252d7eef851d9bdc0ef7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 15:44:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a46b97478533a4d3057b9881c376b303dc243ca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 16:07:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c55b0567da5cff223132aa716e0f8a219ee47338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 17:28:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0de0f0cddf2f4212ade2b1cfcf0935f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd238e4a92e049bea7fcb882f209d1d0 : SUCCESS in 5m 33s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cb75317fc7144479ccb49d6be76bc48 : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/508698bc699e4c13ae83753c6e838eda : SUCCESS in 9m 07s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7dacb91aa96c41f6a82f0fb457c0fa76 : SUCCESS in 9m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca730bda3f1140ffa545c75eb121e6a2 : FAILURE in 5m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9f1754faba7b4e709cf9940f95b17a2a : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d019fc00eee48698c7fb17b417391ee : FAILURE in 23m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/351ede6ae3054cc3ac82f5bb9b51620e : FAILURE in 14m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d523d8e94d84afea62944d42fc60b30 : FAILURE in 13m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e6d50086c0847f58848ca0eaa1a3901 : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ba320d243784dc5b02f53875a3da28c : FAILURE in 20m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/407a6e65ba8f4749bee8c32042141d52 : FAILURE in 20m 21s","accounts_in_message":[],"_revision_number":2},{"id":"532cf2367c33813f21c96c9de06cf6bbc609b108","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 08:32:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7835d01c4623f9c573c0b4a7fa60e4ccaa4389d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 09:36:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4961517ba0139e737840bfbe2ec46ac7e707f9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:55:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e42bed93d7ee460d923c14bdb987872e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d07d29cc22420f95b6d2ea1c1be519 : SUCCESS in 5m 58s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b887ace2755415685fd283198c8f476 : SUCCESS in 6m 32s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/9266621634d943c183c6042873e0c4b0 : SUCCESS in 9m 14s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a47269ca9624a3eb3370324f5446749 : SUCCESS in 12m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ae86a4df514b45aaf539c2b8a22e09 : SUCCESS in 12m 58s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/01cbf22f63a14439bdcd620451461f7e : SUCCESS in 8m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6aa964a12303459db3d0835cfb1f0242 : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5e103d036f64ee9a032cde89d819150 : SUCCESS in 1h 27m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1312f4aed0eb4747b661c4440df8f47d : SUCCESS in 1h 37m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59e9a95a895c4636b57877bf3e4ec51f : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/32df8c0f4f87476babe89cd2fcc158df : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ca8a61b266a43d3b85416d66d3c634c : SUCCESS in 1h 14m 09s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d392a4cbaaa7481ebe4db3bd1112dcc0 : SUCCESS in 1h 35m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8043daa108dfd978790beaf0e0e5bad22b869d2c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-02 13:46:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6a9f6baac08afe9dac25c0b35ac947788799bfa0","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-06-02 14:24:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"10a2159425816afde72eaa0f89ce05afaa433216","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-06-02 15:10:05.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0000a14c4f599d0db2d9127fbdaba32fe4e151e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:10:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f5bfcbeed51484e83440ebe1a3d782b\n\n\nWarning:\n  Change 951132 in project openstack/openstack-ansible does not share a change queue with 951315 in project openstack/openstack-ansible-galera_server","accounts_in_message":[],"_revision_number":3},{"id":"f693b8cc9f77d77decb6c630cbdda5cc48f8606b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 18:12:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1892e1de7c0c8cc3d546ebc8737dcca7deb002f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:22:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a433976f95d742cd909a937308d7d642\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf62e750ceae4a58ac986f84432bc2c6 : SUCCESS in 6m 48s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8c566d72c414369b3cc467f10ca9df6 : SUCCESS in 5m 32s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8761e1b3049d4240a8599bb761e31282 : SUCCESS in 12m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71927dd1032841749606e4072b8c1a94 : SUCCESS in 13m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/eac49016c4d3462bab92bae3cb6153e7 : SUCCESS in 9m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9002a6a3590948a3b9b1fdbf07d9da06 : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2c82330661a48a3a77f8465da462699 : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2daa507019447c0990bb2fc6609dcc8 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf2de5366b3d4409b56f23903f6e37b3 : SUCCESS in 1h 48m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d11d3e3746a54be8bb4a9089059b87ae : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0474b07a9d346b5b4a81234d9846466 : SUCCESS in 1h 05m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0790f12b5a5d4f2a8fc4b1f280c141ef : SUCCESS in 2h 05m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ff0d7359eb6e9f67eed00aba1717d2d2890a3478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 22:10:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f271b9bcb121c99b91355f69d953261f71b514e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 23:55:42.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951317 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"23aa178b75239c98a07af462f19576c3a2d54fbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 07:42:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"df4e3ec520489ac8280770f3452615e780869572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 10:07:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d510b3d9df4436b24f80fa4400a13d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d1376175be463f8eecb4b52e1a863c : SUCCESS in 5m 34s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f29b1ef148d43c18b3474d5adf3edd6 : SUCCESS in 7m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6cac90032691438792bc9bf92326233d : SUCCESS in 13m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a613686c958403393c8b4e1465704ec : SUCCESS in 12m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/797b20ccf992400a876870049e0f2f6e : SUCCESS in 8m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ec03c0fea77344d1b5b43c8997e9a51d : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b3a00ba266440ee9c8972d5c19f3f1b : SUCCESS in 1h 39m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9878603b47fd483a8bd51066855a93b0 : SUCCESS in 2h 17m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66c04ac33f944dfa92292aabdabb4bf6 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/966c11f6ae4d4b628d13b96109392a61 : SUCCESS in 1h 00m 58s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ea93ee488ac4cd8aedd807c90ffe20c : SUCCESS in 2h 21m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc9020e4920d4ecf8229d5aa75be8828 : SUCCESS in 2h 16m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2017c3373cad1b3e5ae62a46fe6a6d6673806d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 10:46:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"48dc9ea5399ff6c28f1fbe155f2a28ccaaac4fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:02:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bb13f3785f54792a8066a747975e5c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711ddc2eb7174f4d83a320118bc65cd9 : SUCCESS in 6m 04s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0406564a06564d8c81be92808ba30d00 : SUCCESS in 7m 23s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/1c3118d1a8db43549830d870786048f8 : SUCCESS in 12m 43s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf61e1dc4020445db19005836df2facc : SUCCESS in 10m 20s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/247ee64e7f734fc391ccdc261f00b015 : SUCCESS in 8m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/62418c6dff7f479286c552e339d67b2e : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7b2884bf7444743962e9d7f74ceccc7 : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/158f1fbb8f004804b812dc66a9911e63 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dcde5f0f801449799f810e42c0cd3f9 : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/269c175c1e8d42a3913852dd74c53cb1 : FAILURE in 1h 11m 35s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb4ffea4626f4dd89a22e058eecf46ce : SUCCESS in 1h 13m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d1421e337704c9ab956350c21c114cd : SUCCESS in 1h 42m 15s","accounts_in_message":[],"_revision_number":5},{"id":"e1c4cf87ae45d52b902c06f6943a3dfd681ce8dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 14:15:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2986ed75c160afa1ff2b9f2a73752b9b80b352e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:46:15.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d56491700b04dd2aee01794457b19f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8149f159cc4ee9a05198b245c48296 : SUCCESS in 6m 49s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4fe3d554d644252826811d5d1b5df4d : SUCCESS in 7m 32s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e6ed6afe1c7441579372eb8aaae0dc48 : SUCCESS in 10m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1a241a02e954653a790317ff1bf23f9 : SUCCESS in 11m 32s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/7518d45dd8d24943a751ca4b293c2c1f : SUCCESS in 9m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3fc86640d99a4004847194ee72a34bb7 : SUCCESS in 1h 09m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e37c77c981a4573b827326564a0e78f : SUCCESS in 1h 10m 35s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e4e974271c147889e6fcf24f8dcac07 : FAILURE in 2h 24m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19b8eae04c3447f18ef89b0e41189c74 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/802a10f313f34d50bbc2eca9c7682110 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10edf79f53cc4a4d8730813f9b5c959b : SUCCESS in 1h 12m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3a7d654d55d4cdaaabb8819e5d9c311 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":5},{"id":"f47dde88327e6ff743190896c84c2e7271e68a05","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 18:07:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a00b522cf8ca6f1a803b199488af67f50db03e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 20:02:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec8207c5b734bb787ee5d15122764fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ba7fb661e34324bc1d8716f1349eb6 : SUCCESS in 5m 26s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75b1a9822cf543be9ee4ccf3847774a2 : SUCCESS in 5m 41s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ede9819e4454673995007294a3f83d5 : SUCCESS in 13m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10702a60603d4325b3d4c00a14609e30 : SUCCESS in 12m 21s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5f55e69a58cb4db79052e1b85ce31e70 : SUCCESS in 8m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9158d429424477fab20f68dc1587fd4 : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64231d74737b4810933209e40f9b3bef : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89df29b485944140868311ba9e8b2f0b : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e67fea14c0ca4fae9545177c025b679d : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/36de5f60b7c64dfaac1a8c3168c7839f : SUCCESS in 1h 10m 03s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88211fa8c978476598aa72058220fcf4 : SUCCESS in 56m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78a357327d584ae3b04cb1b4f5f011b9 : SUCCESS in 1h 49m 01s","accounts_in_message":[],"_revision_number":5},{"id":"bf80d0083ebe0333c9e2eb67a3bb25fb73d87204","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 23:02:59.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8298df229ddc2c02c3e97306aae24c5a4f00ad24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 23:05:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"aff141921647cebb90fb37dbcae639db2c459c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 00:33:22.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/951317 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"d564d00029bb47d23f5d1718c294dc6db829e9f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-04 12:46:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c348cf7250988a30edc1c2c53cc3d821e8ad1b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 14:28:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e67508f39b874702983e56130cc33013\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e227feb14354d9db88acde8784e1982 : SUCCESS in 5m 27s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4579223f773b40b691ea6acdfb85a71a : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2eb5829e6e4a44a0a1545e6a777129ed : SUCCESS in 10m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18413008c5244119826be255928a60f7 : SUCCESS in 12m 33s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae18b1190ec7443ca331eded359750cf : SUCCESS in 9m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/81e0999d54134531b2672be82e7c75a2 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/216eb4d7c6d7441a896f86f57ec092e1 : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d16dd04e913481e8f32ca92e4853f13 : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e11138bc0c58484dbd3867dbb25f96f5 : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a69057516fe74ecdb6eeb159d19e2eb1 : FAILURE in 1h 13m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7828c89c4b6a4bcfb13478771eacd279 : SUCCESS in 1h 08m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e24c6a6d91664752899876e16a817fb5 : SUCCESS in 1h 33m 42s","accounts_in_message":[],"_revision_number":5},{"id":"555e0ec46cd609213afc48b5d6630fdbb0f3997a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-04 17:31:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"84152065fe163bdc25f8a6b894d8faee60adbf8e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-04 18:31:39.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"801edb62dd5909b1764f42222f1a64313d663d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 19:03:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d409c9bc9313431383214a98b876c7ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a391e1f02945ec83beef3d5d043e2c : SUCCESS in 5m 26s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ab04785bda54f96b11723be38b8dd25 : SUCCESS in 6m 30s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/db472a42baad4139bdd9f3bfffb9df37 : SUCCESS in 11m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f58d85f7e262496e961fe718acfd79e4 : SUCCESS in 10m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/b72c67f0f50347148ee30066c6e6a895 : SUCCESS in 9m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba1a4426d2204191a386435d4ea51702 : SUCCESS in 55m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/588a0a5ad823411f80ad06eae3f9df0e : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f2eaa2352454fe580b126beb555de75 : SUCCESS in 40m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b9012a9c4054d478fe8119bcb781db2 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac987227a0a04553a6189e0e8e6c4536 : SUCCESS in 58m 01s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77dbef4d63224e3f82736a8fd823f0fb : SUCCESS in 1h 10m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b334940c239f4ee5991fda1a548db3f2 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":5},{"id":"5b028a011ebeb84a13c2796fef039bf0ab03f5af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 19:05:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dd0a8af4626bcc22abb1a28ff13b1e294eaf0fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:48:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6218ec56b314ce9bdb98b087a04ef23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8cf37746394a9a8aa2c1afe1c3ba7f : SUCCESS in 3m 35s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e01efbadeb44d5988de13d0d0564126 : SUCCESS in 3m 49s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/91f587e0968e450ebd6f54992c0f7bc3 : SUCCESS in 10m 51s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5716e0fa10f8465fa73bc3920fc8a95a : SUCCESS in 10m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/9e457adf0a8941c980445268615156ed : SUCCESS in 8m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/db4719f7628144e491050b19afb7c251 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b3fd589fa4d4570b377426132d83309 : SUCCESS in 48m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8341af2880d6449495ad69d93863d328 : SUCCESS in 49m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d3226a6da74f98a531581869b606c7 : SUCCESS in 48m 40s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e46ddb1847804e48bae1e56aaf28cff2 : SUCCESS in 44m 15s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8530995f31fc4cde98a8d9cb51bfe662 : SUCCESS in 45m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d98b6adf7d1845f99bc91f2134ae53c4 : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":5},{"id":"6a742abc0108bab79978c8e157cf95a7c5e3f923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:48:49.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"73d488da4db0b3bb8fae3434ff7b0098c9278b71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 20:49:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e4a010fa36412b96033f860b1e9c34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469075ab2f2e47d79dd4ee06203e0a5c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a45b26d91f19148f99af94eb77123f81cd4476fe","revisions":{"9afdbabfb02fdb10f61112559fcfedb5e277e94b":{"kind":"REWORK","_number":1,"created":"2025-05-29 15:44:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/951315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/951315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/1"}}},"commit":{"parents":[{"commit":"5653f325603e41ee352cc29adeb28f9eb08b6799","subject":"Switch MariaDB to archive.mariadb.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5653f325603e41ee352cc29adeb28f9eb08b6799"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"subject":"Add support for EL10","message":"Add support for EL10\n\nWith that we do support only installation from distro packages,\nas mariadb community repo does not contain EL10 packages.\n\nWith that no specific vars were added to EL10, except\n`galera_install_method`, so once packages are there, we can\nsimply drop redhat-10 vars.\n\nChange-Id: Ia094e35c35a9f02d777ca443653bf308c53a31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9afdbabfb02fdb10f61112559fcfedb5e277e94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9afdbabfb02fdb10f61112559fcfedb5e277e94b"}]},"branch":"refs/heads/master"},"bacb6ea20ca36d4849cde2f56b202a2210c8851a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-29 16:07:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/951315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/951315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/2"}}},"commit":{"parents":[{"commit":"7805c58274c968bea09b59014a53a936c9ed7a75","subject":"Do not verify MariaDB version when installed from distro repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7805c58274c968bea09b59014a53a936c9ed7a75"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 16:07:10.000000000","tz":0},"subject":"Add support for EL10","message":"Add support for EL10\n\nWith that we do support only installation from distro packages,\nas mariadb community repo does not contain EL10 packages.\n\nWith that no specific vars were added to EL10, except\n`galera_install_method`, so once packages are there, we can\nsimply drop redhat-10 vars.\n\nChange-Id: Ia094e35c35a9f02d777ca443653bf308c53a31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bacb6ea20ca36d4849cde2f56b202a2210c8851a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bacb6ea20ca36d4849cde2f56b202a2210c8851a"}]},"branch":"refs/heads/master","description":"Rebase"},"23172b3db22141e3a0d062825523d7888dc959ec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-02 09:36:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/951315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/951315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/3"}}},"commit":{"parents":[{"commit":"7805c58274c968bea09b59014a53a936c9ed7a75","subject":"Do not verify MariaDB version when installed from distro repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7805c58274c968bea09b59014a53a936c9ed7a75"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-02 09:36:05.000000000","tz":0},"subject":"Add support for EL10","message":"Add support for EL10\n\nWith that we do support only installation from distro packages,\nas mariadb community repo does not contain EL10 packages.\n\nWith that no specific vars were added to EL10, except\n`galera_install_method`, so once packages are there, we can\nsimply drop redhat-10 vars.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/951132\nChange-Id: Ia094e35c35a9f02d777ca443653bf308c53a31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/23172b3db22141e3a0d062825523d7888dc959ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/23172b3db22141e3a0d062825523d7888dc959ec"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6ce10bed0a875041ba854f6de2247d013298c687":{"kind":"NO_CHANGE","_number":4,"created":"2025-06-03 07:42:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/951315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/951315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/4"}}},"commit":{"parents":[{"commit":"e796b5aca7314e112e25c12d56f37bc56a3e8001","subject":"Revert \"Temporary disable molecule jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e796b5aca7314e112e25c12d56f37bc56a3e8001"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-03 07:42:42.000000000","tz":0},"subject":"Add support for EL10","message":"Add support for EL10\n\nWith that we do support only installation from distro packages,\nas mariadb community repo does not contain EL10 packages.\n\nWith that no specific vars were added to EL10, except\n`galera_install_method`, so once packages are there, we can\nsimply drop redhat-10 vars.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/951132\nChange-Id: Ia094e35c35a9f02d777ca443653bf308c53a31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6ce10bed0a875041ba854f6de2247d013298c687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6ce10bed0a875041ba854f6de2247d013298c687"}]},"branch":"refs/heads/master","description":"Rebase"},"a45b26d91f19148f99af94eb77123f81cd4476fe":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-06-03 10:46:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/951315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/951315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/951315/5"}}},"commit":{"parents":[{"commit":"ce844bc4da2f2479374f17564c92c2fd60b4a4c7","subject":"Do not verify MariaDB version when installed from distro repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ce844bc4da2f2479374f17564c92c2fd60b4a4c7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 15:44:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-03 10:46:42.000000000","tz":0},"subject":"Add support for EL10","message":"Add support for EL10\n\nWith that we do support only installation from distro packages,\nas mariadb community repo does not contain EL10 packages.\n\nWith that no specific vars were added to EL10, except\n`galera_install_method`, so once packages are there, we can\nsimply drop redhat-10 vars.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/951132\nChange-Id: Ia094e35c35a9f02d777ca443653bf308c53a31bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
