)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~951614","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I942890cf89b9567067bc601f481bfb516ebb2237","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-04 12:51:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I942890cf89b9567067bc601f481bfb516ebb2237","subject":"Temporary disable molecule jobs","status":"MERGED","created":"2025-06-03 07:38:39.000000000","updated":"2025-06-04 12:52:49.000000000","submitted":"2025-06-04 12:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951614","meta_rev_id":"d3b4b19433f30a0ee99d5131ffa93d3a6ebc8496","_number":951614,"virtual_id_number":951614,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-04 12:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-04 09:04:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-06-03 18:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-06-03 23:02:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 10:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 18:08:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-06-04 09:04:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"85f72efb01371c7c457b4e93b48c243b94a255c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 07:38:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2897079cdf4bf14fb1dbfdbb3c4a422ebc6e8e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 10:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e3ed626523143f2b9861bf0a915731b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6885ea5bd98642c3bda0e2acf35a2360 : SUCCESS in 5m 20s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c26aabcf22e842a8bca2e5220b9b3609 : SUCCESS in 5m 41s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3ffc262a45aa4ea1b1ae7ba2d17988f7 : SUCCESS in 10m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39eab34dda36403f85e7df9d16e36c3c : SUCCESS in 12m 35s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4b758dfe89fb4c03a0a4e69281917adf : FAILURE in 4m 40s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e55ab0fb39f04ae8baf0459da7f536c2 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3175627360034139b6dcfee93c5b6aed : SUCCESS in 1h 47m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bf6c355d33c48abb24f3145aee2573a : SUCCESS in 2h 04m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/398e2212faa1447cb7e0a9a14c42f105 : SUCCESS in 1h 38m 54s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e77e3210b3f74aa981eea9f8a425d4f3 : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/985e956eb7d94b888fad55ee751e6866 : SUCCESS in 1h 39m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d7b9061327a499787d7069d80b4e8f0 : FAILURE in 2h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e7ae24c138632f972a9170fe14e11206132ff75b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 11:07:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a39699582d7525c2a8f38f50ead9835cddefed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:05:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6189e9fc81a4d2aac8a1ee15c3607b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d00be6794624491a41063239a13b7b4 : SUCCESS in 6m 50s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3cfbcfe97ec41dabacee21a29ec059e : SUCCESS in 6m 46s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c9bb873de374e8a8d8e06718c0c9f7a : SUCCESS in 11m 19s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6e3459feda418bbfb1eec94616700d : SUCCESS in 13m 06s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/154fe4a9e516421d852478dc7a677f15 : FAILURE in 5m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ddb0e86010144add8fee030c790bff9e : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db7c4299a9e844899eb34a9f03dc67ac : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35d14ec30f814d3d8281234b7215cfd1 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7bb87a925244db8a43946b1ec3c44c9 : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f06133f384c249a29c72bb578231dd3e : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/223505fca5bc4472a5006081d001783e : SUCCESS in 1h 04m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab2939706179455d84143fe27dd93ea2 : SUCCESS in 1h 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"862cdee41984531caa1e7a0752ed47304c698ad6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-03 18:08:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bde132fd9735262fdc104e3275c15712182fc142","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 23:02:45.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa2214ad4028342804c9299f3024b65a75e51c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 23:04:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d830f907a22192d8ad7c63744f156cbb6d349505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 00:33:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21f4a79970c84d9aaf62e715c7e92050\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b59d15a62214684a8239cdf67428b27 : SUCCESS in 5m 08s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1b0770212854f5cb854b5cedf3e1cce : SUCCESS in 5m 49s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/292373a62e614b109437921dc893e587 : SUCCESS in 11m 31s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b097e4ad1bd245a0abce91f518428b14 : SUCCESS in 9m 27s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/047300bd372d492eb2c2fba279b9d018 : FAILURE in 6m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/54577180456743c984c0b0e7ba577d8a : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e78449a8d0a486a862ab03845d882c6 : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9db7db46db84216881a36d5f22bb07d : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/337de5bcd83e40b88b57ad55eebb4538 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d5563e6e04a54a48932361b42ee47b13 : SUCCESS in 58m 42s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe311f44070d40e79a09ae5468bd32c5 : SUCCESS in 1h 11m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d457bfc7f9c547849d02bc0e610a5a6b : FAILURE in 50m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ea0cc060b07d7b9ae797ed937aa7fa4edfa3bc69","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-04 09:04:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14f841b389a3c1825b8ec926bae976d54d16df40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:57:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc876ee1753046108e2f722aaf327ea5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f631d7657e473b82e9eab07c230f45 : SUCCESS in 9m 11s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e27c02cc30d4a79add3beeb30ee49d2 : SUCCESS in 9m 51s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/bfd5f890aacd4953a859cd6680670a85 : SUCCESS in 12m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d55b5ec169041d0a5e4b335628e27b5 : SUCCESS in 11m 04s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/71d84d013eec44b7905652cdd6a695a6 : FAILURE in 6m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed37c02b9b8e435f87bedc487a51442c : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e259aa7c8aa450f8c736fe719062c80 : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcae5a1b63e0404db6864f5a0a16feba : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c2d14a4df824422ad46b223a14a45a6 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8e58fa0f7a5b4ec5a817120e566cc09d : SUCCESS in 1h 13m 13s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65c3042b970a411bbe81fef8f1cc8361 : SUCCESS in 1h 06m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/853b36b7ffa24fc6a4f0891dd049e20b : SUCCESS in 1h 46m 37s","accounts_in_message":[],"_revision_number":1},{"id":"03c182d153d8faf3ac4de73e0a8d17bffad2fc3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:59:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a932919481edd67cbbc8a64d2acfea657c6f9be5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 12:51:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcabddf9c494f909726362584178f2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6cf79df8eb47f9a0512ba8838d018d : SUCCESS in 4m 52s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ba8c517d62f4959acc4a4fd64899203 : SUCCESS in 8m 35s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/4cb21a28b6f04e4ea8d0e66587804983 : SUCCESS in 13m 58s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd243b73bb747db81026dde41ef9158 : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/932c341566c8447b8959cf7f23af8171 : FAILURE in 6m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d65002e575374f31a3dd0cd6b6ce35ff : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a310e1f7c0745beb150ad4698b09ac3 : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78ce275f46de44f0bcafb9d55bda5ca8 : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6b2f0eb65f44c329dfc6af17fdb3267 : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ec21f28bba04adab027a6a03a19380a : SUCCESS in 59m 20s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbb1cc224ed4a669cba83011aec6b1a : SUCCESS in 1h 07m 02s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f713b2f1b6e494eb220ed9f4c19abfb : SUCCESS in 1h 23m 17s","accounts_in_message":[],"_revision_number":1},{"id":"23541f67770f90c0421e8b76f3368485b2f72344","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 12:51:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d3b4b19433f30a0ee99d5131ffa93d3a6ebc8496","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 12:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5defa995584b48828ac31e10d82830b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034445ad809344e795e26b3dc7ded336 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e35a85ce6b8c23396fe60a96b5b5b3a025cc0b0","revisions":{"2e35a85ce6b8c23396fe60a96b5b5b3a025cc0b0":{"kind":"REWORK","_number":1,"created":"2025-06-03 07:38:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/951614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/14/951614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/951614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/951614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/951614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/951614/1"}}},"commit":{"parents":[{"commit":"5653f325603e41ee352cc29adeb28f9eb08b6799","subject":"Switch MariaDB to archive.mariadb.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5653f325603e41ee352cc29adeb28f9eb08b6799"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-03 07:38:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-03 07:38:32.000000000","tz":120},"subject":"Temporary disable molecule jobs","message":"Temporary disable molecule jobs\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/951317\nChange-Id: I942890cf89b9567067bc601f481bfb516ebb2237\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2e35a85ce6b8c23396fe60a96b5b5b3a025cc0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2e35a85ce6b8c23396fe60a96b5b5b3a025cc0b0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5653f325603e41ee352cc29adeb28f9eb08b6799","is_merged_in_target_branch":true,"change_id":"I1ece14502bb98041bca0e7589d4c0fe7511fa0c7","change_number":949138,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
