)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~951807","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2024.2~I9c30c8c61bb4f1d015baf997955a66adc4df01c4","project":"openstack/openstack-ansible-galera_server","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-08-29 09:49:00.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-08-29 12:34:55.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-08-29 12:34:55.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-08-27 15:38:00.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I9c30c8c61bb4f1d015baf997955a66adc4df01c4","subject":"Do not verify MariaDB version when installed from distro repos","status":"MERGED","created":"2025-06-04 19:45:39.000000000","updated":"2025-08-29 12:36:37.000000000","submitted":"2025-08-29 12:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951807","cherry_pick_of_change":951317,"cherry_pick_of_patch_set":2,"meta_rev_id":"829c3496f42a5ac52f517a995dfd1ef9a80c4214","_number":951807,"virtual_id_number":951807,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 12:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-08-29 09:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-26 18:28:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-08-27 15:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-08-29 09:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-04 22:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 11:52:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-08-26 18:28:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-08-26 18:28:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-08-26 18:28:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-08-26 18:28:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f9e5e7738332e4e01f115b287d97e68ff2c750f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-04 19:45:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"893ff58a15849ddfe798118a0a89c44b398bc2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 22:51:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a722e604650d41df9bca9b04f38d0c64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4991567bf044109dee87b1e98156d6 : SUCCESS in 4m 02s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb0281e21d764eb4831756f5acd35987 : SUCCESS in 6m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ff2971ab055490991e7a4064f656655 : SUCCESS in 1h 03m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4747a09dd49845dda985fe85ad65e983 : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25f519ec4b1d4833ad038e7cea517d5e : RETRY_LIMIT in 5m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b652a0a23fe4b3b95c6aaef20505bca : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fb27573f5a247c9ab92bcd13be34c42 : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb6372d6bb2c465884e06edcabd60e52 : SUCCESS in 43m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae406a3c32424828864d9de65eed419d : TIMED_OUT in 3h 03m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cb9c9c82280493ab308801bc4ab575c : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3845687dc339427bbd607e9dcabe03e8 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1819323010304d2581c2af6e1d6b8b43 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db2d389bacff49dc9b4e3a28b4c8cc3c : SUCCESS in 1h 28m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9fff6995aa74449a82751693e303d81e : SUCCESS in 1h 09m 24s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/456a67679945412b82f6db101ba9be66 : SUCCESS in 37m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f61b98a5034423dae42269d313e04ca : SUCCESS in 1h 49m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9e0ce38984c40c49a12dc4fbabaa8ab : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e86b03fa99a8476244c298771a50bb6389033344","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-02 11:52:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adf83021f224c5e780d7645db6d323e4c57200f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 15:03:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/477647fa93a04f9386fe606a90afc430\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533384739fea44d9aac8eb479e05797b : SUCCESS in 3m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5832fc76ffd4b65bfe651441bac0613 : SUCCESS in 6m 56s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/33384814895d4d4985f8941c10979013 : SUCCESS in 1h 23m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/689a0e4c66c3431dbf4c7605013de02e : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a547503176634d079183e566c6cf74df : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fd5206c09b4647d29e00533f1135a74c : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/929d4ba9b7654976bdbb339ca715195d : TIMED_OUT in 3h 03m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8182b3e1263e4886808b8445dcfd68df : TIMED_OUT in 3h 03m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c32adcbaecf4c1db3612757959cb818 : TIMED_OUT in 3h 04m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f887f0cdb484e0ea7805d840b881c4d : TIMED_OUT in 3h 02m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3805ec61b2004772ac68be37beb60100 : TIMED_OUT in 3h 03m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d090c69f8c54dce87642b797db69091 : FAILURE in 2h 42m 17s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5811eed1aeae44349d9c86f776c3c60d : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c455d01f515645bc8a2e1f5610be30d9 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4741fbcead4463aa2f6a5c2a431e9ee : TIMED_OUT in 3h 02m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1246ca00183847faac665c8cb42b74e0 : SUCCESS in 1h 56m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90029fd0028a4d99993950a71e10200e : TIMED_OUT in 3h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2b055aad751cc269d3b5d52eabec21893d4abe34","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-26 15:34:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"826469a60b7c1c51b0c346c12a76b80fdf58a1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 17:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518d1435d2ee4d3688a7d0bb93d33fad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609ee31e729b49649aa1e6fed8dff323 : SUCCESS in 3m 09s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ab065889c14925977c07cfc47655df : SUCCESS in 5m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/33f778378d7143efbf06add65a4566c0 : SUCCESS in 49m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/accba69703df4d548fd1644c11c14681 : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1b0c0045a4443d2905e1e6b85f5208c : SUCCESS in 1h 27m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/75bedb8e6bfb42c08553bbef35b33841 : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a18d427315fd4d65a4087a8ab05cdd7f : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77f1d525babc44e8a99cf9e09ee95e30 : SUCCESS in 1h 01m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c57898f06f534b7c94b687439a66fa71 : SUCCESS in 39m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e05ee3bb2beb4fd7b40d690ed5f47d1a : SUCCESS in 1h 07m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d2e34bc1e694acfae21fd94993be2b5 : SUCCESS in 40m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a946ac11255428abc28ff23dabf1ea0 : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7144eb4d11274af8af757480a7efbea8 : SUCCESS in 47m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e5f44e88b3884f7188fe1004d52b7ce8 : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9347e79c91c4609b9464d06982f0600 : SUCCESS in 1h 10m 26s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cefd490ddcf4a4e8f5e81027f07c7d6 : SUCCESS in 2h 06m 16s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9893eefe95c43aebc486f6d1f6fda01 : SUCCESS in 1h 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"54ed5383009d150943b13ec13ac92c2b16365f33","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-26 18:28:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac41a6d867f9499613cf26fc385cdca31e0f4213","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-08-27 15:38:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50de2bfd51abf2bcfa28146b7d84f90897664535","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-08-29 09:49:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a6c842ce5ab3d0ccd459a821e3838274954d9c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 09:53:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f3a6913b02e1d92d4b6d4e946198ad592b0ccbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:34:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cafbcebcb0cf4150a59398ed8b3f4d95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153f53859f164424a6db068f7f8ebc10 : SUCCESS in 8m 28s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c757b7626b42838f2e5a0ad2f88884 : SUCCESS in 10m 23s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ab291ace391e42e9b10e0a4a5c3507a6 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b196450b97441c0b36c39dd7be84022 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/71471ab4f3374da383952687ce1a7ddd : SUCCESS in 41m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2771d8c4c1664c4fa1e10fabac53b76b : SUCCESS in 1h 04m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2878db878f7c4ea5a5785b352895330f : SUCCESS in 44m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ffbee3d10fc4f3eb93c4f1f80e26621 : SUCCESS in 40m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f9422b4838644bba08a0de21321429a : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71468fd648404dfdbbef392576964965 : SUCCESS in 41m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3881fefd0e8e41139ed8d9d9d0fbcfbf : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ae54053a5b6446886ac3f729f6df946 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c95bd643d3e94cca81d2bea8f5b3b844 : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a2e3d2dab649879ccc4ddbed159763 : SUCCESS in 1h 11m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/742f8bbcb31647679fde17efec33b290 : SUCCESS in 1h 53m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9517df69d63642d491de587f6a61fcc8 : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fb6d977efadc5564142212337a7277f09a23a3d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:34:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"829c3496f42a5ac52f517a995dfd1ef9a80c4214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718befe29ecd4180b57b9f69dcc95fed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6febd169afbb4df0920925e460d4e106 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d46ef260cc432257a56b7b567666f1ea7fc99fe","revisions":{"7d46ef260cc432257a56b7b567666f1ea7fc99fe":{"kind":"REWORK","_number":1,"created":"2025-06-04 19:45:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/951807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/07/951807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/951807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/951807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/951807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/951807/1"}}},"commit":{"parents":[{"commit":"6684e16be27ccdb2ed260c2eedc7db0c4b185c27","subject":"Switch MariaDB to archive.mariadb.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6684e16be27ccdb2ed260c2eedc7db0c4b185c27"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-29 16:06:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-04 19:45:39.000000000","tz":0},"subject":"Do not verify MariaDB version when installed from distro repos","message":"Do not verify MariaDB version when installed from distro repos\n\nMariaDB which is installed from native distro repositories will not\nmatch our expected version. We also should not really verify it,\nas it\u0027s not up to the role to manage the upgrade cycle for the\ndatabase in case of distro source.\n\nChange-Id: I9c30c8c61bb4f1d015baf997955a66adc4df01c4\n(cherry picked from commit ce844bc4da2f2479374f17564c92c2fd60b4a4c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7d46ef260cc432257a56b7b567666f1ea7fc99fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7d46ef260cc432257a56b7b567666f1ea7fc99fe"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
