)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~953203","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ieba1274b06d06cc275dab94936b3fdfe5a8e8836","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-25 15:35:04.000000000","reason":"removed on reply"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-24 19:28:47.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-24 19:28:47.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-23 11:20:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-24 14:15:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-12-27 21:06:26.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ieba1274b06d06cc275dab94936b3fdfe5a8e8836","subject":"Cannot set fs attributes on a non-existent symlink target","status":"MERGED","created":"2025-06-24 11:44:00.000000000","updated":"2026-01-24 19:29:33.000000000","submitted":"2026-01-24 19:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953203","meta_rev_id":"6d9e3241f1188d01e4c28e00e32407240556c444","_number":953203,"virtual_id_number":953203,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-24 19:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-12-25 15:35:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-12-23 11:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-27 21:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-27 21:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 11:44:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-06-24 11:46:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-06-24 15:14:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-23 22:14:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-12-23 22:14:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-12-23 22:14:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"057d9e60c5fbd77170f3f70891030e7e92dfbced","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 11:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0c306a1d197d145a46dbdbfa24b9d6ef695e249","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 11:44:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ae47e3e4242ffb558989a566238f7f3efaaf4963","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 11:46:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3d16277ee207f8820153459eb272058581a779ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 11:46:34.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"64de1efc3d7150d3a817df6d327cfb69dddbaaa9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 11:46:39.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"5826dd1966cf5a43b43a9995749af111c8ebd320","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-24 12:47:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"416aa6e92761d0d1bbcf9fed7800b1826e86349c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 12:56:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d2dcf47cf39b7bc965d924208d5c793df18e4cbe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-24 13:29:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8dd5e72660e9850ad85ae6159e83490bc9d8414","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-24 14:05:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"51ab324a5270bb951bf8e5140d84a43d4e99c760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:14:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d475389b1442f0bd61c79712df1d6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fee4516a0c46828293808231a5f4b0 : SUCCESS in 4m 55s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bc6ca0d94194cb2824b2412f16850c6 : SUCCESS in 5m 30s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/21a0b04d19cc4373a3e3363c33660d8d : SUCCESS in 11m 35s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754613f1f4a34b02abd4ae621ea08915 : SUCCESS in 12m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/411d8448c4a74c2dbcebded35c32b862 : FAILURE in 10m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4cce72f8a25f4cd3a1779e030231d9aa : SUCCESS in 1h 07m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cde5895b2dc74568ae5951d36b3fe89f : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81322c4ccf5f4bff935ea5da415e6036 : SUCCESS in 45m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6818fe214b30466a8c0f9100c7192777 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/31559d3c1c994163adae3e41ab3c0cb7 : SUCCESS in 59m 11s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25e4a28dce8d4db6a4c07408ef5421a7 : SUCCESS in 41m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6439239741d140bebe67de78e8bb809e : SUCCESS in 2h 10m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c7b9914cccc12ffb5bcc543182c6233f841bc9e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-24 15:48:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5c0f94542c7f9964c4e9f60b1bcf08c3180f1724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-01 14:28:53.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"aad5700f85bf6aafa02833d2eb765cafbd37baa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 16:04:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/283d4bb2dc0e4b6ba545c1812a3d0138\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b38052bcd9d451c993541c63d1483a8 : SUCCESS in 6m 41s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ca31672678241cc8b4998bfbeddbade : SUCCESS in 6m 29s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/bd7698bcad304112ba0948e5dc31f541 : SUCCESS in 8m 07s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/854de5538c8e493ca696bdaffe39d709 : SUCCESS in 8m 59s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/fdde87fbe7a7448aa36879a24ea5a441 : FAILURE in 7m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/49a62cd813e74b5fa5f20e313bef5d9a : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb8ae7ed53614135ad9dc7863d90ad3d : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ab0a625c5b4666bdac15772156aa71 : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb03eb3d92454b0fba5fb8eb3df73f0a : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c82fa347ab5b47e184cbdc2e7e379a03 : FAILURE in 32m 32s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fa406750c8a4adc80c711a289f78cf8 : SUCCESS in 37m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eee810597fe4e0d9c13a4a7fa5e98ea : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":6},{"id":"196cdc8de384e343e3c78f35914c579174cb245c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-01 16:11:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b1f78f5b23a4f5087b967f2d461c1abfa3a0633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 18:15:42.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a86cf2fcf65446392a425bd48ff116f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012a595f12e148afa26b0399d916f144 : SUCCESS in 5m 43s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f35306b57034233927ddc2f1f8f78b8 : SUCCESS in 6m 01s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee676ef1f5b94ef9990ce8ecf6e8ea38 : SUCCESS in 12m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c96d95fd5ff94924914ab03cb93c824c : SUCCESS in 14m 35s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/684d162ed4844a13a3dfeba669a28022 : FAILURE in 8m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/10b45d4bdd9c42b0a34d67c0b45184ee : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/595b1c102e214171bbca753b36daea28 : SUCCESS in 1h 26m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f50af65546f4274afb87ee1027cc231 : SUCCESS in 1h 49m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f06571346f0d48c291092e9539011932 : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9c90525d8c1547159d8359a4381aacdc : SUCCESS in 1h 08m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6da5ae38c1864445b9bb46b943e7c9a8 : SUCCESS in 1h 03m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/593ea3fce32443aa9169b057424166c5 : SUCCESS in 1h 47m 52s","accounts_in_message":[],"_revision_number":6},{"id":"37a458586d14423181b4742050908a3573cfc738","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-01 18:17:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3d8ac0780eac069055e82bdcac138fa4949d9fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 20:18:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0da2178e70214e5b84cb72a3fc7f5b08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f094876932403a97baa739bf6d23ec : SUCCESS in 7m 21s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/221a6c58d22848c29a445fc80eee3415 : SUCCESS in 5m 46s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e79491406924ae19ad57a7a00a48c39 : SUCCESS in 13m 52s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9afefb7c57444bb8aefcc7f4b4e9769f : SUCCESS in 13m 53s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d7b584a817904217bed6e7b441f16fab : FAILURE in 5m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/07518c2af57f4d2193df4fce78280ebc : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc4b8a1cf06e4df0b7724f6ce7ff49ac : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e2b250e4a934081b8872eb0feacc4d9 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdef711e8ce04057ad5a7fad8c79628c : SUCCESS in 1h 53m 34s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f961e7b617d844b186a6f35b937a0b0e : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/902c78d1dedc4d68b440b31473d184c7 : SUCCESS in 1h 27m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1eb67bc0c6f4913bf79e0be9a1ec89d : SUCCESS in 1h 53m 23s","accounts_in_message":[],"_revision_number":6},{"id":"643308fd91700d81c565e6eeb4bd56e47cea3730","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-02 11:50:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f33e30e0f3f84763df07086e39b34b7094633a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 14:58:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bba0ed59ba4e4b6182e201316e483dca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512cd47149d94494a84ff307f0d863c2 : SUCCESS in 5m 53s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adde22a15ed9469a95dc2a3e43f32cf6 : SUCCESS in 6m 31s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8522f396d3b5462d880366514eceeaf8 : SUCCESS in 9m 46s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac688f1ad0974a9085bcd880f5ed070c : TIMED_OUT in 30m 49s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/91b1ee395fdd425284c08f1ae8779b13 : FAILURE in 6m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f237ba3385e24d1aa9ac6c4f58fc8f64 : SUCCESS in 1h 19m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaba82b334644bddb1993f0755dc122c : TIMED_OUT in 3h 02m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33af3cb455f9424087e01e227dd4e62a : TIMED_OUT in 3h 03m 47s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35843fb391384fb7a508796a36281b3f : FAILURE in 2h 52m 47s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2c3a6711f03e411088377e71b2dbf8b2 : SUCCESS in 37m 56s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce71a9cc4143464ca952c4c48c9b2277 : TIMED_OUT in 3h 02m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52182f8053534fc3b3fe6b4ea5bf5fab : TIMED_OUT in 3h 02m 50s","accounts_in_message":[],"_revision_number":6},{"id":"89bf0bc45667635f1dcb77f2bec4a5d3830803d4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-04 20:59:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a0680ac41b9dc27a19f3af4a1fff4cd9ed0d091e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 22:24:37.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05aece4aa0fc4cec91710ea25f6a3a80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecabfa4510e64eed8c523a9317769482 : SUCCESS in 6m 29s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e926241bb574428b9e8e7fc7a83ad67 : SUCCESS in 9m 28s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/598f769c9dcf4b09836d4bb6e7fb2f5c : SUCCESS in 11m 02s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/526406411233421daee1758ac0a883a9 : SUCCESS in 11m 32s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/82f90164e5834204b6295259b85c0edf : FAILURE in 8m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bfe6b36bf0ff4ccc910a92dad6dec883 : RETRY_LIMIT in 8m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8e3f07691114118b020b944edbef98d : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b44a14a42d7842ec878605b5cb3016b1 : SUCCESS in 46m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a70d3112645f4757813c2c7edf9747ab : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e9e6d094f12a4f15adefa580f2484257 : RETRY_LIMIT in 7m 58s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f7c54ce523c4f1aa50bf4b6d261b803 : SUCCESS in 1h 14m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bc1c4d5e92f4734b75fad0dde06efc6 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":6},{"id":"21a3560e54372ec1f84c67d27dfdaca039e8e491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 20:28:49.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"0ceab0e8e27ba03d02cc493f82d39fc3aa89798b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 21:57:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76456f24a52b4a7cb19f5e788ce9b829\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d8e288aa2a42c08682afced7566aa2 : SUCCESS in 6m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a74ef3b1e6d4b499a2069fb4a903b66 : SUCCESS in 4m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a338805440af46c485e3dee9f22d081a : SUCCESS in 6m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4abb6dd298aa425c97fc48b02b11474d : SUCCESS in 14m 09s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5fed3cd194ae4a31b47bcdc0ddfb9495 : SUCCESS in 8m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e77e983df7e444d3b85fef0055f5c919 : SUCCESS in 1h 22m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16733ba73fdc43ba9c3c08daca6a7f81 : SUCCESS in 41m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32cd7a58e1c846c8bc9fbf0bf0398fc9 : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204f46a697da41448d33198cf0b31198 : SUCCESS in 42m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/56d59122b13449f0adb4e0d0d77f59d6 : SUCCESS in 38m 20s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5df63338fcf3442ea44e648d07d7d72d : SUCCESS in 39m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8034799bc7a24aa398f1767f77cc7eb8 : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":7},{"id":"2a85b4e62eff15722cbe023e6a07ed913e2c45ee","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-17 23:01:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"7073db42fc6bc86d252bc230b4223c36fb94c04c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-26 18:33:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d8a28e8c2f9d58fa1388b29da1e829c86ba9e5f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-11 18:52:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06aeb1474ab0832ab14c8435398769552c022010","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-09-12 00:11:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ecfa80efba9d8ba1afb84768b741d327af130d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 00:24:55.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"06dec47f554e88e99be4da782200280f13f76c0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 00:25:40.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"683bf54daa2a9c0c8ea32a8e8356ae349924ee52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 02:15:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1038ce8566df411d916edd8cb0f9dc3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41918c65a2346f7bea046f65bf08b4b : SUCCESS in 5m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1673972cfef44beeb398c064ad5fc97e : SUCCESS in 6m 26s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/dfa7c33d92634ff2bee750d3ca08cfbd : SUCCESS in 10m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc9d777d6a5b4e9da28f0fe48829a5d5 : SUCCESS in 9m 28s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/768e158614fa480f81c352e1f35c05d9 : FAILURE in 8m 29s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d75b4006d6a44ff9a8664f357ff78656 : FAILURE in 30m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a109d79a883c492088b5a5ca266fae29 : SUCCESS in 1h 03m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b0118a2d3724cdb952d9c9765539067 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42e0b1914a224b7fa9964b0a30485eb1 : SUCCESS in 42m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caed42e21ccc445491a128f2e550a7a3 : SUCCESS in 40m 43s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a206d2359f7a4ca4b774899dc0a4411d : FAILURE in 7m 35s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/95e7bd5bbf23495895c06d4cb3b62722 : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40b50e531dba41d3a4066818a40ca091 : SUCCESS in 37m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c6343454767458e9a3d16ae72ea9cee : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":9},{"id":"2517dfa5f6f4b74f6247ea8818a89983862def94","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 17:17:23.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"413c05709584e36b675df4fcd60332571055395c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 17:21:19.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"7b4e527008a83b728d9f51b97f90414927fe48cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 19:13:59.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c4225506c6e4decb3b0955b151e771c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda668544c7944a1b7f3173d8ab257ee : SUCCESS in 5m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f32839b1094aa4b56434d68600b9a0 : SUCCESS in 6m 28s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b3d2bb08e24b4b3f865153f3fa5f94d2 : SUCCESS in 6m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69e3cbbe539242d3bfc06ebf21afd248 : SUCCESS in 10m 13s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2d6790b1db4e40a5acf33beb8d68638b : FAILURE in 5m 20s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a3a408298d664b33a0c7299e76e1aeac : FAILURE in 20m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/47d3525b50c94010b897abf7bcee5ea4 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23e5310094a544dbb91c26fbe76fec27 : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29685e53a44143628e8b9c1be002cf9f : SUCCESS in 1h 00m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8304ef5127942fbbba9d069e01e1f0a : SUCCESS in 1h 01m 26s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2629d29013964f2aa176de938078be0e : FAILURE in 8m 39s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/159af93f41ef45ceaf8d4fe58648a787 : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c80dbe4d89204ec3a9831fe268a97ed3 : SUCCESS in 1h 12m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9cd9e3a102340849ed84bb0c52604f2 : SUCCESS in 1h 49m 29s","accounts_in_message":[],"_revision_number":11},{"id":"5c1c63d980efdf3266b86abb2982e32e1f24ea9d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-09-13 14:06:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"05c1a1b02cfcc69383dd7a4dd934e388d85dd648","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:55:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"5a02c4ea85e299e986b1b692479ef2099e3a8d6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:55:59.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"24520695efbe7a6cbdeef7fc225b800effefa938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:56:02.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"66ed22d83c969938fe3d6339626b278e6891e38f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:56:44.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"3c160196685f7a52e92c23a2a8a472ac1384668b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 23:57:15.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"96d329299d1d220bb3f95455b14739a02175b1c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 01:54:34.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9683c8305af485ba89769a3a1a803a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89dfdf80b0b484b9aa205f611f6269b : SUCCESS in 4m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feed55e01391439d845cb2ea6a8c5b8b : SUCCESS in 5m 23s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7eaba353a3594cd1910b306f84e45b14 : SUCCESS in 11m 06s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/65aedcffdb8c4cb7bf5cb8ba19e33f45 : SUCCESS in 11m 21s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2266ab36b31147a4aac876fc1f5e3eda : SUCCESS in 6m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c680956b5b534a9484c7d8aa70c1793c : TIMED_OUT in 30m 52s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/715358c2d9434daca540c00825564eb2 : SUCCESS in 12m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2e9549a6d5ce4c67bcc1ac5bd80eb1d0 : SUCCESS in 43m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/95c13a77cfa945fbbdf42cbe370fb10b : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1dedf52f7b4349ee87da49a3386f2776 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/97864efc7f1f4255abea25611fe90f92 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2ce5726125c4e97aeebbfc3e6ebef89 : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeea04a9077449d2b32f00afb1c48779 : SUCCESS in 41m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f55ea8b23da843b8a65618e1e3e54aba : SUCCESS in 1h 13m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fd86e2923424138b341336ef43095f7 : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/effa74a1249f4add9df775efc23c5bdc : FAILURE in 6m 26s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/70375ab6338a4a6b99552e7cdb488a23 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/186de53875cf491dbf54b27ef2c8be03 : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/909cb48817a0468da45c093d37076cad : FAILURE in 6m 14s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2361f8eca0b4ef5be1e5125ac8c4fa4 : SUCCESS in 58m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efadf9c9fd1746e4a3893f329f6bbdf3 : SUCCESS in 1h 45m 03s","accounts_in_message":[],"_revision_number":15},{"id":"eb39e68c6412a2ae9b704b5b5e779e8a8fb1443e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-12 08:34:38.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"9ae419976bbaf64134cb17d32b892bfd92d040a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 10:45:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c6cb346bbe94999ae43fd6524052268\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d835c06d9bc4b37b478918c74de86f0 : SUCCESS in 5m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bff42f29a2d47fab729f39ff80f8935 : SUCCESS in 6m 37s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/25078905786341448d5a1996db342f04 : SUCCESS in 11m 35s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f599ce5fe8e64a8bbb53de51bdbe2983 : SUCCESS in 11m 03s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/967cab90676c4317a29b000931c320dc : SUCCESS in 12m 05s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/88682d4c7d1a47e3b2717ceea897af07 : SUCCESS in 10m 10s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4142891d01e4e0c9b75011fbd5f7332 : FAILURE in 4m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d4b0074c09eb44ba9e3dfa809dec02c7 : SUCCESS in 43m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/19bb0e43f2c34f3d92b5289392218456 : SUCCESS in 1h 12m 36s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e87c18100b42433d9d34ebc847ffc820 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d71a84cfb7a4359bd901fc9dbe16f1d : SUCCESS in 44m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f05b7538c794b5e93711ef8cbb3d295 : SUCCESS in 45m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd338b476cdc4455b055bc5e0764e2e2 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39e5c524335143ca9addc516cb1539c3 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90b28cc6fc374beaaf06aea39d1b20a3 : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e012caa80b694f2b858378d1c2b00e26 : FAILURE in 5m 55s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fb5cd0752f1b4fc38c68c1865625dd18 : SUCCESS in 1h 04m 26s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/10036b1bdb6a4fc1ab18b5a34f8aef37 : SUCCESS in 40m 32s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d2d26334a9b48e489c20a96d06967cb : FAILURE in 10m 44s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/814d7d54b1b3414d8e683f5dcccd4285 : SUCCESS in 1h 12m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d98852e3076d4eeea8f2a19d1308da41 : SUCCESS in 1h 56m 18s","accounts_in_message":[],"_revision_number":16},{"id":"33b38b72ffe4c1bed8440e679330bf7791f00b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 10:48:39.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"6d6a7aaf2e7dbf6f338354181d3158426a48d95f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 13:12:20.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e55e0886e44fdd88d6ad6582e4f458\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1aa987c80147c7b7891ac702c0dd50 : SUCCESS in 6m 21s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/062ce7b5a33349f083a961c716d3960c : SUCCESS in 6m 32s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/684fd4ad5d0742df9b28a5e9e4b8d230 : SUCCESS in 11m 27s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/02d3c2459ed54540b265b8c148fbe532 : SUCCESS in 10m 37s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c0385ab861946d98b1c7406d146a8eb : SUCCESS in 6m 21s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f02fb6303ce14349937f6b32240e1a30 : SUCCESS in 9m 46s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/95de9b95bb3c4105890687d67209cc38 : SUCCESS in 6m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae05489ceade4008865cfcd847935b01 : SUCCESS in 43m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/152c8685e19346b98a1add9b64cef1d8 : SUCCESS in 1h 01m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a5402ee4f76a43a1b4af1e99261963f2 : SUCCESS in 53m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da3959adb49040b8b275e83a2cb30f48 : SUCCESS in 1h 03m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbae3db88b3841469aba4810935711e3 : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8f141a75f6a47f49740ed2eb2ef68e9 : SUCCESS in 41m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcfb172bba0f44c58be4778a9fcd1c7b : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93e0d5b88221402cadd9530a1c822cd0 : SUCCESS in 43m 19s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/943f735435304eb7a2c19516010d5465 : FAILURE in 7m 51s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d730e928923435fa396f2d21bcb35fa : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/88e256367eb64644bc936214cd092d84 : SUCCESS in 39m 59s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7d3ab953cfb1428e9f3ef0867b6c0292 : FAILURE in 11m 32s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a925d02ab81f4c49bce280404d7d4ca4 : SUCCESS in 41m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0614f784609463abd4eac87ab92c203 : SUCCESS in 2h 15m 59s","accounts_in_message":[],"_revision_number":17},{"id":"f69a475443c31695e8f82d92ac6590496ad441f8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-12 14:12:42.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6f006eb60697445808a0e3617def73c375f55f71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-23 11:20:36.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"b4aaf4f7acc9c96c59d0116f7f06f47f11dd413f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-25 15:35:04.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"02712321d4a717b262de4f406857b12b98f2d78a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-12-27 21:06:26.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"0382719e2c1b3764677dd38ba5d4b99ee40a48ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 21:08:26.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"b0a458021525266adfda71d2840ec55b9b073048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 23:17:12.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/611f5cc018c5444db8855979c3d2879c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff566826d3c4784b8e4b0acc5169f3b : SUCCESS in 3m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d658601d4333429a8f1d344fef3b660b : SUCCESS in 4m 38s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7a22b3ca12ac435994b44dcc265c18db : SUCCESS in 6m 06s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/55d18da25b274f9b853f8e7914128407 : SUCCESS in 8m 57s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd7dbb83797468eadfa38a87aec2cb1 : SUCCESS in 10m 31s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/36510ba0186f497cb929c308c02c1fb5 : SUCCESS in 9m 13s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e60aeea901e44cbd9823d5375e18f614 : FAILURE in 18m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5bfa9c3571844ff18eec33954c1cdad8 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/dca628be66c2466899c1af71814bcc82 : SUCCESS in 46m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b7994aa2f6646fea05ce2f9576aa4dd : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c67203f939ee43d6b0e7357f9462b027 : SUCCESS in 44m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c01500d647c46b1b0a24324f48e9ac3 : SUCCESS in 40m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d0dea6ca16243f3bf8d6a912df64ef2 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/281a1790cb844756a1777f52fbd3224d : SUCCESS in 59m 39s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f42378484024988ac46a20a85b7e80c : SUCCESS in 39m 13s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/09563a8cdb754a76974ba7713374d9fc : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b74c9cbe19dd4a7da31155b980232577 : SUCCESS in 1h 10m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3137dc9361a46eb83de11859b0cd5e1 : FAILURE in 1h 02m 49s","accounts_in_message":[],"_revision_number":17},{"id":"f9c4fdbf6b61968f042b2845f24990f9874beb77","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 20:26:06.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"681aed8c2c3809c13917efe420817d353b69912c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 22:01:38.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45d9355e90d24121a0c4314abb4c335f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0a099b371b4c36b5e1b19d7f02e93f : SUCCESS in 4m 25s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf9ecbd8d3e84685bc032f6ffe0c6aac : SUCCESS in 6m 05s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/82cc3aa4744447a2a0345f233987e4da : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/71436470149d457c892b288c1d7c02c1 : SUCCESS in 6m 41s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50b0756764b34108a2394d003f378d60 : SUCCESS in 11m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/428aeeeeab3c4ff3b5ea086e0e9237e8 : SUCCESS in 8m 33s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/304ea2ca7e694a9080884a4c1907861c : FAILURE in 4m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/be3502544e7e428c9a8dc7e6dab99530 : FAILURE in 24m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/522cc3433a8045ff92d5d2e742a32614 : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8e978a67a6e043a58eca0b18d62ada5c : SUCCESS in 57m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2b6fced955ef46aa9c169c2fa04180ca : SUCCESS in 42m 44s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f7463febbf046d49f586c6021df3fbb : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65213aa0a5ec45f68d783ac5231e514a : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/715bcc6245cb4cdab91bb92e70510694 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f888d9663a554948b2365f920b1ad85f : SUCCESS in 57m 10s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d18de67fd934abf8e4f073c81016b27 : FAILURE in 7m 43s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7068b5efd1304b7397891006eaf1e93d : SUCCESS in 52m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/fc16e33fdf5f43b2be2959a53107eeaf : SUCCESS in 53m 10s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65b797655e904a42a27ad2745dd16e82 : FAILURE in 7m 48s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/547504dfc32e45d49053723915b2aff1 : SUCCESS in 1h 08m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8116c1f111754a1cb5d0ab8b2078d231 : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":17},{"id":"9a0737afd5de6ae5752702571076d944f6a0da32","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-05 14:40:38.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b52f470069d1102646e704beccceee0b495bff3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:34:44.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/174d1ae91abb4db8a380e795d504d4d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a3b4b64dcb4f0683399fe2167fb824 : SUCCESS in 5m 28s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae0f071b4a5f492ebc63debd6d0eed93 : SUCCESS in 6m 31s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/f59de6dc10be42a99c897c522d755595 : SUCCESS in 10m 41s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d76ea438893340a2b4a06d075cc7a15e : SUCCESS in 10m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d4bc01ac8d540b9877bab4612dc4535 : SUCCESS in 6m 14s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/03fa9098162b46f1bf039a65bf1e742e : SUCCESS in 9m 00s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/901a6dfc0033443ea183926c0f623c83 : FAILURE in 4m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c4cc81a86a9c488885d6738130870f64 : FAILURE in 15m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2700767f38fc4050897feb45c5256788 : SUCCESS in 1h 37m 35s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/16895a5a80834388b997cb1d39f6dc19 : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16c1f61a88424cb38e8a37ef9d3c2aef : FAILURE in 47m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4f21032274894dca955f404c9168e223 : FAILURE in 39m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/474e1c42e2d04f14b6d4062d1a1d4c77 : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96882a5ffe67447485e325a87c32862f : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/409f1b4cbebc45b08bbbe8b0a23b6427 : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef3a9f19561e40e4a0162b0364f70d00 : FAILURE in 10m 38s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b33db209138c49deb00bbf5521c22782 : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/734b8568818b4d6ab5796acbc6c4f853 : SUCCESS in 1h 02m 08s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cec8854944a847da88f51f924a3493f6 : FAILURE in 8m 24s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b716d2413da0412aba68a920a73cace8 : SUCCESS in 1h 06m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b6c683d5c2e49a6a0603922d553d228 : SUCCESS in 1h 40m 37s","accounts_in_message":[],"_revision_number":17},{"id":"3ddcfb71c452068f2dcd8816bfd01f4a3115c0e8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:26:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"333e490c4ff0994bc2a8e79ede577d3bf32be449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:50:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f983a406d81049bfad06e01369e59d45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caaf78af9f7b47f7bd0533bb84989ac2 : SUCCESS in 7m 48s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7490b99869d74ef6966a32f2695ee962 : SUCCESS in 14m 52s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e63f534cf3e54f948d43bad7dba39aed : SUCCESS in 13m 21s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/35399668fbaf41cc94fff47010d91772 : SUCCESS in 11m 39s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b7a421747fb4c5a9df5f76382e2e029 : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2881ba752ad142308732d3c46a195f45 : SUCCESS in 9m 48s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5f6e9ebbdeda47f39f52c69451ee7986 : FAILURE in 5m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d991675481d74063ac3ecf6aee1a5d05 : SUCCESS in 53m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/52eaf5c29eb947efa23aed2ffc2f40c2 : SUCCESS in 45m 50s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/38726259c5d44d6ea17a6fbba9ab0735 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b1e26c5daa084560a13baf823eb13202 : SUCCESS in 1h 37m 47s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7341a1c88d9840e9a99824cca3fe84f5 : SUCCESS in 58m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26737d31de80480b91f499ee550d1e80 : SUCCESS in 1h 18m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21d38aee8b2f4fdba81e2a7be4dfe4f2 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08fec9fa03a74ce887ed67485136a1a7 : SUCCESS in 1h 06m 54s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f3c4e4c9621f412488bbed1b266f3c29 : FAILURE in 13m 09s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fedcc8e3aa694d50b78e5318d7b47739 : SUCCESS in 41m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9b80f12b2bdc4c67b956b3673421ea0b : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ff75140ee0b4f689c003ece8257951a : FAILURE in 8m 56s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56b372dfa27f4246a54375de7fac0e3e : SUCCESS in 42m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a9a5e70ab3490f9806bda2db036bce : SUCCESS in 1h 46m 57s","accounts_in_message":[],"_revision_number":17},{"id":"45f51fae8d9da78833f426c655104ad0e91e4438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:52:47.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"8c32df551a78a556b19fad46630151a1fdc77660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:29:22.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1d293f3907f4c57ba84ccc70ad58400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a736545c30954940995b7c4d6ea55990 : SUCCESS in 10m 16s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b83476fa06a42098a38f55c41b8fa35 : SUCCESS in 16m 23s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/963644c75d1e4293b858e711efd5c42d : SUCCESS in 14m 48s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/21e7b36917224ba6b83b1a1a97f13115 : SUCCESS in 12m 50s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8a02a119217493cb1986eafe15fa80d : SUCCESS in 7m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/df1cff8c311449949f01af85f7104ade : SUCCESS in 15m 42s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/10d60c5c20ee4f0ab0207ef0d1b8b625 : SUCCESS in 59m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/654fb0aa7df04f3aa11e8d5ffbde15c9 : SUCCESS in 50m 16s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3d8b73b553c84f74affb1179142a7c8f : SUCCESS in 49m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e4e6cb0ea86411aa03c40cd0d1f06bc : SUCCESS in 1h 36m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74b84f218b774e358c7fd830f2e14f31 : FAILURE in 32m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f138e143d44469f90a5f299fd1c1ef9 : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aa3928b05764e5c96faeabb6f20b28f : SUCCESS in 1h 30m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cf4443e64e146628ff4462a926b9139 : SUCCESS in 1h 15m 34s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5fe0d16e790d48168c5f57a6d7596eb0 : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/56a93569479148ce982ccc825048f31d : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad0c045428f1485097c68897aadbfd39 : SUCCESS in 44m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b0307db8b6f4d0f9c739a0b6fd86503 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":17},{"id":"e840ba7ff67f222cc374ca027d6e4b8aa56d5564","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:15:39.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5cf1947d241f81dc248e25cc984ae5941f70afb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 16:25:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d981c5de53a54f6980030c05db5f6c67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2008fef3a3764e58884e2f9c34ce2f5f : SUCCESS in 6m 46s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97d248162d5a482e9dcad556aa8ef08e : SUCCESS in 12m 56s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8770d3beac2b44a0bdd2d20630c4c7ae : SUCCESS in 11m 42s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/41cfcf2ac2f444ec8ebb080b9104997d : SUCCESS in 11m 55s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d43db1fd1742d4bbbc9fc905d5513f : SUCCESS in 11m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/68f95d62d3f745f982ca7cff500b745f : SUCCESS in 10m 07s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06e4c7ed151e481095c631e5c2bbfc8f : FAILURE in 5m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/11e73e1c2a2d45d08f686749b4daf883 : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/163a5204a5c4449fb7c8537edd720c76 : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/92798804210a48508e53ad287562f290 : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9faef2bcab1f4c939a2205d7fcc20543 : SUCCESS in 48m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f5cd4ce29a4e42cd89a66dbbc159f169 : SUCCESS in 1h 32m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e7817b745fc48249ddcfe1193a7a099 : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eb40bc7b7d04722abc9a1cee0ce21a6 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6688427ba9084a109cb5007359d6da0a : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a1b73a6836034743b9917379836a0efd : FAILURE in 13m 52s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e08dbdd4eae346d98f10fe0a20744bdf : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0c2b859a6b494ca697ddec63e29ced12 : SUCCESS in 1h 03m 21s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9fe4a1f834e4aaa94d324af18d38e78 : FAILURE in 14m 06s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798084f2694a4deba281551df0ee040a : SUCCESS in 1h 09m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9c323cebf404ff29f13da672f02ff59 : SUCCESS in 1h 48m 24s","accounts_in_message":[],"_revision_number":17},{"id":"597154fe610058e7eb7737474e53d65b64019754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 16:27:36.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"fa9fffa27cd6b84a849c19bea57dfbf9d2b47b81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 19:28:47.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fea26e48abf4a638b6562a5c77600ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4991fce678e458b8b6806333673d5ce : SUCCESS in 6m 15s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70359cd5137b437883cd442d9a3f4476 : SUCCESS in 12m 22s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a16f8096ae04b8b9f33a4c64fd7fba2 : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/8cd57baae7cc47ddbfc5d5d1c238b5e2 : SUCCESS in 11m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66a5613369314bcab5766d5ec627d388 : SUCCESS in 12m 22s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/aaa90510c2c84bb8822624380fd98602 : SUCCESS in 10m 10s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e8c646b2f3384930adf5aeba673f57d3 : SUCCESS in 46m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d4ad8e45ed964aa68456b647804dc4f2 : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5c90cd77f189460c8a354b9b0e6dd610 : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/830145ae162a4f9eaa53cf9053f34e9c : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68cbb3989ecd4def91a7224e1b48ce06 : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57211636472c42f3a03500d13a5e0dbb : SUCCESS in 1h 17m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77de5464488a41d195cde7b221cbb156 : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eee05546263d4b3fa6749396292535a3 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/69401bc38a5b459c9e17c414d50942fe : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/66800f4bc86d475d83b239d05ede9ada : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5d154f6c28143b48bfdca37fe3a4427 : SUCCESS in 1h 14m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/730ff1cc01984d818f48745bc3e8e0e0 : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":17},{"id":"917ce9e552e7d7aa9687951a9de39bbbe8409f43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 19:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"6d9e3241f1188d01e4c28e00e32407240556c444","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 19:29:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2c1df678e24cdbbe49e32ffea14159\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326a35f7c782422d98632dcc46ead18e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"48db7e4507a91d3dabfceb198075d47863616f26","revisions":{"3c9f5703f50314ab2210c0dc842ab27163442474":{"kind":"REWORK","_number":1,"created":"2025-06-24 11:44:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/1"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"subject":"wip","message":"wip\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c9f5703f50314ab2210c0dc842ab27163442474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c9f5703f50314ab2210c0dc842ab27163442474"}]},"branch":"refs/heads/master"},"061da3bc9cc74df9a4a221a6282df87c4dab6a5d":{"kind":"REWORK","_number":2,"created":"2025-06-24 11:44:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/2"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:52.000000000","tz":0},"subject":"wip","message":"wip\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/061da3bc9cc74df9a4a221a6282df87c4dab6a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/061da3bc9cc74df9a4a221a6282df87c4dab6a5d"}]},"branch":"refs/heads/master"},"409b0681137689b0f24d521c08a29e94231932b9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-24 11:46:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/3"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:46:26.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.","message":"Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/409b0681137689b0f24d521c08a29e94231932b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/409b0681137689b0f24d521c08a29e94231932b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ebda46dbb579c8682d0c683498d77bd85a9bc200":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-06-24 11:46:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/4"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:46:34.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebda46dbb579c8682d0c683498d77bd85a9bc200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebda46dbb579c8682d0c683498d77bd85a9bc200"}]},"branch":"refs/heads/master","description":"Edit commit message"},"51d20a06ae3485e7a100150e64905e30c8482547":{"kind":"REWORK","_number":5,"created":"2025-06-24 12:56:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/5"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 12:56:50.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/51d20a06ae3485e7a100150e64905e30c8482547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/51d20a06ae3485e7a100150e64905e30c8482547"}]},"branch":"refs/heads/master"},"13a7b0241f3843a89e1126eaa81f9b415b8be0c1":{"kind":"REWORK","_number":6,"created":"2025-07-01 14:28:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/6"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-01 14:28:50.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13a7b0241f3843a89e1126eaa81f9b415b8be0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13a7b0241f3843a89e1126eaa81f9b415b8be0c1"}]},"branch":"refs/heads/master"},"798bb8e60fa321f12265b2443c90cf3bd1a0c453":{"kind":"REWORK","_number":7,"created":"2025-07-17 20:28:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/7"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-17 20:28:49.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/798bb8e60fa321f12265b2443c90cf3bd1a0c453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/798bb8e60fa321f12265b2443c90cf3bd1a0c453"}]},"branch":"refs/heads/master"},"e912fe63e0b8bce80b7cc8ac02cb3be6559534c4":{"kind":"REWORK","_number":8,"created":"2025-09-12 00:24:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/8"}}},"commit":{"parents":[{"commit":"a45b26d91f19148f99af94eb77123f81cd4476fe","subject":"Add support for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a45b26d91f19148f99af94eb77123f81cd4476fe"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-12 00:24:54.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e912fe63e0b8bce80b7cc8ac02cb3be6559534c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e912fe63e0b8bce80b7cc8ac02cb3be6559534c4"}]},"branch":"refs/heads/master"},"ff6527b81ed04ab69d644cb94f6f16dce8072fcb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-12 00:25:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/9"}}},"commit":{"parents":[{"commit":"a9718f24ed5c101bad2d3e6bc6b5f5a175630c25","subject":"Use management address when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a9718f24ed5c101bad2d3e6bc6b5f5a175630c25"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-12 00:25:40.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff6527b81ed04ab69d644cb94f6f16dce8072fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff6527b81ed04ab69d644cb94f6f16dce8072fcb"}]},"branch":"refs/heads/master","description":"Rebase"},"8605d4786d8ccffec3447cd8a3f13017fbf8cd87":{"kind":"REWORK","_number":10,"created":"2025-09-12 17:17:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/10"}}},"commit":{"parents":[{"commit":"a9718f24ed5c101bad2d3e6bc6b5f5a175630c25","subject":"Use management address when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a9718f24ed5c101bad2d3e6bc6b5f5a175630c25"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-12 17:17:22.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8605d4786d8ccffec3447cd8a3f13017fbf8cd87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8605d4786d8ccffec3447cd8a3f13017fbf8cd87"}]},"branch":"refs/heads/master"},"51383a5c7eee1c6971be0d24d88ea9af0c6f0d47":{"kind":"REWORK","_number":11,"created":"2025-09-12 17:21:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/11"}}},"commit":{"parents":[{"commit":"a9718f24ed5c101bad2d3e6bc6b5f5a175630c25","subject":"Use management address when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a9718f24ed5c101bad2d3e6bc6b5f5a175630c25"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-12 17:21:19.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/51383a5c7eee1c6971be0d24d88ea9af0c6f0d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/51383a5c7eee1c6971be0d24d88ea9af0c6f0d47"}]},"branch":"refs/heads/master"},"2dbdf12b4257b635d1ca6462c2ee43d55c240be8":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-12-11 23:55:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/12"}}},"commit":{"parents":[{"commit":"a9718f24ed5c101bad2d3e6bc6b5f5a175630c25","subject":"Use management address when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a9718f24ed5c101bad2d3e6bc6b5f5a175630c25"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 23:55:59.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2dbdf12b4257b635d1ca6462c2ee43d55c240be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2dbdf12b4257b635d1ca6462c2ee43d55c240be8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3993a697960aadaff35faa118d01e64c4b5be9ce":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-12-11 23:56:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/13"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 23:56:02.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3993a697960aadaff35faa118d01e64c4b5be9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3993a697960aadaff35faa118d01e64c4b5be9ce"}]},"branch":"refs/heads/master","description":"Rebase"},"1b64298874687cbb4460062086c056cc3fbf83a2":{"kind":"REWORK","_number":14,"created":"2025-12-11 23:56:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/14"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 23:56:43.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b64298874687cbb4460062086c056cc3fbf83a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b64298874687cbb4460062086c056cc3fbf83a2"}]},"branch":"refs/heads/master"},"1ad14386b16cf4992f55dc91f62ec53f5cf231d9":{"kind":"REWORK","_number":15,"created":"2025-12-11 23:57:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/15"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-11 23:57:15.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1ad14386b16cf4992f55dc91f62ec53f5cf231d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1ad14386b16cf4992f55dc91f62ec53f5cf231d9"}]},"branch":"refs/heads/master"},"fec9e5a58076bc00b57dc18c38385ac0340d6928":{"kind":"REWORK","_number":16,"created":"2025-12-12 08:34:38.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/953203/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/16"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-12 08:34:35.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fec9e5a58076bc00b57dc18c38385ac0340d6928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fec9e5a58076bc00b57dc18c38385ac0340d6928"}]},"branch":"refs/heads/master"},"48db7e4507a91d3dabfceb198075d47863616f26":{"kind":"REWORK","_number":17,"created":"2025-12-12 10:48:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/953203/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/953203/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/953203/17"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-24 11:44:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 10:48:39.000000000","tz":0},"subject":"Cannot set fs attributes on a non-existent symlink target","message":"Cannot set fs attributes on a non-existent symlink target\n\nTASK [galera_server : Update the local file system CRUD] *****************************************************************************************************************************************\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql\u0027, \u0027state\u0027: \u0027directory\u0027, \u0027mode\u0027: \u00270755\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027path\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027state\u0027: \u0027directory\u0027})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/usr/lib64/galera\u0027, \u0027path\u0027: \u0027/usr/lib/galera\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/conf.d\u0027, \u0027path\u0027: \u0027/etc/my.cnf.d\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\nchanged: [infra1-galera-container-ff4a7448] \u003d\u003e (item\u003d{\u0027src\u0027: \u0027/etc/mysql/my.cnf\u0027, \u0027path\u0027: \u0027/etc/my.cnf\u0027, \u0027state\u0027: \u0027link\u0027, \u0027force\u0027: True})\n[WARNING]: Cannot set fs attributes on a non-existent symlink target. follow should be set to False to avoid this.\n\nChange-Id: Ieba1274b06d06cc275dab94936b3fdfe5a8e8836\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/48db7e4507a91d3dabfceb198075d47863616f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/48db7e4507a91d3dabfceb198075d47863616f26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
