)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~956778","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I7a506b58353b01a4bd1bd49847345b22e6becf94","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-08-18 07:06:16.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-08-26 18:09:44.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-08-26 18:09:44.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-08-18 10:42:49.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I7a506b58353b01a4bd1bd49847345b22e6becf94","subject":"Add mariadb-client-utils to packages for removal","status":"MERGED","created":"2025-08-07 12:06:24.000000000","updated":"2025-08-26 18:10:54.000000000","submitted":"2025-08-26 18:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956778","meta_rev_id":"45ae8c45fd484a9f9c6f302e3308def429cee014","_number":956778,"virtual_id_number":956778,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-26 18:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-08-18 07:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-15 02:07:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-08-18 10:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-08-18 10:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2025-08-07 16:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-07 13:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-15 02:07:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-08-15 09:07:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-08-15 09:07:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-08-15 09:07:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"f01fb5207a4d57eeb43afeda9e8a68ea7be7b015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-07 12:06:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba5804fb46d4cb8935ea19a227f7e35dd661d9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 13:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bd5833ef414cf0b038ff335dacfc6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66ad2ba1b08440eb6a78139d3580c1d : SUCCESS in 7m 17s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a888a264cb5c4fe18b86f2dc9c35b541 : SUCCESS in 5m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7b06cc9c501b4428b893b80d674a7296 : SUCCESS in 6m 10s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2254df5a44084e00b2ad95aa66c6c74b : SUCCESS in 6m 19s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4eb47f8442cb45e495598d4cc0da2da5 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f2ae591c40344eabeb67c71d9ffa367 : SUCCESS in 44m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c0c55571d4451eb844e9feb0fccf8c : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/582f52d9b6eb4f6aa6d0f29a08935887 : SUCCESS in 40m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b25b5c463f44797b816143a46a2da09 : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c15cba93bb484335acdd9b926a5cb3a2 : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb5323590d740ceb17143305a41fb32 : SUCCESS in 58m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45c5f16190c74d958375b81e605dd9c2 : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6061cac49567d9dd96db9b25acfb515dc4f253b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-07 13:56:41.000000000","message":"Topic osa/el10 removed","accounts_in_message":[],"_revision_number":1},{"id":"6118e7dfc23ba20154bfe5e4301a127fc914a630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-07 16:47:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"94f8c8c6694cb22f0380b86f96df5539558ea908","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-07 16:48:03.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"5c4cd0c5b2ecd639dfd95cc1ba122c63d3c94894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 18:25:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a158f08cbc54cf196aeb3124b242e5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c1ff0cb3eb43649ee77b74fa22d426 : SUCCESS in 5m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fc937f63610448ea05a89dfc87c6faf : SUCCESS in 6m 56s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/36c0be89f96043d39c5bbe295fca194c : SUCCESS in 12m 10s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d55bfc9f100c4008b27c9d014b610c96 : SUCCESS in 7m 21s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/b26430ebf4804a4eaa97a66a4f054b5a : SUCCESS in 9m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b81be39a8f1482aa1a9a2a71a54fefb : SUCCESS in 43m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdfad3a875d14a2c96f872f6ed26628b : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e0f91f0e2324811b7726cb34ea2d619 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da3c241b187a4fca88ed0eed8d7b2a06 : FAILURE in 1h 20m 49s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9155d94003594976ab61c62c9d8fde22 : SUCCESS in 41m 35s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/804226f9120347a7a5b59abf1e036ca0 : SUCCESS in 39m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d83f91da3f414fbf970935c99b15f9e2 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9028495864955ed3796efb23679a8c74dbee066c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-13 07:23:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39bc3e27ce11c37569784c297a573fb2a251420a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 09:23:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edcfa2d3b5254547a9ac83170de35f25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d97200371ed49379c46bd01ac8650bb : SUCCESS in 5m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92f5de9141a64e95888140d5b420f42a : SUCCESS in 5m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e538218a9c9345d6b2c0a0e48a177893 : SUCCESS in 12m 51s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a819d49607e2462a9bd14043cfb9cdce : SUCCESS in 13m 22s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2950467811fd4827b19e07e36c31c715 : FAILURE in 5m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/95c470360417456db61eb7c0c3046961 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aaa268c6ecd435bbdd5cf392f5b650c : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9209a6b46f94f9094ff3af1ec83ee69 : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bce9c75174c42639bd964ca1dcaf4ac : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b582ebf3bf334fb1a8a9f3396a8e0ef9 : SUCCESS in 40m 28s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/506221c553d94cd59105d1923a503b87 : SUCCESS in 1h 10m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68dc288403034b9e8a50994f2d687b0c : SUCCESS in 1h 45m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fc604b7085c99a1d6a1fd972c91b08a6c8b7bfc9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 02:07:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abd701541054f3691fa91d0788ee70d36a5fd557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 06:17:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ff7a63ddf2f4b66b7f3f2a0dd1bd67e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b5f301582a4e1c8eedf3ed05c8c0f1 : SUCCESS in 3m 24s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5938b3ff47f4f02af32ca26d734cb45 : SUCCESS in 6m 27s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b89ffd1ae7e44244a3ec7fc39e34bd70 : SUCCESS in 7m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80c2ca8a31f94aa393398a812881decc : SUCCESS in 6m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c4d6a3f03014796b8a19b96d7e0a0d0 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bab4023d64fa4377936d89ee294b1b91 : SUCCESS in 1h 01m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dc64d3975314b7b9b1533d17c0c07c9 : SUCCESS in 1h 23m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87ad4364557d46ba895f7803d7d13945 : SUCCESS in 45m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b49f4f33b3d441e99f4a86b8eeb7d4be : SUCCESS in 1h 09m 03s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3d7269b8aa3441b98af40397fc2883fe : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b06d0725ee3f498596d10d8a68c15fc7 : SUCCESS in 57m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04639990342449d9b6b1f11412ffe504 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7ff9710967bbd32ad570c613e0e085ee1a83d8d8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-08-18 07:06:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2fb1c464766afe4353d89a940631511d348b06e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-08-18 10:42:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70b8c528236533bccc9ae5934df0a954c06ac887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 10:44:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b10bd47937cc87d6b26ae5fe1a44d3998c12ffe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:13:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82b832a469294821816e25db0e0dbf1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa38a682325547d48ea95180102c850d : SUCCESS in 6m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b406ae7162364e30b1e75558743a3253 : SUCCESS in 4m 14s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/951079dee8ce4a40968228e406dc7a52 : SUCCESS in 12m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa5f13a7cac9499f99f6684b9afbe647 : SUCCESS in 7m 31s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/785a63a2bb4a4f509a5f9f309463f578 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f2dd384ce57440d5bf1817457a978a38 : SUCCESS in 50m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f7ea2c010f404d7c87fbd2b2b930f98e : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/692a478e4a3e43e4b297fa6e6bcf179f : SUCCESS in 45m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6384b06a75a94e1eace0dfc19baa8d39 : SUCCESS in 42m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe09bab22784414f9187e40becb5b9b5 : SUCCESS in 43m 08s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a82dc86211eb4921896f0f9e1157c539 : FAILURE in 6m 27s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/10a0576c526643a5b1bde24b28ae9295 : SUCCESS in 58m 36s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e171e823620949c2a5e9d64fd55cd7db : SUCCESS in 39m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb53cf0774df472c956393af3dca3479 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":2},{"id":"8cab50d254de4114a81cee06183490e30b326338","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 13:21:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f3bb409e9d7e2abc1faf360b895ba495f603377","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-18 15:31:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00580762aa6e2ba7e4149307806b4ddc62aa3789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:05:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e511791f2ec0448fa299f457a74928b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f7a982713b4999938955412898ebd9 : SUCCESS in 5m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee0535b116ae4a5292daa6700be7e571 : SUCCESS in 5m 14s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/918aacd17d5c4d788418ecb77f5fc7a6 : FAILURE in 5m 53s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/433207ff45ca4baea1978269ed9bfaa6 : SUCCESS in 8m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1845b7e79564b6c80c25ad7294314be : SUCCESS in 9m 32s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/fe4c0f5085d34d1eb03967fc2b019699 : SUCCESS in 1h 00m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/675d073bb3a3407888ae205ec898373b : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cb7aa82d60a445e85c4b5e4f325d487 : SUCCESS in 1h 04m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/845ce7dafd304d80bbfb5c73c013b917 : SUCCESS in 44m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24dc5464b2994558badc9dc319de2217 : SUCCESS in 51m 11s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ac2be6ba77a54f60897b645be13545cf : FAILURE in 6m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/856ea9e86bd4462799bc1b0ef85b57d8 : SUCCESS in 1h 12m 19s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fb7b2490e0b44e7afd7054d9246a408 : SUCCESS in 1h 09m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c24fc7cbbf442c498add79b322d07e9 : SUCCESS in 1h 48m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bc5823145827c7c643d9c8fdfe4775610a9bc3a3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 16:20:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5901bca318e171c991c7b767dfe7e88c7f61a6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 21:23:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/033a5c52fff74d97bf40a34aa36c3164\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7e6f9385244158816aa1f32d5ea0f7 : SUCCESS in 5m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c81df01bffe648149f519facbba1d849 : SUCCESS in 6m 36s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd7abed9afe4408ea95c1468d0a11e31 : SUCCESS in 7m 34s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b65265344d446b3a4f9dc51c91146db : SUCCESS in 6m 35s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/98eb14e6a026464cb5f98b9f952af4c0 : SUCCESS in 9m 25s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f97d7c011952488a95961c6f22ca9fa3 : SUCCESS in 1h 06m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/67d4f5b7028a427ca66a915eb43e2499 : SUCCESS in 46m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ff4b325ff774389874119e55e8417c4 : SUCCESS in 44m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f31865c2b544efa86e68e8537315be8 : SUCCESS in 43m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c137e2a6dd1e4b58be1f6cdd87e73f8c : SUCCESS in 41m 42s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c9ae5ad160ec41e8893f00e99fc9c9f2 : FAILURE in 14m 12s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b1d8986b9ab940cfbc14d7f66ad49900 : SUCCESS in 44m 55s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0171bedcc02b4f9aa1d67e3e7d9d383f : SUCCESS in 39m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36381c91bff1451e8b74eb6216a8d018 : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d5a34bff3b9a6c1925e8beae0dac38b588544b77","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 21:46:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"323fd45d5b20e0927e7461de2295822739dcef1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 23:39:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/556867b8520042faa4ee7676fc5b40fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4d9c2dc64e49f699cd53dd63a051b8 : SUCCESS in 7m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6105daf574da49c4aeae41569103fb2d : SUCCESS in 8m 57s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5493722484334d888755331c0a2a60ca : SUCCESS in 12m 25s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f000a941e6bb4fa8b434c3716fc39e93 : SUCCESS in 6m 17s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/7b4b8c10aa1641d1ab08eae8cd7ccb71 : SUCCESS in 9m 55s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7a219c0a92ad41fbbc7e01771c87dad5 : FAILURE in 1h 20m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/83d7f13d61bc43c882a78ff178f5e93f : SUCCESS in 45m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2d06c73c2184388af7edabd9f7a8b38 : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9db89d35d32f411d9035bb517c210eb3 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1d8c536004e45bfbc0f44dc33e21dd9 : SUCCESS in 1h 22m 42s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/164f36dc445e454f9c6729d04b394b17 : FAILURE in 13m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f3a5e64d3caf4f82a4e6be2a86f3396a : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f9bfdbb11754d0a8ed26a3150718787 : SUCCESS in 1h 16m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ada4a16d461f40679a12aafe4b4c2fa0 : SUCCESS in 1h 50m 28s","accounts_in_message":[],"_revision_number":2},{"id":"832f3fe842d6666ef61d2d9c78e1527a74cc2ede","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-19 15:05:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d18c062c26e911dd2216377e7f0e30c1038493c3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-26 14:31:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47a11e95ec1df72f66f8e1d05b4b5a44837282c4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-26 14:31:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"041c57d200ca83b69dcd1f94124c46176542d024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:08:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e0dfca371c4e38a7603bf7e379d141\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e94c1e930b4a70b3aa3fa32fa1cfbf : SUCCESS in 5m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e817068e34054f2c9fe5da2ccb81b85e : SUCCESS in 10m 02s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/042af230010a4ac29a024b63a5195935 : SUCCESS in 11m 30s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd6bcaa0786c4a4186d775b31699d988 : SUCCESS in 11m 27s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1cfad5829cf740bcad849fb20b4db7c8 : SUCCESS in 9m 37s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e5402bbe8b5c4fe8b078bce118740c50 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d6ed3ebbddad4cd190cdda3ef8e90443 : SUCCESS in 57m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df1083dd36514f208e59381a53198215 : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/425b3141b4dd4681806eb42293210f44 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8a4e1c75e3541839e3a8c6fdfad3395 : SUCCESS in 41m 29s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8335c29c89e24effac62f9705884df73 : FAILURE in 11m 50s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/defe11ecad754e8198cbc1529feaaeec : SUCCESS in 1h 04m 37s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab871f16953c4d96a3fb1472e1d1389c : SUCCESS in 42m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfbc67a50f624165971e9a6c2348e45e : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"93bd22319064585509c1d4d86bb73c621666151e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:10:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2172b8801a16888c0eeaacc85a225daa46d0286a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 18:09:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db71f5f024b41a4ae3a5d144b766fd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c6a9b08fe34e838a293e76e3b511e4 : SUCCESS in 2m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b3abd657e04a439e22980228256c04 : SUCCESS in 5m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a08f34b55d9f405db52be0e7fd24ef5c : SUCCESS in 7m 50s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba40ab0c9e4c4966a922c2b5f73baddb : SUCCESS in 11m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8ab24baa29544d3098f88b9d8af72378 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9962e412fdd7449c8d8a51ed1255fb43 : SUCCESS in 1h 20m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5b7d228856249b494e0488abab7fb4d : SUCCESS in 1h 06m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f81df29d82944d3fb15ab3c5ae461f35 : SUCCESS in 1h 11m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c49028e9ccab4964ae35f9cde60bcb56 : SUCCESS in 1h 05m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0178410f109f4b14b3aa4cc8314ec68d : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/508bcd234e934a72a3db6220cf6fc5b8 : SUCCESS in 1h 04m 19s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6cda63b0b9a4f59a6f3adf50ef323b4 : SUCCESS in 37m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e214c0a9cae84873ab39e88aad5ed542 : SUCCESS in 1h 54m 19s","accounts_in_message":[],"_revision_number":2},{"id":"688dd73257a3a2af8a2787c7a79c2aa4884eb838","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 18:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"45ae8c45fd484a9f9c6f302e3308def429cee014","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 18:10:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc15245d4ae648eaa58614199fc5442f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9062c3342eb4453943a7dba9d2e059f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47e00b2e53a276769d37e1993e18d094949132d4","revisions":{"c411e4ea1434d8194a026424829742fc658f8cee":{"kind":"REWORK","_number":1,"created":"2025-08-07 12:06:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/78/956778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/78/956778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/1"}}},"commit":{"parents":[{"commit":"d9e6e8dbebec8a753b3761ed1a89189fb713f5e8","subject":"Bump MariaDB to 11.8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d9e6e8dbebec8a753b3761ed1a89189fb713f5e8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-07 12:06:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-07 12:06:11.000000000","tz":120},"subject":"Add mariadb-client-utils to packages for removal","message":"Add mariadb-client-utils to packages for removal\n\nThis packages does conflict with MariaDB-client when distro-\u003erepo\nswitch is made.\n\nChange-Id: I7a506b58353b01a4bd1bd49847345b22e6becf94\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c411e4ea1434d8194a026424829742fc658f8cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c411e4ea1434d8194a026424829742fc658f8cee"}]},"branch":"refs/heads/master"},"47e00b2e53a276769d37e1993e18d094949132d4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-07 16:47:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/78/956778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/78/956778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/78/956778/2"}}},"commit":{"parents":[{"commit":"fa66b49e009b9570773c99d26b1c13b9d8c195f5","subject":"Bump MariaDB to 11.4.8 for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fa66b49e009b9570773c99d26b1c13b9d8c195f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-07 12:06:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-07 16:47:58.000000000","tz":0},"subject":"Add mariadb-client-utils to packages for removal","message":"Add mariadb-client-utils to packages for removal\n\nThis packages does conflict with MariaDB-client when distro-\u003erepo\nswitch is made.\n\nChange-Id: I7a506b58353b01a4bd1bd49847345b22e6becf94\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47e00b2e53a276769d37e1993e18d094949132d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47e00b2e53a276769d37e1993e18d094949132d4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
