)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~957659","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2025.1~I7a506b58353b01a4bd1bd49847345b22e6becf94","project":"openstack/openstack-ansible-galera_server","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-08-18 22:07:03.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-08-18 22:07:03.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-08-18 11:35:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-18 22:07:03.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-08-18 20:00:57.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I7a506b58353b01a4bd1bd49847345b22e6becf94","subject":"Add mariadb-client-utils to packages for removal","status":"MERGED","created":"2025-08-18 09:28:39.000000000","updated":"2025-08-18 22:08:04.000000000","submitted":"2025-08-18 22:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957659","cherry_pick_of_change":956778,"cherry_pick_of_patch_set":2,"meta_rev_id":"e5c989b2ad1341f839f65cdeae4067b3ae680639","_number":957659,"virtual_id_number":957659,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 22:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-08-18 11:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-18 20:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-18 20:00:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 09:28:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-08-18 09:29:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-08-18 09:29:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-08-18 09:29:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-08-18 11:35:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-08-18 11:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1361d0c77e309708f71da054d66f8ca63e7339dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 09:28:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"96419fba0ac93cc55e11c2a4b258ac73a3994991","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-18 11:35:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c0bef8ce8f535d0cf2fa525bc64fed0df2ba5ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:37:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b14ba0c69ad419cb8cc478e110c5c9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803e75bddfcc4fffafd1fd36a577d08d : SUCCESS in 6m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95ec16c8a78244dda63b6ea27a070159 : SUCCESS in 9m 56s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/33be2c153046431fafb9092a00a14c92 : SUCCESS in 11m 45s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/915090a5f4e44ee98e993a410b16d1b3 : SUCCESS in 12m 48s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b94ee9f9e274d1388202847200dd2a9 : SUCCESS in 11m 32s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2281f14d3bd647c081a6d34b71818ddf : SUCCESS in 10m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d6ef5b2252a47a19d2fb8976617861a : SUCCESS in 10m 32s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c88110005dc245388bc533449d2c9e0d : SUCCESS in 9m 27s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4889d4eeba354c64a12c83c4a5513f96 : SUCCESS in 42m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd4a9447c64a45a48190b948ce4df50a : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/77928a2d7035429b88364b99f7dcd319 : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/06d837950a1f4044bbe0fa368382b169 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae8ad46ae8274dbab1be8f18fcdd0fbd : SUCCESS in 43m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c07ca45d90d049589ee5e6e65102ecce : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2144a32164784d4b92f95155bbbd85fd : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56132fff03b04d11ab6182885315afba : SUCCESS in 1h 01m 15s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/341f227b027245cba635cbd269071965 : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2a54f943184472380bc81307f938ee5 : SUCCESS in 53m 00s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d837921c1c74a799917b1ff2b294453 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b83e8bc977de4ee08a140c46cd7aff4b : SUCCESS in 54m 30s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/548c7b3f3a4a4643af83b2059379edea : FAILURE in 1h 12m 54s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b5eceba92de14010b4c04e8aefb51ea4 : SUCCESS in 59m 39s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f201687670154c318adf63bc59f0f61a : SUCCESS in 2h 00m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b39337e4e7e34395a7d13fe678b03c9a : SUCCESS in 1h 41m 41s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89ec12b4520c4157a197b4f53018b1f8 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"15af5eb94569a8d9b196528ed72d9cd84a791c2b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 11:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f872a56a092061b9d9760b088f8e6da88de7dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e6daf099a46499e81637261ff09276e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b43dd4c43c4f69914af0ce9bb11cce : SUCCESS in 7m 11s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d8cb83df2c949118278150bf16c9cc7 : SUCCESS in 7m 20s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/4332564a5be74c678c68f737cf8838c9 : SUCCESS in 12m 15s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ed46f08c61d4a7a8ac7404fe5e26a8f : SUCCESS in 9m 43s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18e3069273bb4504a66a9f9752770a29 : SUCCESS in 5m 51s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a910cf442cac41a5964d940a1e4c939a : SUCCESS in 12m 43s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2edb4bd3f52d4e43b6ff3aa1ae84fe60 : SUCCESS in 10m 13s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/de2e4fb9ef6f460aa112f1b19f8e25c2 : SUCCESS in 9m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1bfc8981975432f8f4c701525d4d541 : SUCCESS in 42m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c0c488b2dae4d5da9d192416e7061a7 : SUCCESS in 50m 54s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50778f1a63714cc49fe37cb050df1a49 : SUCCESS in 42m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b09cb48b6644d58a0586052a91cb47c : SUCCESS in 41m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/109c0457aa3341c0a169f14121a05a1c : SUCCESS in 40m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dc00c1cca4c4b10990910b66c815345 : SUCCESS in 1h 09m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c1e100c62d34f2089fc79f88dbbff4a : SUCCESS in 39m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5124bfbdd2e48e0aef6a9c1ee849b1d : SUCCESS in 40m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a21f4bc0c4e4dfa9165afa7520be153 : SUCCESS in 1h 07m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59e2ddf9d339489c83804863d1441b1c : SUCCESS in 1h 00m 24s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd409858db3d46c49143b87d837c3a49 : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/92e78b6b357a43b290d3b840bd93f0ca : SUCCESS in 44m 01s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58e6a1280d454183b987799c36066e0f : SUCCESS in 46m 10s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/482714db2a0141f79d49bc444d60b641 : SUCCESS in 1h 41m 37s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19b94be093bd404eadbbcf2d25f4ea2b : SUCCESS in 1h 05m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7baaf3d5e9a4da1b6c63f5ad444d623 : SUCCESS in 1h 28m 21s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869bdd02a70443328daebb7b2ed8ec41 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"faffd19655ab716873841a35f968d3c282cd6f09","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-08-18 20:00:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c39817e45c09431ea8094a86348726d34a427cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:02:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cef4ca4a0327961dc1cab951ac1490703536d227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 22:07:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f27026f3a9486495598efcceff46fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b751af2b513445c3a464cb81d1e75f68 : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/836b9c2a2d30459eaac8d8c193bb7478 : SUCCESS in 9m 37s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ab270928ef841309316ff09f5e7fcc2 : SUCCESS in 6m 43s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/46614bc1b5a941c586488762ecaa9e30 : SUCCESS in 12m 28s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28e6d28c9d2442e1b73b1021ac5fad79 : SUCCESS in 7m 04s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dbc3f8c90224eba95d4a4214a27b1a6 : SUCCESS in 7m 25s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69172bab337e4cf998ce29d720bfcf1e : SUCCESS in 13m 16s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8453972aa1ce421d802747baea3349ac : SUCCESS in 9m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f23b6b53f4f441eacf5188aa39c22cd : SUCCESS in 45m 55s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c38980ca49204469bf2e98ad9495bd27 : SUCCESS in 43m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4d7e31bea44a4c0683b0f7e20a16d31c : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7154557b982d45b589ae238f2fd7f69b : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/596b63cabc1944cda0da677597744076 : SUCCESS in 41m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dddbc0d546648bc9ac69b789e5f12d9 : SUCCESS in 41m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798205ce14164660be132da9d532cad2 : SUCCESS in 42m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edf0595570ae42e183c8b33bc6fb6d3a : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c004771d5dee4bb2b4186e28c4cd17ab : SUCCESS in 44m 59s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0e0f00a81044216bccfbaac847413cc : SUCCESS in 37m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ca33cdfeae8b4c5f914177704a3d227e : SUCCESS in 36m 51s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/356559a20c7d42a8af3bc15ddc5b7204 : SUCCESS in 1h 12m 38s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b557eb0e7c0404caf60bbd45e88f2c3 : SUCCESS in 1h 02m 29s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/62fa5d82f49e458b9aa07fad90961d17 : SUCCESS in 1h 58m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49677392f7f540438180c5efae6dcab0 : SUCCESS in 1h 41m 20s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaf2ff41469e4e0cbfa128e4931f660b : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8495abc1a51e093237c6b94d830ca3ec1e6afcb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 22:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5c989b2ad1341f839f65cdeae4067b3ae680639","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 22:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a860e115ec43fe9357c36c5bb78c18\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726e835fa5d8459f8a834723df20cec2 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"731ace7ab12ad09ef1c0945c3a2719c18ed9e88a","revisions":{"731ace7ab12ad09ef1c0945c3a2719c18ed9e88a":{"kind":"REWORK","_number":1,"created":"2025-08-18 09:28:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/59/957659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/59/957659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/957659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/957659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/957659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/957659/1"}}},"commit":{"parents":[{"commit":"ab5d15ed970c46901104c9d59e41c0f4a22745f6","subject":"Bump MariaDB to 11.4.8 for EL10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ab5d15ed970c46901104c9d59e41c0f4a22745f6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-07 12:06:11.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-18 09:28:39.000000000","tz":0},"subject":"Add mariadb-client-utils to packages for removal","message":"Add mariadb-client-utils to packages for removal\n\nThis packages does conflict with MariaDB-client when distro-\u003erepo\nswitch is made.\n\nChange-Id: I7a506b58353b01a4bd1bd49847345b22e6becf94\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/731ace7ab12ad09ef1c0945c3a2719c18ed9e88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/731ace7ab12ad09ef1c0945c3a2719c18ed9e88a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
