)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~959027","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6aa4ea1b0c4151fdde2dbe26c02949cd12b77e4c","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"deb822_loop_label","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-09-24 21:38:29.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-09-24 19:42:33.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-24 21:38:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6aa4ea1b0c4151fdde2dbe26c02949cd12b77e4c","subject":"Use loop label for deb822_repository","status":"MERGED","created":"2025-09-01 10:24:13.000000000","updated":"2025-09-24 21:39:14.000000000","submitted":"2025-09-24 21:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959027-deb822_loop_label","meta_rev_id":"791d13cca2ff906972d27373aeea7b3dfb0daf1d","_number":959027,"virtual_id_number":959027,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-24 21:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-09-24 19:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-18 12:00:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-09-16 20:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2025-09-24 19:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 13:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 20:31:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-09-18 12:00:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-18 12:00:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-09-18 12:00:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b9f85814137417fd402f8b7f8b4af8e06ee63aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-01 10:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e41200f12c3893edcb02ddbfa8045bdf19af5fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-01 10:26:35.000000000","message":"Topic set to deb822_loop_label","accounts_in_message":[],"_revision_number":1},{"id":"87c98f5bd2d93dacd93f28e5c203c649f599ae1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 13:33:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13fdc35f39c7479087c7279dd8bb6238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e292fe3edaca45d3bc212f8675946a18 : SUCCESS in 5m 49s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d5a0b7ac53745b9bcd9bf70173fb878 : SUCCESS in 3m 59s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e40a86a9bf74d63833a22e413ac7966 : SUCCESS in 6m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72bcfdf3f4774d4e903cc03fd3763242 : SUCCESS in 11m 59s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/b8f8c29e2db846c5a5541161fef82dea : SUCCESS in 9m 54s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/32d770eeedcf48e2b81c7201a48bb5a2 : SUCCESS in 41m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ed41121404c54b2b99255e54fb6b1430 : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0183419aae694e53916fa405353efac0 : TIMED_OUT in 3h 01m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a22fc2892e5941f8b7d0cac4c8fdf304 : TIMED_OUT in 3h 01m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a13f707291964199a8098831fca01ade : TIMED_OUT in 3h 02m 27s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/42e1a479d5a44e32a12b4da0b0f1204a : FAILURE in 7m 13s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bd7c4b262a5c4f1e82f81d18ce1db8d3 : SUCCESS in 1h 08m 25s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf0c9b0fdc4d4676ad7524cc53d203ff : TIMED_OUT in 3h 02m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96ea5c3a3f774fd29c91e0d29d0fcc34 : TIMED_OUT in 3h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"17b9d32db2180fe94e56346eead552d452325422","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-02 16:04:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c8f2f6c7536cbffd5872f95867dda5a492609d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:21:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a5dd6ed53f949788d3228e950a3eb88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd26342c6665484d9bb5d5c06765b2ef : SUCCESS in 5m 09s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/638882fd428344719f045301c152b226 : SUCCESS in 7m 25s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/4be1d6c068b64869a6d7d8bca5e40818 : SUCCESS in 10m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/505b35089b24475ab4e7fd49e8233a6c : SUCCESS in 10m 41s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/62916fe71a874bfc8b37fd3ee2a61e3b : SUCCESS in 9m 11s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4cd9f57e287245abbf4f2167571e8499 : SUCCESS in 51m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/46f10c32d5d24b27ac14a1c58e305bf1 : SUCCESS in 43m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67fe1b24d99d45dc89be7389ebb62cf2 : POST_FAILURE in 1h 01m 55s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8cba1706112478886d7a8a6f0240761 : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/184a09a00e3f446faaea1ed50d6735e0 : POST_FAILURE in 49m 30s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1747d567cffb426aacd19d866fd1c8a3 : FAILURE in 10m 59s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/dfea8db0405341e59e18ee096f13c9a9 : SUCCESS in 39m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c0b2ba53cf74e92826f1f206d2d0fc3 : SUCCESS in 39m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72bf168462384f98b440c36de6d0c674 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c71866f6d8a30ba6446c93f88c1b54588ab3d134","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-16 08:23:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58e422e4ca7c843c9b374eaa99e2c52d99fc6fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:27:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ddad91b23594d1f8b70abe2a5be81a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0708717d7ea740c2a4d9ab01998e148a : SUCCESS in 7m 11s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2350ea1daacb49a38817913badf956b8 : SUCCESS in 9m 09s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a644fbebd3f5415fa248d2036f378941 : SUCCESS in 11m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2845223bc61544da909ec586cb74bdf9 : SUCCESS in 10m 19s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a54a3cc3a3e6404994f7faa66ac3afec : SUCCESS in 8m 55s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/138c2a0ae9eb4e98a6a0c43f7fc7966c : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0a7198bfce9f4d909a9bd0fe02cd3ad6 : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4965761fbac48feb69556f45b4e0a49 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cd754839a084299924f65bc4bfbdb69 : SUCCESS in 1h 05m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ad065774f9b4fe9ac62fde6318fe230 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c01ef56f0eb04d8eaabdf626946e252a : FAILURE in 10m 01s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f69dd575b0bb40e28aa6a1dc95accf96 : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/468ec6e0295840d2857bd0170401b2fa : SUCCESS in 1h 01m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7da4a456706741b7aaea8c2752c6f10c : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9a3faff265480406104292a92f96924d54452fb1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-09-16 20:31:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cfc7af730fd9265e251ab22b3fc5776e7162f6c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-18 12:00:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae51ee061ee0004184634d57ebb90cd72c523f97","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-09-24 19:42:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e991a0c37209b43be3add885fd4ca93f9bccaf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 19:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6370f09db88b48147bde2d49d9b6af20494fccf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 21:38:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a00f6fda4b4c229c93b3a9b55cb6f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab752b35a7ad4b0ba20537d3e62ca327 : SUCCESS in 5m 36s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812990fdd23d4d698b057e75ec942061 : SUCCESS in 10m 26s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a4552830e854d80898db865463fd4e7 : SUCCESS in 11m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bba5500576484263ba12e28bf7858bce : SUCCESS in 7m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f4be0897a1334627940d2c2f1096fbac : SUCCESS in 9m 03s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1ed677728476492d8e5c33c8ac8137fb : SUCCESS in 41m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e003899d73b9458b97c59b63d65ff50d : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6271ec7173e492b84b39400ea445a73 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dba7b6027ae46b38b1a36b85601d6ac : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef59c4a46c3841e2b52a5e7cdaee8efc : SUCCESS in 1h 22m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/07ef7af238614e7e916541bc2f9da6a7 : SUCCESS in 1h 11m 17s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00e5c470688b4012831167e830b739e3 : SUCCESS in 1h 14m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba76f9e84ebd420a84ef29f001a24a67 : SUCCESS in 1h 49m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6cfe63ddf82e8824de5eb2802cf1d326084df3be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 21:38:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"791d13cca2ff906972d27373aeea7b3dfb0daf1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 21:39:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c258310b16484c91a48624b37d972de7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be1d038851648c1b70604234bcf8815 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d887be4b801f3beaa45897d8aa4717e7b7008a58","revisions":{"d887be4b801f3beaa45897d8aa4717e7b7008a58":{"kind":"REWORK","_number":1,"created":"2025-09-01 10:24:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/959027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/27/959027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/959027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/959027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/959027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/959027/1"}}},"commit":{"parents":[{"commit":"67f506eb73de711dfb1fe67861521395d41f6c5a","subject":"Merge \"Add mariadb-client-utils to packages for removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/67f506eb73de711dfb1fe67861521395d41f6c5a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-01 10:24:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-01 10:24:06.000000000","tz":120},"subject":"Use loop label for deb822_repository","message":"Use loop label for deb822_repository\n\nThe module has quite massive output due to included GPG key\ninto the structure, which makes it barely readable.\n\nUsing loop label on the module will limit output to necessary set\nof arguments.\n\nChange-Id: I6aa4ea1b0c4151fdde2dbe26c02949cd12b77e4c\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d887be4b801f3beaa45897d8aa4717e7b7008a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d887be4b801f3beaa45897d8aa4717e7b7008a58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
