)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~960387","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Icbfe298cfae63dd9ad15d62d165357cabd164fe9","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-05-30 17:52:43.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2026-06-02 00:41:39.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-06-02 00:41:39.000000000","reason":"Change was submitted"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2026-06-02 00:41:39.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-01 21:33:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-01 08:46:30.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-02 00:41:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbfe298cfae63dd9ad15d62d165357cabd164fe9","subject":"Refactor compressed backup creation","status":"MERGED","created":"2025-09-10 14:42:27.000000000","updated":"2026-06-02 00:42:33.000000000","submitted":"2026-06-02 00:41:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960387","meta_rev_id":"69409dcd707f346f973d70b9c24b7ce4a98e01b9","_number":960387,"virtual_id_number":960387,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 00:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-05-26 14:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-26 23:11:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2026-05-26 14:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-10 17:46:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 18:51:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-11 18:51:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-09-11 18:51:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-09-11 18:51:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-12-12 21:29:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2025-12-12 21:29:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"13aecdce7e9d8d6729561da0b4f25a7ddd14cd82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-10 14:42:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b15d833358bb645dd3be88a253c673a04701610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-10 15:40:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fa4eb8f1f4a99c071dac3ac86e6ae0908e20d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 17:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13dd632fb8954abd85e3e415adc5c061\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb6f75ba34f4672ac02e0a3e2747b3c : SUCCESS in 5m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2885930087754591a656f40b0a05bc28 : SUCCESS in 4m 14s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0e0aa9fce484d86ae247c8c7aa06c04 : SUCCESS in 9m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48bc6a833e4248f2a712f3fdfb541202 : SUCCESS in 7m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/91337bc9ca914ea1a932cbcd39f06768 : SUCCESS in 9m 46s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e935553b4437484faf0f5aa79ab56544 : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b63f07c42c64bc1ac5cd94bf94d051d : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55388025e85d4c098cb86d9390884e63 : SUCCESS in 1h 18m 45s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93cec88082e84514825bc64799e9bb84 : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be0a85a5c5ff4432a7a4a430d43a9533 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3a0c396115984fdd91f0d24e0ab2dfbe : FAILURE in 6m 51s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bc9124446f774617ab8359a984efd506 : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7da7715daccf4bb88372f20c5fabbd56 : SUCCESS in 1h 15m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91b120f0f3a74ee2a5bbe6aef6944df6 : SUCCESS in 1h 58m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f88b1a7fe11723e6c68aed89ca4517e02dd01849","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-11 18:51:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6209e0c45dc0e974319ab444590ab33944527d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-23 10:53:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"48c276d06e4e094b9f7a7aca64dfe0329cf2834f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-23 10:55:08.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"261a0aba01ef79a641d949a58abab9d387d7da84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:31:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63fd588f16ff4361b6f95ddb9531ff4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d00b893f5ba4f3eab72d48206ba5ee8 : SUCCESS in 5m 14s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/303875524e1e4b608e3d69824c646769 : SUCCESS in 6m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2008efabe66143e293cbed1288ce9ff5 : SUCCESS in 8m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6898eabe61449d5be285b4eda072bfc : SUCCESS in 13m 36s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/32e3008f56a44cafb676dc2313cebd05 : SUCCESS in 9m 19s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f1a11691b084f74ab0e4998b1094b49 : FAILURE in 4m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c4937e625c524e0f9d57f6ffbb20ec80 : FAILURE in 1h 15m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d60ba7f63acb4a8fa7b7cc629ada2733 : SUCCESS in 55m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/90a937a950b9418bbece63d9ddc8e0ff : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13a7f578d202439f8331648e0bd5d18d : SUCCESS in 1h 02m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e0d2be6bc0a47bf8647119da3c76e7e : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93b37c8872fa4eac92af7d0c82a04f74 : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34cc5946cfdd4c35bf6ddafcfc4a080e : SUCCESS in 51m 41s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/892c84fbdf1745b584d970a8c26dd08a : FAILURE in 6m 59s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/643d6c188a5646c883ae8cd8a67b178d : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/662f8652946047c494f35bc3ec76db0e : FAILURE in 10m 48s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c75b970873c743b791969461da6c6064 : SUCCESS in 55m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5043ccb9464142beba318012bfee5551 : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":4},{"id":"a5c4830cb17a66e6aca43e4dfcf51fdf6baff7c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-06 14:16:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"71d567798e698b7d802da76267bbff1c847b9b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 16:29:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71aace22f18f4cf29a482c55b2ab7269\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dac9d67b8554f468b4f186f5bd90a7c : SUCCESS in 5m 46s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4364e1b525bf44ed8f7c0df440f48809 : SUCCESS in 7m 17s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e9708ef859bc48a6b84085bcc7016c6c : SUCCESS in 12m 27s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/51454463982c402c888af2b6e0b7d59c : FAILURE in 4m 30s (non-voting)\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a64945e21a1d4c5f9ad01f9460c7adc3 : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/44545ab133bd4ee1be526dd9ac6d7cf8 : SUCCESS in 10m 02s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d60ad354e8349f4a66b7f32fc6cab6a : FAILURE in 5m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4abf37505cd344049ed78765a70abf99 : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5fdda745257d4948bb2a99001bdd6780 : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/093f048dded544c3821ab40a1ed9b666 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c2f218fd538140eeb316975a65a4f51d : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e7fcac1a64f9429f8fed38e9cd7ee7ec : FAILURE in 35m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3e7329051d04bdbab4bdf3a48518528 : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a76b950de894437fa4e3f689977fb362 : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73446847fb474a0dab6a0e4c7d524c0e : FAILURE in 53m 51s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef2262598cff441b958bed3618e9ace3 : FAILURE in 10m 39s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6207414c4ca344fe8630f7440e1f752e : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7b05874788de4c7b8e1f951ddd394c15 : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b2cb3f4844ed4f659c7a806b13967e86 : FAILURE in 12m 20s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7a523a3b3ef42f08f16059a0d71f3eb : SUCCESS in 1h 05m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d908c8f211b44b9e9723a203265e873e : SUCCESS in 1h 54m 44s","accounts_in_message":[],"_revision_number":4},{"id":"f2d264b20e69c82587798de8a91064aef53d5b45","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-01 08:46:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2aa6b1fe017ff932acbf0302090db01b2485cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-01 11:46:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fc215c698e16cf7741083ba724b32899a3fe591d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:14:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7204ceabda4a76b0fb060556812a52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18ce11a81cc437383fc3e4c955e55bd : SUCCESS in 4m 52s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb9cdd8bcd0e4af9a2bbba602da4db5e : SUCCESS in 7m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5335d008181443e3a0c22e93e507d44d : SUCCESS in 12m 42s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/664dc04e602e438bb16f1eb70a77a2a6 : SUCCESS in 12m 16s (non-voting)\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d47273c9969f488cacb7f31d2909b57e : SUCCESS in 10m 36s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f6a0f7e4718949f4b04b5687a2510c94 : SUCCESS in 9m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/38dec9ba7ff74674b34a7c779fe0a9c9 : FAILURE in 4m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b11286a89c0a4dd4b09a9106cde2fed2 : SUCCESS in 58m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab93b7bce68f4e4393425d9a4c6a410a : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d86e0d78963749c6af9beb034d9516bd : SUCCESS in 43m 14s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/34b40d7a398d4d4a885317d53bc56a9d : SUCCESS in 1h 00m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ccdec9cc355542919b62c40dd7bd5612 : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abcc9b5fb0574f26a35b2fab1beeb3e1 : SUCCESS in 1h 18m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb1f7f7d1de74acaa45775d6cf0c0c57 : SUCCESS in 1h 11m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41287dd4bc4a4ba697eb1c9d25b7eeaf : SUCCESS in 1h 10m 03s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/45ed759505f54299bdea75a48afe3537 : FAILURE in 7m 33s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/26a930e60d854a988ef0cd3e95f7da0c : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e860e607181343309dcc00c29e104148 : SUCCESS in 1h 06m 13s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0eb0b090399c4fa6a99984a0995d6894 : FAILURE in 8m 15s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/236ddff85f4c4745b22e63d3e311cd62 : SUCCESS in 1h 08m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0f31b1ba328487aaa4dacaeda14f9a5 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":5},{"id":"54324ed9631931886312aa09889050c7b6626f99","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-01 13:14:58.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"801b25ce0fba30c53a544028ee2ba4faadd1157b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 22:08:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7a131807b9b6e421b1de830840b74d8fab0cd45b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-05-22 08:06:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"63cb723bba189517bcf6e835ddd3aa4734311ead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:09:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5e48f4c1f950aa4ed65bda7639d7ee8626cb7a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:49:33.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/514a577411d0479ba05712a048e6c676\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba9322d940e438fa24d966e411edf25 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e93d8c850f84e578e5c11a2a85ff73f : SUCCESS in 2m 29s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2afaf6fdc4d741e89da71358a858e6ad : SUCCESS in 10m 16s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/d820084a22694115aed4448bdac16caf : SUCCESS in 9m 35s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d124ee38cc244b22a7f71e08800deba7 : SUCCESS in 25m 15s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7bf00796bb24c3fbdcd8a012ce0df28 : SUCCESS in 12m 35s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a1110fcb1d2744e1b25923c632d71b27 : FAILURE in 4m 29s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a41a277ede214767ab6ffe2d2e40e3aa : SUCCESS in 10m 13s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/db7258d1635440b7b9fcc196aa68161d : FAILURE in 2h 08m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/63f80e4f365543558b9c438cc46b4def : SUCCESS in 1h 31m 26s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/02497c991c174ede943d4abf73b0aef4 : SUCCESS in 2h 05m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a045d7efff2041d095702583b9393206 : SUCCESS in 1h 41m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/257a87bfed9443ed9a7acffa35d81f75 : SUCCESS in 1h 51m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ddf984f7c247649296b6df59d2ffc3 : SUCCESS in 1h 43m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5c3331e9c9a48cb920663bdd8c3d9da : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/146affa528f34168a8afce90fc004fc1 : SUCCESS in 2h 37m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/57eee4fd8c3d417a861634401f883c6c : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/df3184a0d6c84d44aa7b218b798c0180 : SUCCESS in 1h 55m 57s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce8a245741f94f14a39a7a16d29b207a : SUCCESS in 1h 02m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0517d71cf60a440c80f426c27d416913 : FAILURE in 1h 15m 56s","accounts_in_message":[],"_revision_number":5},{"id":"ee6f0d53f1ee9ce0ac77d42a67372d70d28c2554","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 21:09:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"325495bb1bc4d7246c8d39c752174509b2e5f7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 00:17:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f885e9152e7455bb45109148f25710a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34225c490cb4638ab75b5bcd1bb6fcd : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/039af6136ab042edb34b4bb8761ed08c : SUCCESS in 3m 32s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d18d6f4d7bb24dfa8e9ba915dc2da679 : SUCCESS in 15m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c788ff33eef94ed9b475690b815e4bee : SUCCESS in 14m 18s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/7b66163c11104cf0b971b07664ce96f3 : SUCCESS in 11m 12s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27d7029b20044c52b3cca8f9e29e8c9f : SUCCESS in 12m 47s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f41b8240ff6f4c2a91797171e4a07e18 : SUCCESS in 10m 18s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/cd346fc3f3944601acb75b3ff64a3932 : SUCCESS in 9m 37s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/32ec8d5c193c4a5ab7fe4ee9fca58888 : FAILURE in 5m 57s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2c0b934210884118945cce8de221c2fa : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f7ded327b2e431a9e2aea484abecbe4 : SUCCESS in 1h 52m 46s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c7c5f8efb2474ce2a90ff59ff2b6131f : SUCCESS in 1h 31m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ea72151880e242af9d9a7f0ac3375d68 : SUCCESS in 1h 36m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4f2d5be41caf49a3829c7700d8e992f5 : SUCCESS in 1h 47m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eddad0be5664ccfb84be45c2e3f31fe : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47abc7e668ee4e7b8eba9e5d0e405de9 : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb06017dd66a460e8c667422a9123a89 : SUCCESS in 1h 49m 46s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1ba63bbacddc4f82b6e3623f640b720c : FAILURE in 8m 48s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/85e32c271f5144bf83f186368a0d09de : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e31891c9009345d6b5ff427193a27858 : SUCCESS in 1h 45m 26s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/681bc28b2f1e4a3ab86358e3a5a77dca : FAILURE in 11m 58s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ddd9d153ef14bb09daf973701ce7ae9 : SUCCESS in 1h 08m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066663524ce146ca8be38cb47501f7f9 : TIMED_OUT in 3h 04m 30s","accounts_in_message":[],"_revision_number":5},{"id":"1602a3906dd47eab39e0cea04b0202a98c92758b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 06:46:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"81e0e39e597ffdd0c9fa740d36e502afb2de9823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:09:54.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8287c0b3aee243968c442b4bfe02cd99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ef01054aab461aa6238c02b92662d5 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bc2b5300e6b47259e1478a90c02083e : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c999b5b613eb402aa2a498249937f42e : SUCCESS in 11m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a249b57c16241659b6ae90847033f86 : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d1bfdd6948c24af58896fde3dd72e8b3 : SUCCESS in 12m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b14d984c11f45f38014041d06f71e87 : SUCCESS in 12m 22s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/307c34394c8f4405a973a8eb42e5e5d1 : SUCCESS in 11m 30s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d60b83c8d12b48458f47270a78a8ea83 : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8918918c1515438b96013bf9c1e2a842 : SUCCESS in 7m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d7f50cfbca4441aa9b59f67f5fda0949 : FAILURE in 1h 59m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0b197a1317df492b881a67423cb11331 : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6eb8615a25ad44d58735b8119abaa62d : SUCCESS in 1h 51m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/223778c8bd774ad0b00c9d461710259e : FAILURE in 1h 23m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/93e622f6b6a94d6494225bd85ae30202 : SUCCESS in 1h 45m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55eb999813af454bb1c18c149e7fa66d : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22ce433a9adc46acac0a88769460223b : SUCCESS in 2h 03m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ade340a0e304529a03c81cce4a8693b : SUCCESS in 1h 58m 34s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/64c98a3dd8d24b01bc6046bc8c8f46b1 : FAILURE in 17m 55s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3af55b5ed2d447a79779e6db758be8a2 : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d4792ace898c41bb91ad4c7e021c33da : SUCCESS in 1h 41m 46s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24755f05c7c744b18185ade967adca23 : FAILURE in 7m 18s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac25b26c0fa45519e437532a92863a4 : SUCCESS in 1h 47m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc612daca55b43be939d5ede93a6b450 : SUCCESS in 2h 16m 37s","accounts_in_message":[],"_revision_number":5},{"id":"4f4cf3296770fc317ea9748fa7c89564848fb305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 14:02:09.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"72ecff447dd44945037a062e50539aed1d852787","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 14:02:13.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"56aea6072dfed54d5f2af6c94012fea023dcd5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:07:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45fd2375ae8f4d13978eb7097f28475e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078ce4b1c3e54ce1a1faea798de46a0f : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a21ad0aff0514e8abb085c8d0ea1cb64 : SUCCESS in 3m 58s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd5a3c05edf9489f8802fc272656054a : SUCCESS in 11m 31s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c310baf02dd493b83dee8c1b1eb3ff1 : SUCCESS in 11m 51s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a00995015534499bb4dfb9e54901dcb8 : SUCCESS in 12m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/498b9819975748888061da582996924a : SUCCESS in 12m 31s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/44debbd159194bf7b54fb6012147cfb3 : SUCCESS in 13m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/45003e24215c462697ba7edce075f86d : SUCCESS in 10m 32s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c3bcaf2022a74ff2989695d0860cb551 : SUCCESS in 10m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/67a579db04ef48bd98223dff0df34ca1 : SUCCESS in 2h 58m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/41055577098443cfabac829874fdfc13 : SUCCESS in 1h 45m 44s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0395d2d23557434ca4c9615ac0fde131 : SUCCESS in 1h 55m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2331a0854e8147668200a185292375d7 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/15d7496760f1425e91e079b7773375c3 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb50bb3dd8234be39c4677e66e4fac4f : SUCCESS in 1h 58m 20s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af33dfc4fc014e37820309ad0f145bc3 : SUCCESS in 1h 50m 44s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5e3b04440224992b52b0ad2ee7f519d : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8002d5f0eef64b5fbf2851ed4444b6dd : SUCCESS in 1h 51m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c10687777b8b43459f0d1144154acf0b : SUCCESS in 1h 52m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec83058cfe284fd9adecb942ec705962 : SUCCESS in 1h 55m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d5e7beefd224691b53c17bc4e0b02d5 : SUCCESS in 1h 57m 44s","accounts_in_message":[],"_revision_number":6},{"id":"58e554530d0598cd217dd575df327cb3f0e1d82b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:10:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2a2cc4ad2353046a30506bae73370944d396a4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 19:32:30.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/390f7657b3b14a6e8a1c5d185b1e59a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc465688c69447086840d7b077680d1 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa7a71c5b9cd4c85bc0e46d8d7e5db67 : SUCCESS in 2m 49s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b4e845be631417ea1a32cf7df4e4c0f : SUCCESS in 9m 16s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/4b486a41eca14d31ae499342f542cc40 : SUCCESS in 11m 05s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/83f0ee34cad8428eb64fc54a93963e0c : SUCCESS in 9m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/813289c81dd943dabd0c753db4405f7c : SUCCESS in 6m 42s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/893d13c276bd4597bc384a3357ba5b38 : SUCCESS in 9m 55s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e3813348eab1493a9060a95b76207b32 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9b00114d9814496c9a7fd562a8ee0e79 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b21b460f46d4f34b27acb43b8c9d688 : SUCCESS in 2h 16m 23s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/23167f54eacc418884216aa5e2ddcb70 : SUCCESS in 2h 05m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b5c7752c200b473b8640295d97c3ee1e : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e50438a67374449790409f73e3f716ae : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd26244715c24f68a664136a7e805d23 : SUCCESS in 1h 33m 24s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1ac840a4426455697345ba031260b27 : SUCCESS in 2h 04m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ccd4e68e3e94cfebe6d6d5427804e0f : SUCCESS in 1h 58m 12s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/394447f285b9494a96f6430ffed026b3 : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/699d0fec5f674a6d96bc32187bc0300c : SUCCESS in 1h 47m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67367d1f326f4be88c2e0cf97e9c8605 : SUCCESS in 1h 40m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f39138dc3af4dbf9142d73a7e9e100d : FAILURE in 1h 22m 05s","accounts_in_message":[],"_revision_number":6},{"id":"f3c312899fdc7f15b22b813c36fa50c9e138a625","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-26 23:11:27.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"308f60c266db76f05d350124da11165a5e63cda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 01:49:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f35625ed6d24975af501a204a19e25b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7037c2f609f4cbfbde06f4118d2924a : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b85b228113b4b14958b96a8511043bb : SUCCESS in 2m 45s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6341e388d7e4511947fdc29d9f48e4c : SUCCESS in 8m 15s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/75e2fa3a479b4019a5dffa604751956c : SUCCESS in 9m 34s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/121cff39c1ab47b8a006627db2aa1df9 : SUCCESS in 6m 07s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d45480eba0249f7b75dd92364d46cb7 : SUCCESS in 7m 28s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b71c9bb6f4514d81a0789a00e2d27eea : SUCCESS in 12m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d7964084335f4227b407dc80c4daa25f : SUCCESS in 9m 52s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbfb9f6ce6d9487e8eabacb8dc3b81f8 : FAILURE in 4m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/90b0a5cb72c748b389b8df959717a3ba : SUCCESS in 1h 06m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/afb67803738545ab8f859968d37b3e3c : SUCCESS in 1h 23m 36s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2938e9b11ff646999e818e2a8b83d89a : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42a45e9514bc4dce8b987807108fb79d : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e37d6e8154f944f990ca6b1b1148081a : SUCCESS in 1h 32m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55b367aba9ee463c92a998ad36e2e14d : SUCCESS in 1h 51m 14s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a45cab340f7a43eb8714735b170664cd : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8049c98d51f471ca1000ef008b27539 : SUCCESS in 1h 52m 15s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/985869cba623452688ca039560dede7f : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b5a2db77efb24e9a84b314a2a1a5568e : SUCCESS in 1h 51m 14s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1db17f4d5e6441cb6c16414474017a8 : SUCCESS in 1h 36m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7ce692a2b6e4caa910bbc99a42f130a : SUCCESS in 2h 29m 49s","accounts_in_message":[],"_revision_number":6},{"id":"c7b5dafca82f784a6aee41c593364fb7b40cde2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 01:52:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cc455323ff53ab90ed63a1f1d3c9de5cd8e8c458","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 04:27:47.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76d6903e5d25466587b3ebf1b22b52eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db5522f3adf4cff8df9478eebae2068 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eaad6c8c2a74634833623e79d8099ce : SUCCESS in 3m 28s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70b7c34fe6c34933b07cc97ba5363b4f : SUCCESS in 6m 08s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/922bab9648954208a0adf782ff7d6e1e : SUCCESS in 10m 36s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/91ea0ebc3419474ba17d9cdf38d4558f : SUCCESS in 9m 19s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7989c7fb9ece46d9aba3567604b24d65 : SUCCESS in 13m 14s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5d835668feaa422499c2dc295ffb3616 : SUCCESS in 12m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f21346215894c65b967824982c90013 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/349121942fb54103b0a03d0d468827f2 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/88e6b86a09844ac884a0ace8ebbef833 : SUCCESS in 1h 50m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/98faeb46477544f298b9f10399c5ad47 : SUCCESS in 2h 28m 59s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/47ac60b073b94708a469da99e0801f37 : FAILURE in 59m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24eb90d71c2747658be2000851218a37 : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da6627cb821c40f690b9f9c373475868 : SUCCESS in 2h 28m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b286702cac4a4453bffddd993e42c53f : SUCCESS in 1h 58m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d04f919a9b344a2bb52ef7948c025bb9 : SUCCESS in 1h 53m 09s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a191fb50c07b457394cd18c007112ee1 : SUCCESS in 1h 37m 08s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/63af348236f14ce683b67001235a205f : SUCCESS in 1h 41m 16s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f03e83e8b2542fea6c1665ee5ab5346 : SUCCESS in 1h 06m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4436a8befbf842f7bf9f5e74c6ec6166 : SUCCESS in 2h 19m 18s","accounts_in_message":[],"_revision_number":6},{"id":"3c6f6cdb3f02059369171b998b4af26d2512c137","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-30 17:52:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed6912d384550eda94bb60705d00806a3ece63cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 20:15:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bc14eb5c1e24098a37a1c325244b8cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8200bb3f96840d7839376831e442a5b : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d6b0cb3892b4cf187e6e8bc2dcb32ab : SUCCESS in 2m 17s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bee7f32932244b9f90972004968bb976 : SUCCESS in 10m 25s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/11de8cffd8ca421b861f835b5ec4ae9f : SUCCESS in 11m 36s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/781b4b1136a7467f9ac17b20fe25387d : SUCCESS in 9m 25s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81759c73007545879a6f3052dd77ea19 : SUCCESS in 6m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5d448a970a2a4cc5bae4a0302183a7cc : SUCCESS in 6m 16s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc3612b3a90445f792e4774559a20699 : SUCCESS in 10m 20s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/40931ea077b94862b7702dac7485186d : FAILURE in 4m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/941eb4fa7aa5457196a1edd706524068 : SUCCESS in 1h 11m 56s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9538a7cee3c94b60807c9ff872e03fda : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/49fe322a7dc14e38bef7a2559f78305d : SUCCESS in 1h 54m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/76634f7fef994c28a45085fc28979529 : FAILURE in 40m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a45b69841b5047d4a65b0a1f09fa64b7 : SUCCESS in 1h 36m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfc4ae87961f443e9826e25d40839080 : SUCCESS in 1h 46m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f47189b470714005af90dee8f328c7ad : FAILURE in 1h 22m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/165244de202046fd82861b607c211cb8 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fda629d97a814f7485327f0441e9924b : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c55cb71dbbb145dfa7977ccabab0d53b : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cb1706aa9574c708267d98dc3ff8821 : SUCCESS in 1h 21m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8370343eb3d4ec3bc740e804a044289 : SUCCESS in 2h 13m 54s","accounts_in_message":[],"_revision_number":6},{"id":"b836be002a5ec73c08bc987d5eaccf7aef58c882","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 10:13:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"db8ff7e028964c3ec64b6091fe6ee04b3d219d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 11:46:37.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2da620e128a34b019f6260e71fcb807f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0a40f1651e4f4a932acd51d72fef7b : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46788b4efb8142acb4121e24c2dcbcb6 : SUCCESS in 3m 24s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f2c3ce76e5e431bb9f1836179e78ef2 : SUCCESS in 8m 44s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/121fb53eed934afa881128c18a68dd85 : FAILURE in 6m 29s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d1ca8de7eadb4045b8c8e8f7705401c2 : SUCCESS in 7m 40s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d10492fefd44dfa923d6e243c52b183 : SUCCESS in 10m 34s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7b1e85c704754400912d20dda84ccda1 : FAILURE in 6m 30s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/9b45f9656f264a76a8838105b2b902c2 : SUCCESS in 9m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54bae426d7824b70a2dd529ece592bf7 : SUCCESS in 9m 59s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4ccd394dcba14b468fe4a9a95840c23a : FAILURE in 1h 05m 10s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/763509277cdc46f4baa5f682c346ad3e : FAILURE in 1h 03m 29s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4eb79b9648c4abab402ef3f52ec1405 : FAILURE in 43m 23s","accounts_in_message":[],"_revision_number":6},{"id":"2e73a61b21aeb3004964d095dd5f9f17d23c81d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 21:33:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1a55bbd33cd7afbcc4c6ff81e5665d22d8344e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:13:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb98428f0464af7b809c855642c787d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494512e0566e45ecb0735c6e3cc00aad : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b53e580d085b451f8cc55ea7727d2174 : SUCCESS in 4m 37s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbe45f7004a844439e94b794471e3787 : SUCCESS in 13m 03s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5f918eb4bb84452b4e8146e31bb5203 : SUCCESS in 12m 39s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/040a2f16a0d94bc79e50afdc6cf18f7b : SUCCESS in 11m 11s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aa6f7d1f2984029a0fd262192f5d512 : SUCCESS in 12m 27s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/33418dfe422b4da291e7f0580ad12752 : SUCCESS in 14m 47s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/898b65fe653d41269c2c7692e4637d9d : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c3569c5a7c35423e95dabea4d209ff79 : FAILURE in 5m 36s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b6ec2319998049ed96f1cf4b4399580b : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f2b3bc67cddb4f29bc99d9b94b614edd : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7b4f50349b04283b1b8191a4c110a3c : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":6},{"id":"80185187ba6a6e797dbf76a1fbd17eda0bd20563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:16:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"015606efe84254e37d52ef031f14e9ff530c0a79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:41:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/652daf3a33674255ab5553f05d838dea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccdd866867f6424eaaeea7849c2c4c31 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8080b2763014839adb90dc6ddde11cf : SUCCESS in 4m 04s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f6cebb1556f4974a33b32f21343fbb5 : SUCCESS in 8m 52s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/03e364a0c05b4761906de2d2ea98a2ae : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/5ca389a191ab4306b9c9f28ced7c8787 : SUCCESS in 11m 34s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67f66d8397b74a98a0f9b7c1659b0204 : SUCCESS in 12m 33s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c441005cc952448d92fcb234f46be64b : SUCCESS in 13m 04s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5cb3853838424a64adc1bd46d5d959eb : SUCCESS in 9m 47s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/eadf2e94c2b541e888933dfd3383186d : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a7cf2b8a9d8c4c6f935626fd0c30c80d : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86e1b2addf0a42d2973425275097e38d : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":6},{"id":"bdd45ba1ad4edfb2b998758ad8dca8d32dbace92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:41:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"69409dcd707f346f973d70b9c24b7ce4a98e01b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:42:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b6d630b9d5f48009a70c95506dd6d73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621dd7939c5f499abf7b3ac4823a171a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ff1562ab4a54f15e46e9d639f1205267786e704f","revisions":{"e455214750773f579b8ae29f7fff41ddc887d196":{"kind":"REWORK","_number":1,"created":"2025-09-10 14:42:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/1"}}},"commit":{"parents":[{"commit":"9d8e49c6f8dd42a4959775b20660ad4f2171a4f9","subject":"Replace normpath with join for backup script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9d8e49c6f8dd42a4959775b20660ad4f2171a4f9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"subject":"Remove Popen wait() for compress scenario","message":"Remove Popen wait() for compress scenario\n\nUsing wait() with Pipes may result in process deadlocks [1]. Just\nhaving communicate() should be enough.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e455214750773f579b8ae29f7fff41ddc887d196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e455214750773f579b8ae29f7fff41ddc887d196"}]},"branch":"refs/heads/master"},"9e032d13a6fab89a4ec9fc8c73afde948d83e185":{"kind":"REWORK","_number":2,"created":"2025-09-10 15:40:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/2"}}},"commit":{"parents":[{"commit":"9d8e49c6f8dd42a4959775b20660ad4f2171a4f9","subject":"Replace normpath with join for backup script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9d8e49c6f8dd42a4959775b20660ad4f2171a4f9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 15:36:55.000000000","tz":120},"subject":"Refactor compressed backup creation","message":"Refactor compressed backup creation\n\nThis aims to get the code more readable and easy to read, while\nkeeping functionality as is.\n\nIn addition to that we aim to properly wait for processes to end.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e032d13a6fab89a4ec9fc8c73afde948d83e185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e032d13a6fab89a4ec9fc8c73afde948d83e185"}]},"branch":"refs/heads/master"},"6abd1806a94574513245eca0759821bc4cd5e1f4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-23 10:53:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/3"}}},"commit":{"parents":[{"commit":"ad0cec3355b3430ef7f964803c869342b3749765","subject":"Replace normpath with join for backup script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ad0cec3355b3430ef7f964803c869342b3749765"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-23 10:53:53.000000000","tz":0},"subject":"Refactor compressed backup creation","message":"Refactor compressed backup creation\n\nThis aims to get the code more readable and easy to read, while\nkeeping functionality as is.\n\nIn addition to that we aim to properly wait for processes to end.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6abd1806a94574513245eca0759821bc4cd5e1f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6abd1806a94574513245eca0759821bc4cd5e1f4"}]},"branch":"refs/heads/master","description":"Rebase"},"cc335669905fa935147a05b8be76d983ed25692a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-10-23 10:55:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/4"}}},"commit":{"parents":[{"commit":"ad0cec3355b3430ef7f964803c869342b3749765","subject":"Replace normpath with join for backup script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ad0cec3355b3430ef7f964803c869342b3749765"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-23 10:55:08.000000000","tz":0},"subject":"Refactor compressed backup creation","message":"Refactor compressed backup creation\n\nThis aims to get the code more readable, while keeping functionality as is.\n\nIn addition to that we aim to properly wait for processes to end, as currently\nthere potentially can be corner cases.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc335669905fa935147a05b8be76d983ed25692a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc335669905fa935147a05b8be76d983ed25692a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0f4a9846db6656bbc3747f7dfc17ca3288ba8755":{"kind":"NO_CHANGE","_number":5,"created":"2025-12-01 11:46:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/5"}}},"commit":{"parents":[{"commit":"56f70e5a304452abda779c4b532c14258bd9d343","subject":"Replace normpath with join for backup script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/56f70e5a304452abda779c4b532c14258bd9d343"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-01 11:46:51.000000000","tz":0},"subject":"Refactor compressed backup creation","message":"Refactor compressed backup creation\n\nThis aims to get the code more readable, while keeping functionality as is.\n\nIn addition to that we aim to properly wait for processes to end, as currently\nthere potentially can be corner cases.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0f4a9846db6656bbc3747f7dfc17ca3288ba8755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0f4a9846db6656bbc3747f7dfc17ca3288ba8755"}]},"branch":"refs/heads/master","description":"Rebase"},"ff1562ab4a54f15e46e9d639f1205267786e704f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-26 14:02:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/960387/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/960387/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/960387/6"}}},"commit":{"parents":[{"commit":"d6c4f4332a96273b5d41d4c030823bcfd343e961","subject":"Upgrade MariaDB to 11.8.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d6c4f4332a96273b5d41d4c030823bcfd343e961"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-10 14:40:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-26 14:02:09.000000000","tz":0},"subject":"Refactor compressed backup creation","message":"Refactor compressed backup creation\n\nThis aims to get the code more readable, while keeping functionality as is.\n\nIn addition to that we aim to properly wait for processes to end, as currently\nthere potentially can be corner cases.\n\n[1] https://docs.python.org/3/library/subprocess.html#subprocess.Popen.wait\n\nChange-Id: Icbfe298cfae63dd9ad15d62d165357cabd164fe9\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff1562ab4a54f15e46e9d639f1205267786e704f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ff1562ab4a54f15e46e9d639f1205267786e704f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"0f4a9846db6656bbc3747f7dfc17ca3288ba8755","theirs":"d6c4f4332a96273b5d41d4c030823bcfd343e961","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
