)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~970814","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ifc5c8a5c9fe00f2ad7daf8012814cf528fe3c3e6","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-25 00:07:19.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-25 00:07:19.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-23 11:02:32.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-24 20:29:25.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-12-27 20:54:41.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ifc5c8a5c9fe00f2ad7daf8012814cf528fe3c3e6","subject":"Remove outdate code for SUSE","status":"MERGED","created":"2025-12-12 11:08:18.000000000","updated":"2026-01-25 00:08:08.000000000","submitted":"2026-01-25 00:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970814","meta_rev_id":"37af1fb7c61c37e7e3ee1368ef74217092d95fc9","_number":970814,"virtual_id_number":970814,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-25 00:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-12-23 11:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-27 20:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-27 20:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 13:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 13:05:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-12-12 13:05:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-12-12 13:05:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-12-23 22:14:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"732a08ace79737ed41b6d05c07f981618f2f56e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 11:08:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41d2883080a76717e4271f40d833917de2e7e0da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 11:09:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"62a237b050581b4d182dc5adb7dc99acc38c7d0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 11:09:45.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cd2611913c24fe63e1016103390d127240d51f55","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 11:09:48.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"d2aa586fcbe5fd0e57ceaf70e75ee40cdc4e429c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 13:00:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f700efec1e8f4691aea6ca0b9c3914f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f031780aeaaa4341b71f1352ccaa12b2 : SUCCESS in 5m 38s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a23c78c554f4fef8ea8370660536e0f : SUCCESS in 6m 21s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e5209ab6a284a86974f65a9422158a1 : SUCCESS in 10m 27s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ab381a9408d44ae0a4528cccfb40775d : SUCCESS in 11m 52s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79a2c67186584400938c2d09b2e5ebd2 : SUCCESS in 11m 33s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/68e3f6c83f084052a794b2824e9fa0f2 : SUCCESS in 9m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e61b80d2eeb24a498b0246413467a74c : SUCCESS in 6m 36s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4ea61b87dc44418fa111ea748d1a2efd : SUCCESS in 41m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7046176a759c40db86b1992338d7b5c2 : SUCCESS in 43m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c8e87e8fad6f48dd8750f02cb5df8efe : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0415adafb474888b1468cc40ca955f7 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a7750665fe47483cb4257817e7947675 : SUCCESS in 1h 24m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcf101b24a544da3b67c13f3ef4f1be8 : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24bc855ef8274ffd9ce5bb9d3e6b4fc6 : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5a835e5e23f4fb0a2566555faee9541 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9279459babc5454684698ed5214463ec : FAILURE in 5m 42s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d8a6e3a3072547239547e83a23cebead : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d715acd47c7d47e593f95622282edb66 : SUCCESS in 1h 03m 10s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2e38b74094064487a6e29e2a33dc0107 : FAILURE in 10m 33s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cefbb8093acb4906832c49d451ad6d67 : SUCCESS in 39m 17s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9bba03c27c94984a3ec23d9d072ffce : SUCCESS in 1h 44m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c25b094e7aae3837e43800da1e3b851bcd807723","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-23 11:02:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"33b34b627b6e82ad1f95594c3d8704a97cbad912","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-12-27 20:54:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"205cc7d71654edd641b255c0fe1d5c89a83e4c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 20:56:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2333491ee3488f43c021d4e13734f2573bd4846b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 22:30:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/375d4be462e84ace94b5ad2f40d97924\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bd625cddce438d86bc5c031a592742 : SUCCESS in 3m 36s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/631d8078540a4b81ad9cdd1f7440a45f : SUCCESS in 9m 00s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5f94958151744e24b58782e38475e224 : SUCCESS in 14m 14s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d18e9e14c9854382b35207a9c4e51a68 : SUCCESS in 8m 58s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac5058d917f14bb49876af24a82b3cd1 : SUCCESS in 11m 11s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/02b0790f9d5943139dab6b6f9dbca29a : SUCCESS in 9m 24s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5a2a7182ba6b47c7aa51354d6b1bfe16 : FAILURE in 21m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/160223814a5547a68bac570656f03e84 : SUCCESS in 59m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/eecde0e2f8924926a28ccbdfd13cb3f9 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2570c8b0c66544b2883d8a9cae23d5d7 : SUCCESS in 1h 01m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6727a2009c754e42b1327cd619448b43 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d6832a14cc145cf81b2e783256a0b31 : SUCCESS in 41m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5768acd1df0744b68198c21946311599 : SUCCESS in 1h 14m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf6acd70b6c748cba13391bdf4eaf1f5 : SUCCESS in 42m 22s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4d24baae6a6543dd9b7c973e51581280 : SUCCESS in 1h 10m 16s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e5cbd7fdddaa46deb438af025b2199f7 : SUCCESS in 40m 35s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44ed00f8d0a84d57946e268c8ea0645f : SUCCESS in 1h 04m 00s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3020925ad5b342cbb4d70b151a823ada : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bbf133fcf129d914f459f1e5711f42432e2140f5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 12:26:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aef41694ddbc856bab2c11cb945d7f433e9dd5c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:04:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6e866a28a6d4d5885b4e016f5e8404f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4628f2b6d7774c3fb39ea98757491cc0 : SUCCESS in 4m 25s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/687aeb5c8aa04ea9923f3d8423ecff2f : SUCCESS in 12m 07s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ceca4be551b424895cc4808f67eac35 : SUCCESS in 11m 39s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b8efd9909a8a415490464a524529f93c : SUCCESS in 11m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ec99e5144164b88bbeb86f875deebf8 : SUCCESS in 10m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/96aa91f55149430b801d7578b6f268c3 : SUCCESS in 10m 48s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/723c6979b26c414abbef204d82cf97f0 : FAILURE in 4m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/39537d5cbe634c0d84a81816fab5fd80 : SUCCESS in 48m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e3780d5e945746c3b3cc3c92cd725fea : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6f0f36fbe7ff4e068790d55eb79c9eee : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b8983f9afb804e5fac42231bef450615 : SUCCESS in 54m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a49dd5edf7b48dca5e2c1f74650e285 : SUCCESS in 56m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b586a792d4044818e5373e29eac41d6 : SUCCESS in 1h 14m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c63a187452a46e9a57f630784d75ca6 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9b4505b20f34fdead3772a0c8641bd8 : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/69aa79b6c3b545d2ab4269601928e52b : FAILURE in 12m 00s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ba3d4b71d9a74a4e964886135576ecd8 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e9746eb2b03a452e896ec1d5ba3f8338 : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d12c77e50bee45f5b23cb6ff0b06c1aa : FAILURE in 10m 33s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a3f98c1a644456fac35e1dd469f321f : SUCCESS in 43m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c43a9629ca64f0d94e3cb49688e64db : SUCCESS in 1h 47m 18s","accounts_in_message":[],"_revision_number":3},{"id":"64151874091a9cc1f83515688492230d1f63f412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:06:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"faa6ebf018754afb600f6dcea327dd9cecd7abb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 17:53:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4af1b0ced4846f398d1e1027ca84e84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f8c7d75fdb4781aed97ba6d40e6d0e : SUCCESS in 3m 39s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74c0e0e315a845ed8f2228a2d8e0dcee : SUCCESS in 9m 52s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/234951dce7a34e9a8fc12ef1da2a258d : SUCCESS in 14m 12s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/152ca2ebe3894442a5cfa5b731c1e992 : SUCCESS in 12m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a56c64561e4a439d3c4a6598c8606f : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ebd2b55be1e744c7982359b6e32c8ec6 : SUCCESS in 9m 38s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e5086483236a426cb9948f3f27b77f72 : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b9c9e3f17ddf4061943b82a2f88d61a5 : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8ea3c926b14f4c06b5d37668567c65e3 : SUCCESS in 51m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d250362b15ed48f7ad7738180e8f02c9 : FAILURE in 36m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/514532666d064d03a2e381163ed37c5c : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f97e929a0e5a4ceaa68d37b1368eddfb : SUCCESS in 46m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b6132f802554d778e979b9fa2f25195 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec97535a4d8543e789e3a55f82100dda : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/902b2c8a1d7543e3ba91fd00e55498b5 : SUCCESS in 1h 02m 20s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/614cf4efc57d4ff78a65abf28d0a72a3 : SUCCESS in 46m 12s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/007e546d2ec0497b8c9091da704b1542 : SUCCESS in 1h 02m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9918ed44a63c431c9f655a820016384e : SUCCESS in 1h 55m 26s","accounts_in_message":[],"_revision_number":3},{"id":"0ee4384979d041449be9dd5338633d38f3bb075a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:14:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b51228a433790db307b3a62ca9994b28564b32f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 16:11:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e565c2bb47424c01aba163019e93ae8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d68fd4a47904469802c9e9263b2b75b : SUCCESS in 3m 54s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcd286993cfb40cda45cc6af9f979582 : SUCCESS in 12m 36s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/26aa679ded2d45db9ad946d647485cfa : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/2eec6a9cb6b94600ac66c08fa76d41e2 : SUCCESS in 13m 31s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bdb5d0bc0cf4e7a9c86714f9e36815b : SUCCESS in 6m 54s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a0a897f377cf43d0b1efcc2eb9852c56 : SUCCESS in 10m 45s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/95119f1ea5254ffca773100a5192ba2d : FAILURE in 4m 34s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ff17cbc6fbb4f80bccc4b1ce6a777b6 : SUCCESS in 51m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9bfe5b352d0547c5857aa616cbffda4b : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2a0c7f7d4d7a4e54965e9194671d4c8f : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f24720ca8bd64bdca626c1bc781e862c : SUCCESS in 56m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23e141bf84864842a65413c9fd75a099 : SUCCESS in 50m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66961018944447ffaaec7c35ad46db09 : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74d54ee5ae5d486493663202895be8d0 : SUCCESS in 1h 38m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4b1e45248aa417f9688942fc60b523d : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/211f07c494864899a75891aefeb93e45 : FAILURE in 9m 37s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/387ab1a8460849bda73153fb8476c33a : SUCCESS in 43m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7fa4c2a4410f41e082d2ecfa940afe36 : SUCCESS in 45m 50s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50f147df5aff486f84687b3d34f06d5d : FAILURE in 8m 43s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a2275901f5b46ee9289e9e5c7c720b2 : SUCCESS in 1h 05m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7680b1fb42fc43babb2b8e33fed83385 : SUCCESS in 1h 54m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b53b160f5274ee5b03ec90fbb2970856c2f37fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 16:13:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53c466cb7f12673930a234a484b96f2bda50df42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 18:23:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec680e8ecdf24cb98eae1a6799705207\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2740e892404a478c8a62970c31d80bd2 : SUCCESS in 4m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dbe7dcb9b5447dc8c12a72c5445850b : SUCCESS in 8m 09s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a31e95826c4e431a9988e9b32988bf0d : SUCCESS in 10m 31s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/cd328538fa214b96a3409a3bd2dd6935 : SUCCESS in 10m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/063ec4090def49e58cb781f9930b381e : SUCCESS in 10m 27s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/bee01144bbbb4fd4a27b9435a0475108 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/09d9f1fce9554979bfce3e283162de9b : SUCCESS in 47m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d6734ddd64f143ff9a5984f24a24f512 : SUCCESS in 45m 15s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/689e25a82c2d4923a6ab6154c2c3d486 : FAILURE in 1h 18m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52e075df39064ccf85bc9522e801992a : SUCCESS in 1h 04m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e311a4f782a41ca83957e12a3fd0676 : SUCCESS in 48m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aed96b79b1994f23a22ba94c39369070 : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67af1a2149a24295a3c87bd2f69fde19 : SUCCESS in 1h 15m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c79912fdb69426ba6226ab368091730 : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/51b1ef1faa5646ac9fddbf4b17de5342 : SUCCESS in 1h 10m 41s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/274ce4e771e8498e873da841c05ab326 : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bbef30d132140f9b6982a2f91d74247 : SUCCESS in 1h 06m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/146a36de823542d08e6e1f584e9a58a3 : SUCCESS in 2h 04m 54s","accounts_in_message":[],"_revision_number":3},{"id":"7b774ceeb4bab3e24b4f2e87cc82afd212c9fe91","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 20:29:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"299d6c41a3e40785c4ca1cdd13e7313a9f05be7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 22:13:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2f6731a2724d26be5bbbc6fac4decf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0f494a9a7f42e19095ceedf1c0a322 : SUCCESS in 7m 41s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7b8461e36c64cc9a1c72b849a42f229 : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/35e5a1469c1d4609a0ed339044ac950d : SUCCESS in 12m 14s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/fb6d930ea3a74c7aa20b9fb5f1a6241a : SUCCESS in 12m 18s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46c0d7f701fb452d831908c27f02876c : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/945171c81c3b4accbda1458835493b71 : SUCCESS in 10m 28s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbba19974155440097c89f2d6c7fa86e : SUCCESS in 15m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8b3990c33d794d169a09af162aa476ce : SUCCESS in 1h 30m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/271bac513ffb4573aa2685eafdaf9718 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d90081e3e5234101b211a9837cb60b71 : SUCCESS in 53m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2897b6ba4dc94021bc80e711e896a933 : SUCCESS in 1h 12m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6790d01243084ac1b836430baadba487 : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4e61d70758e44eba30f2ca6f9924b3a : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a57508216cb2453eaadc5f876f9554ed : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bff52b7f4554d48b06b23caca397b9b : SUCCESS in 1h 10m 28s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/babc7247569345088c2eafac70f06d54 : FAILURE in 13m 09s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a3c7155587c345b29a006e7b4ddb6eae : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a099027911294a76b9a82885683f370a : SUCCESS in 1h 15m 34s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/daa09a01401f4cf48c0ada95f9cb47d3 : FAILURE in 17m 06s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00c38f846d2749719ce3389367ad32c1 : SUCCESS in 1h 19m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9050aa370fc340babd38b6c9accecf48 : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ea9fa0e0c99afa87aa6759da6d56459eec2dcacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 22:15:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e2415a7066321b8289bc40cca5fd02986a408263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:07:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bbeea9b9a534771b9451d8dc344b7ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f25126628b4637b85a284e39bfd228 : SUCCESS in 4m 51s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2415052a61c41b7b220dc09d25f1a01 : SUCCESS in 6m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9dd536693534d6689d0525ecebdba0d : SUCCESS in 5m 44s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/1d1bd6e9c93d47049d9dcdc28f36748c : SUCCESS in 6m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99574f4af390410b9aa76fc5580168e7 : SUCCESS in 11m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/749e2be41b2e4365b6644dbc8b2011ba : SUCCESS in 9m 55s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/551cd24550a34476bbef58483e4e2fcc : SUCCESS in 44m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f04b5e00e8a241fbb4fc507835b209a4 : SUCCESS in 1h 00m 32s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d0d53a7100d74987850528dbfa9d5cab : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8f25e122c3f94a3e8f680673461a136f : SUCCESS in 46m 16s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f3d0bf23654404aac425e17bd80283b : SUCCESS in 1h 28m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f126d0dd05b412a841587a3fca9faf4 : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5935d10547e2485d89baf7a74cc49769 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac8b1cc80a3d48aa8fc0c7cfcc10c345 : SUCCESS in 1h 12m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/be17847067b34c28840867430c3d7582 : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b78c51e9b2704203be86cddd8ff4acd7 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f9bc7c54a0c466eaea020a6d36eb993 : SUCCESS in 1h 13m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e49a052b45141eda7cfb31eb6a8981e : SUCCESS in 1h 47m 16s","accounts_in_message":[],"_revision_number":3},{"id":"439c1897faf2c142db532b8f18b04509c1f87274","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:07:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"37af1fb7c61c37e7e3ee1368ef74217092d95fc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:08:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1225854188ee4fcba1df446a7cd7a514\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4c57aae9cf4a13976a50283b450a13 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf7588ac8ea4714524a542e378a64f87d23368c9","revisions":{"e2404421f1dae6177d0adb1fef7f119f22a19b8a":{"kind":"REWORK","_number":1,"created":"2025-12-12 11:08:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/970814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/14/970814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/1"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:08:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:08:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ifc5c8a5c9fe00f2ad7daf8012814cf528fe3c3e6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e2404421f1dae6177d0adb1fef7f119f22a19b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e2404421f1dae6177d0adb1fef7f119f22a19b8a"}]},"branch":"refs/heads/master"},"31c038e0de95998260d57d661f855a8c61708a58":{"kind":"REWORK","_number":2,"created":"2025-12-12 11:09:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/970814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/14/970814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/2"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:08:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:09:15.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ifc5c8a5c9fe00f2ad7daf8012814cf528fe3c3e6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/31c038e0de95998260d57d661f855a8c61708a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/31c038e0de95998260d57d661f855a8c61708a58"}]},"branch":"refs/heads/master"},"cf7588ac8ea4714524a542e378a64f87d23368c9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-12 11:09:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/970814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/14/970814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/14/970814/3"}}},"commit":{"parents":[{"commit":"b73d64078db90f63f371f4a56789521db21617a4","subject":"Merge \"Copy mariabackup script instead of templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b73d64078db90f63f371f4a56789521db21617a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:08:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-12 11:09:45.000000000","tz":0},"subject":"Remove outdate code for SUSE","message":"Remove outdate code for SUSE\n\nSUSE/openSUSE now unsupported.\n\nChange-Id: Ifc5c8a5c9fe00f2ad7daf8012814cf528fe3c3e6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cf7588ac8ea4714524a542e378a64f87d23368c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cf7588ac8ea4714524a542e378a64f87d23368c9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
