)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~973144","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I8b8557ab111c853775e1fc1521a136a224a1e746","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-25 12:46:13.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-02-25 10:48:38.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-09 07:48:25.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-25 12:46:13.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-02-25 12:46:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b8557ab111c853775e1fc1521a136a224a1e746","subject":"Remove deprecated innodb-file-per-table","status":"MERGED","created":"2026-01-12 17:16:16.000000000","updated":"2026-02-25 12:47:02.000000000","submitted":"2026-02-25 12:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973144","meta_rev_id":"d71323ce48afb864c75516db50cf3b3ce155c8ac","_number":973144,"virtual_id_number":973144,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 12:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-02-25 10:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-02-09 07:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-02-25 10:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-08 21:03:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-08 21:03:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-08 21:03:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-02-08 21:03:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-08 23:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df8d165d67e81a9296df8c6ff37eb5812f743f7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b49e5a20f5a0f3f57807b7109c70f9f29c987b9d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:19:20.000000000","message":"Topic set to osa/bindep","accounts_in_message":[],"_revision_number":1},{"id":"d8400c1bc0a33bd6460da99e616051d2c165838a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:19:54.000000000","message":"Topic osa/bindep removed","accounts_in_message":[],"_revision_number":1},{"id":"869e9dde26d42c39d731ce52461eef9badd7ba8a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:19:57.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"8f9f7ba23d3ef63e11b472bf941ce6031bd1956b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:19:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2639ddabe09c85b1f825592f5f01e31864b96df6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:20:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"06437fd156cc43333ee1550a6a67eac583d3d0b7","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:00:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"6e6d10db6fa06ae94a8e30c1ebcb1394dd344853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:01:08.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b722ebaedd3da49a5c717eabe6276b184391f8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:01:20.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d83fe5fa23f3dd43b6775de364b818c1f18a44c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:02:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"4b321f9a89ffeb682c1b91fd898cf21c19485f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:03:09.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"44a06d476f2f4c7618f1dd7267e87bb45cce5a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-08 23:06:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26233e25f7334c70a23650de96c221d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf6c6a496fa4346937197d14c6c6f14 : SUCCESS in 3m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f818588a78a24393b860ce166d8eaf84 : SUCCESS in 9m 03s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e6e2c1c1c264edfbce8b533fddf58ab : SUCCESS in 11m 00s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d0fd15aa32364702b386c38cfe84c828 : SUCCESS in 10m 43s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d978bc86615484b91bde11fe18204dc : SUCCESS in 6m 47s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/01398f00cf124ce69432a7abdc7c8fae : SUCCESS in 9m 06s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7de3a2ecf8484d16bcabc9d16fbce26e : FAILURE in 4m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/72ab82946e5d4919856ee0c7b00dde76 : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/72997bc9d7a24a1ba8bc272db05a1eb8 : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aca5334cee5045239ea0d391600db82d : SUCCESS in 44m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16d4c75ed71149a89b77d652c95caf4f : SUCCESS in 1h 02m 11s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/307b4249aab14a72922c97764543e11c : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f249cd2bbc449daeb36a036a9a1451 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25761853194c4d4c840e9bf24f767e35 : SUCCESS in 1h 00m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/055362725e9148548c3ba590ee25585b : SUCCESS in 59m 20s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bbe88447508541be9c2533ab4567afdb : FAILURE in 7m 21s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/426826b1eaff4bb3984b17d87f1eec7b : SUCCESS in 1h 04m 45s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0b5cf629e16c452299e6e5e42948a8e2 : FAILURE in 56m 40s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/115cab55e6e6457e8a552c043985b484 : FAILURE in 6m 45s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8387007b6ab4e988e26e9c2ea88f8db : SUCCESS in 1h 13m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3de9cb4c90604d0082bf0f3fa0afbafe : SUCCESS in 1h 55m 21s","accounts_in_message":[],"_revision_number":6},{"id":"88dc79ec508683e72dfb1b8593bcce63c063131f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-09 07:48:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"417cfa4b0bf05c95de13582f7c7754b197672a41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-09 07:48:43.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bfe7ad35026d5c7c9e9361a76c9bac4373a8182e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:15:43.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/785cae5f0c484f06af5c7cad26f1a77e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b15d150b524303bb8b36414294340b : SUCCESS in 5m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb51a676f86640baba9603b1c203c70b : SUCCESS in 9m 00s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef52231615a341bdb5e8eb3338fa80c6 : SUCCESS in 10m 37s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c36669545424428aa519bd4d49f17e79 : SUCCESS in 10m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beaf0ef68cf04d2bb7e0827251323b34 : SUCCESS in 13m 10s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/96557bd119d44103bc11cf6e73d48e02 : SUCCESS in 10m 01s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05c47cee35e24b1c92383cf82c68fe9f : SUCCESS in 12m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0a62c04c74954ed8a0957a4867ca1c66 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4d55a899684745598bbc7f51c14566c5 : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d9dbba9b73f143099ee3118b130ac0b6 : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46b31ec3aad245c79704536d5a3f30af : SUCCESS in 44m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cda767499343465ca1d14e7ccf92a414 : SUCCESS in 1h 07m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae8154c4b0f84d83b159e13b0ca337ba : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/768d5c9525c7453f991add2dbc39e06b : SUCCESS in 42m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38d93b4d34c04522b559f5204a62e642 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d467f1f9e90c4a11b9b81ffb80f23e60 : FAILURE in 6m 51s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ada462642084604a268fc62e34d336a : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a1db9f228b5a4d5ca8ad4013efd8e19c : FAILURE in 29m 32s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a86e6d87e2c4741b83686d941d775e5 : FAILURE in 8m 08s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f36d06ad85048b3924dfa064d501f22 : SUCCESS in 37m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e878df2990b64a0498faf8941e4eebe3 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":6},{"id":"2d36410118790fe60d24427e3b73f43fadb8a07d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-24 20:41:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"378a97b577a5e1c13c3685d8dd1be3ad9f7dbc35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 22:33:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6bc22e8528540349740fcd399dae61f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e812d2a03d00412b98d05ece3a309a2c : SUCCESS in 5m 19s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d62607e4bd544f0c902a2cfa4202ba55 : SUCCESS in 7m 56s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/66cb82c8754a4ff8911cb70a068b939d : SUCCESS in 6m 03s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/593f16d699f54a4a8c2299252f1b5049 : SUCCESS in 9m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3129f67c29df4089b678023a6016ca1c : SUCCESS in 7m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c0a2dbfd3da43679a78941317cdb227 : SUCCESS in 9m 10s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a42990ff73d54f99ae24f5ccf34c6723 : FAILURE in 4m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b7ddc620acde4738ba8e2c170b72d82c : SUCCESS in 59m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/87fcbed26a394b4399538bae30e0c4ba : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aa044230838e4af08e8f8bc348860df3 : SUCCESS in 1h 03m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74021c7edcb449a3918ccabfad5ab3d2 : SUCCESS in 39m 54s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d15fd1cd4c6b44c08fa3da6d25c66ffa : SUCCESS in 1h 00m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd31a8a0539f49f4b1331e3b3f2d5fb1 : SUCCESS in 34m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a4a4c1f94bc4a3fb9f21b4d216a0758 : SUCCESS in 34m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baa137ad312244f68f503659df79ea8d : SUCCESS in 57m 40s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9ec4ae8fa1d54fc08136bb7256de86cd : FAILURE in 6m 31s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8426d4b96dd54a4cbc9ed95aacb4915c : SUCCESS in 54m 09s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/17f3b180669342a38419744860d3e7b6 : SUCCESS in 31m 47s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3439a0d65e2d40e69188bea3c4d1186c : FAILURE in 5m 59s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fcc47a8f73946c79a8fa45a5a69a393 : SUCCESS in 53m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa8b5bff7aeb4fa58d75274a5b01e32e : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":6},{"id":"c0bbed459e66a4589bc8c6f6b9153d11cd3a1f8b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-02-25 10:48:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b5b5ab9a06d881c65263ee1cf8f87237d28a0a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 10:50:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9e62c1634d959a2587a673672cec76c7f5788dc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:46:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68b5346396e46b99b89a85c9234ecc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cc84f4f907407889c6551491430fe1 : SUCCESS in 3m 38s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c709e5257b44424aa0741e973142ca3 : SUCCESS in 11m 41s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce38b234f5964f9da3d995cd42f163f2 : SUCCESS in 12m 09s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/64e8d4f43f5840adace06008f30093d6 : SUCCESS in 13m 20s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce6d6dbe72df4a6a8bd30be38b437b12 : SUCCESS in 9m 10s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c9cccfc0bb29454fbbcbd785b7e5de5f : SUCCESS in 9m 35s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bb9fb88076a84296ad67b5178211922e : SUCCESS in 55m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ffad8eb7d5be49b1a2435145c9762bd7 : SUCCESS in 1h 06m 11s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/135e1b50249d44f89e7494e980277ae8 : SUCCESS in 58m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0e43a5b9b51498c829802fb35f14f97 : SUCCESS in 37m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd765c2b6ccd4beb97f31339582f5e7f : SUCCESS in 44m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f38b26834b541619d23fa8e97a9979e : SUCCESS in 37m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/555aa849ae4d4b0d946c9fce7bcfeb61 : SUCCESS in 1h 39m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2a106ccedf0485a99082539511f19c1 : SUCCESS in 56m 15s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1307a1d1afba4964944a4c20b354d671 : SUCCESS in 52m 28s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1ba2acf0343042128f848f2017a0abcd : SUCCESS in 35m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/646270c1ddb7479ebeacb1dacfeeaef6 : SUCCESS in 34m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e670975eb056447cb10a2e1f1e4982de : SUCCESS in 1h 46m 27s","accounts_in_message":[],"_revision_number":6},{"id":"9678e9b7571047645b27c2a7a30088e4aeb16c53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:46:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d71323ce48afb864c75516db50cf3b3ce155c8ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:47:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd535176d2440bda22048a15f94f66d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbde295359143cd8fc92dd2898ee00b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","revisions":{"1bfb626bcdb81d949f0a9561290ee1a3a77d4ab9":{"kind":"REWORK","_number":1,"created":"2026-01-12 17:16:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/1"}}},"commit":{"parents":[{"commit":"ebd7fd89988ac38e7e5b3e08a758fdd24efd3518","subject":"Merge \"Replace normpath with join for backup script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebd7fd89988ac38e7e5b3e08a758fdd24efd3518"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"subject":"Update outdate information in bindep.txt","message":"Update outdate information in bindep.txt\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bfb626bcdb81d949f0a9561290ee1a3a77d4ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bfb626bcdb81d949f0a9561290ee1a3a77d4ab9"}]},"branch":"refs/heads/master"},"009b1d2b0d50ad3681f056da32564468c1edb980":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-12 17:19:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/2"}}},"commit":{"parents":[{"commit":"ebd7fd89988ac38e7e5b3e08a758fdd24efd3518","subject":"Merge \"Replace normpath with join for backup script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebd7fd89988ac38e7e5b3e08a758fdd24efd3518"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:19:52.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/009b1d2b0d50ad3681f056da32564468c1edb980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/009b1d2b0d50ad3681f056da32564468c1edb980"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2f01233329086061821c5d49d3999554f6129ee9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-08 21:01:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/3"}}},"commit":{"parents":[{"commit":"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b","subject":"Ensure conditionals are proper booleans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 21:01:08.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2f01233329086061821c5d49d3999554f6129ee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2f01233329086061821c5d49d3999554f6129ee9"}]},"branch":"refs/heads/master","description":"Rebase"},"e8dbbb8feb2e0ddbdbf011c6b66e4775d647e112":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-08 21:01:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/4"}}},"commit":{"parents":[{"commit":"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b","subject":"Ensure conditionals are proper booleans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 21:01:20.000000000","tz":0},"subject":"Remove deprecated innodb-file-per-table","message":"Remove deprecated innodb-file-per-table\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e8dbbb8feb2e0ddbdbf011c6b66e4775d647e112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e8dbbb8feb2e0ddbdbf011c6b66e4775d647e112"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8f6879cab08f8b37c5dec709a293fd06e6671796":{"kind":"REWORK","_number":5,"created":"2026-02-08 21:02:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/5"}}},"commit":{"parents":[{"commit":"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b","subject":"Ensure conditionals are proper booleans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 21:02:06.000000000","tz":0},"subject":"Remove deprecated innodb-file-per-table","message":"Remove deprecated innodb-file-per-table\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f6879cab08f8b37c5dec709a293fd06e6671796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f6879cab08f8b37c5dec709a293fd06e6671796"}]},"branch":"refs/heads/master"},"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-02-08 21:03:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973144/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/973144/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/973144/6"}}},"commit":{"parents":[{"commit":"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b","subject":"Ensure conditionals are proper booleans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:16:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 21:03:09.000000000","tz":0},"subject":"Remove deprecated innodb-file-per-table","message":"Remove deprecated innodb-file-per-table\n\nON became the default in MySQL 5.6.6 and deprecated in MariaDB 11: “as there\u0027s no benefit to setting to OFF , the original InnoDB default.”.\n\nChange-Id: I8b8557ab111c853775e1fc1521a136a224a1e746\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
