)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~975250","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6d6012bed90bb40a4aea6d515f6a1536a3717301","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/core-2.19","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-02-05 08:08:11.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-05 10:36:44.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-05 10:36:44.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-02-05 08:30:13.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I6d6012bed90bb40a4aea6d515f6a1536a3717301","subject":"Ensure conditionals are proper booleans","status":"MERGED","created":"2026-01-30 09:38:25.000000000","updated":"2026-02-05 10:37:45.000000000","submitted":"2026-02-05 10:36:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975250-osa/core-2.19","meta_rev_id":"b3aa2bfc51b917db87441c7f5131c9616c0b395d","_number":975250,"virtual_id_number":975250,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 10:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-02-05 08:08:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-04 10:55:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-02-05 08:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-02-05 08:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-30 11:13:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 10:55:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-04 10:55:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-04 10:55:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"11c0d211ffcb890be268bbbee9097cf606151442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-30 09:38:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8290a2b3ff504d1c966f5db277ce0bcda635ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 11:13:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ccf776bf1c64dc18ca264ed15ddb99f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7e98fc98ce498eaa6442aca5b6d028 : SUCCESS in 5m 19s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adef02fc59554cc2be2c59abc8e3e3ba : SUCCESS in 8m 42s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/30b546c222df436695910af7c91a57df : SUCCESS in 10m 16s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9c6057363fd04860894e5975361e7ce1 : SUCCESS in 10m 11s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fb5f3fd46414976a7c56ac238172e30 : SUCCESS in 11m 54s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ed08aa027fd94f50bef115072b956031 : SUCCESS in 10m 07s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ce974728f2b4114b280b0104e8c7bd0 : FAILURE in 5m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7ad57214855a4069a0920d98375d6bb0 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fc5faaf491cf44139ffb83640a474d64 : SUCCESS in 50m 02s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b5f722df152f45c081d48beaf79e182b : SUCCESS in 49m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/548aec127a8f4fbba4515e5b37134496 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/32b6fb32b01f4237a05d8cbd62dcf178 : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63470d6a02d0470c94923cd1722790f1 : SUCCESS in 41m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c8960c462d4473598476754a0643278 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faf803ed12724d90b9cb21e792a7a4d8 : SUCCESS in 1h 31m 21s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2a51ef45304343e7b0f8fb6f7479fa2a : FAILURE in 11m 19s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/08492b4fdd154f1a87d047093e438e4a : SUCCESS in 39m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/250611110c9349b5a3bc383e475717fa : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0e545cb04174453da4b13887f5b7353c : FAILURE in 6m 31s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e2344e3e05c4e3bac7adf584bdda312 : SUCCESS in 55m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1eaaebea7774c1b956bf14394dd3082 : FAILURE in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"20bffd62e025c4a011e945a2f66283f77d6681ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-03 17:25:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49f761f0b5b2ed1a98d9065e131a61680cef4707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 18:43:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2579408bc7be496692ceceecd4ef4ec5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd836d03fde7426fa512b835478a4acd : SUCCESS in 6m 12s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47d1969f491b497e8c382eb6f34e2970 : SUCCESS in 9m 42s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0466bec855e4ee6b7760c7999d9f563 : FAILURE in 5m 31s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d7350f50b76f4a3cbe9e09b80adf9de1 : FAILURE in 3m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5246633ac87c4f00aa5b9e22ead814c8 : FAILURE in 3m 16s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/bc48a2ac26914019a3ede1d595be5373 : SUCCESS in 9m 57s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dcbbee846b3141469b5ae39aad4f9e13 : FAILURE in 5m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/218390d537074fe8b12665ee1166edcc : SUCCESS in 47m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d5191b6675e849f98a15b89e7333058b : FAILURE in 32m 19s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9fca430079774633b652ed09f823dd84 : FAILURE in 33m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db5959f00b1244d8b858cfc9ab82aa74 : SUCCESS in 48m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da6ea63dc18d4f3c9a97cbc8906e8251 : SUCCESS in 48m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83b2732cab2f4c2598f2a2983470fb62 : FAILURE in 22m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db024e85d12443a99b94cf94c1efa095 : FAILURE in 40m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/417926a058fa460c957e1d183fa356fb : FAILURE in 53m 47s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4c6bd13d94924aa3b6ad49de4fc73904 : FAILURE in 13m 03s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a65b04dd8304f0abaf040f4e32ed4b5 : FAILURE in 34m 17s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3b5035b8ef524160a8e39e13bb9ad8dc : FAILURE in 32m 43s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/799c246d1a734831a0fdea37e3c2a2ab : FAILURE in 8m 17s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d470d13641624270a2d4bd4b4c19b70d : FAILURE in 36m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/111c19554c0844f8beae6f08fc20cc23 : FAILURE in 1h 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"940c10be727ae296f466050fc0250701e44ab2d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 20:05:46.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"624a72b024e670755805975c423fba609bebf047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 22:24:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f89a9bf87d1448e5bda081a81dd31054\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51035d4b27948a58734c122e30603ee : SUCCESS in 5m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60354f3b266846149f5c1891a993ec9b : SUCCESS in 7m 02s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/1fbcf44c5d5e4e2c88338d6017d8dfa7 : SUCCESS in 5m 43s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b5613e6fc21c46e2adbe999ffa8206a1 : SUCCESS in 10m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25f99e923d2549248260e2ace5ee12d5 : SUCCESS in 11m 07s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/63e0df6f1131412e8fee7e5ad8864988 : SUCCESS in 9m 15s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c5665b6b82eb41a5866f13f2b6c9ce7e : FAILURE in 5m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/923cca54ec6845deaf147aa6dc0a140a : SUCCESS in 1h 04m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/adea6afd284c480eadee8bdc338e4bd8 : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f8ac846df99c4cb6a920f907bd9019d5 : SUCCESS in 1h 06m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e1bb77181db418e90f9a3c473588468 : SUCCESS in 45m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f8b8ea3201924863985194e77e06c549 : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a18218d70fd4c199c2a9c080f812e14 : SUCCESS in 1h 11m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0691c0c39d2b48bb86598438416a2021 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e97836b704574d4d854fadae64a5d1d3 : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/da30cd8d885344968f83a6a9067c57ee : FAILURE in 8m 44s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5579280185b24c98b8f0b063688a0cf4 : SUCCESS in 1h 07m 54s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3455b0fb26554419ab656e2a9af73c9b : SUCCESS in 56m 51s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1c6f53cfaedf4f33b73f8491606e0b69 : FAILURE in 7m 55s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e01e0789cec40b3bff7a4e1a8ab40be : SUCCESS in 1h 15m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38e7f137ea1840ddac241e2cbe07647e : SUCCESS in 1h 57m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b408efe9bb0266a5c9fb94c509f232ae5be7ef90","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 10:55:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f5dacd46357c9e3b2a7cafb2d2625221d236070a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-02-05 08:08:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cdd4c6b20f8f2024c4f77d7de0e20cdf51a5c755","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-02-05 08:30:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e046bcec5484c41da1e7e58bb04601c4915a4f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 08:32:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70b65362d9c1e06dc6369eaf18d467bc71e3da05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:36:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb75b85e203d44c5a30c9a8aa8f7e813\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2990eb47ffb46e3aa3a1f5dd48dfc68 : SUCCESS in 3m 34s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a4d6cffbe834dc3850ac21c8129bdad : SUCCESS in 9m 49s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5eba9b9fd9f4e49b0aa3c450a4e3367 : SUCCESS in 6m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/2c083758fb7a49af8ceedef92520a029 : SUCCESS in 11m 02s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7aaa7fd74f8410784b4b1cb698bcbaf : SUCCESS in 11m 15s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/0ada51dc85f74010bfeefe289dc69a6c : SUCCESS in 9m 51s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e0cd06e69be94c72a05792f2e788a4a8 : SUCCESS in 46m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8bf2fc2a74f34934bfdfbb67479f0dac : SUCCESS in 46m 25s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f245e327e5184736ad7c55b13bc8d63e : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/346f02c3aa0c4c7ea13ff2d001b7db14 : SUCCESS in 1h 07m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9fa05d9b5a214d6e8ff480941abbc078 : SUCCESS in 1h 28m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e9146d1428e4970bca2d536f0b62ae7 : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82bb033105f24f27b86de896fd995f5e : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee65fa47fada42a6af1dfc02d0ec7768 : SUCCESS in 1h 39m 50s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/faf98c251ce74b6eb204a7ed6431c074 : SUCCESS in 41m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6d31adc5799b4004b22701251dbb9c2d : SUCCESS in 1h 02m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4e115a105504eac904ee9d8a1e85d99 : SUCCESS in 56m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48c9c695dfc340e7b0b6ca07828095dd : SUCCESS in 1h 59m 32s","accounts_in_message":[],"_revision_number":3},{"id":"abd03f07ef7a1db9182d60058201a886910e997e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:36:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b3aa2bfc51b917db87441c7f5131c9616c0b395d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:37:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d277f678254180803f1402bb0a2430\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4cb4698c07d489e83e1f8e39dcaabda : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b","revisions":{"1021c0067eacd79ec05c835d9a52342190364edc":{"kind":"REWORK","_number":1,"created":"2026-01-30 09:38:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/975250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/975250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/1"}}},"commit":{"parents":[{"commit":"855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c","subject":"Merge \"Remove outdate task - Clean up legacy repository config not in deb822 format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-30 09:37:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-30 09:38:17.000000000","tz":60},"subject":"Ensure conditionals are proper booleans","message":"Ensure conditionals are proper booleans\n\nIn order to avoid broken conditionals error in Ansible\nwe need to ensure that variables are proper booleans at all times.\n\nChange-Id: I6d6012bed90bb40a4aea6d515f6a1536a3717301\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1021c0067eacd79ec05c835d9a52342190364edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1021c0067eacd79ec05c835d9a52342190364edc"}]},"branch":"refs/heads/master"},"c698f1345b6bdb844b7800bfe252f5e8818763b9":{"kind":"REWORK","_number":2,"created":"2026-02-03 17:25:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/975250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/975250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/2"}}},"commit":{"parents":[{"commit":"855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c","subject":"Merge \"Remove outdate task - Clean up legacy repository config not in deb822 format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-30 09:37:16.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-03 17:21:56.000000000","tz":60},"subject":"Ensure conditionals are proper booleans","message":"Ensure conditionals are proper booleans\n\nIn order to avoid broken conditionals error in Ansible\nwe need to ensure that variables are proper booleans at all times.\n\nWe also move `apt_package_pinning` from meta to actual include in\ntasks, to ensure that `galera_repo` variable is properly defined.\n\nChange-Id: I6d6012bed90bb40a4aea6d515f6a1536a3717301\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c698f1345b6bdb844b7800bfe252f5e8818763b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c698f1345b6bdb844b7800bfe252f5e8818763b9"}]},"branch":"refs/heads/master"},"d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b":{"kind":"REWORK","_number":3,"created":"2026-02-03 20:05:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/50/975250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/975250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/975250/3"}}},"commit":{"parents":[{"commit":"855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c","subject":"Merge \"Remove outdate task - Clean up legacy repository config not in deb822 format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/855ff192dfcf1a9ba083a71fc4bd1d5054bfbe6c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-01-30 09:37:16.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 20:05:46.000000000","tz":0},"subject":"Ensure conditionals are proper booleans","message":"Ensure conditionals are proper booleans\n\nIn order to avoid broken conditionals error in Ansible\nwe need to ensure that variables are proper booleans at all times.\n\nWe also move `apt_package_pinning` from meta to actual include in\ntasks, to ensure that `galera_repo` variable is properly defined.\n\nChange-Id: I6d6012bed90bb40a4aea6d515f6a1536a3717301\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d851b2e5a27091cf03fe7f88fd8a2927a4db5f6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
