)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~977963","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I1dfb829df92de5174e6fd123ed8a4717a6972cd6","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-03-09 16:50:26.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-09 15:06:15.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-09 16:50:26.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-09 11:11:47.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-09 14:03:02.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I1dfb829df92de5174e6fd123ed8a4717a6972cd6","subject":"Use READ-COMMITTED for transaction isolation","status":"MERGED","created":"2026-02-25 13:52:57.000000000","updated":"2026-03-09 16:51:50.000000000","submitted":"2026-03-09 16:50:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977963","meta_rev_id":"4de201fe2d3392b3c37d4c553288605371c94213","_number":977963,"virtual_id_number":977963,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 16:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-03-09 15:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-09 11:11:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-03-09 14:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-03-09 15:06:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-03-09 08:59:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-25 16:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 11:10:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-02-26 12:54:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-26 12:54:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-26 12:54:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-02-26 12:54:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"2a6fbfc863770e868b86e9db7caa39ca826fa036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-25 13:52:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e33e8a674cdaaa3937de88876d5238e0923c6354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-25 14:15:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da85e713b8b002829597cf326285a0c2fad1afee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 16:30:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bfeb6fd631f4529bf611361f9973a3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4250933e2f4372a13d3c3a0ef25d03 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0253828393e249e5affb3993c18894e6 : SUCCESS in 2m 07s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1826ab671d94befa6995f09259c0153 : SUCCESS in 9m 30s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/58f36ab1edb840ed83897add8341b5e0 : FAILURE in 3m 46s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d720d14fda8947a7b420f18420efa5d2 : FAILURE in 3m 41s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17a177e2f4524b9f80694cb1ab404371 : FAILURE in 8m 03s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/10e65a0bf1f1498da5c0d5d189747434 : FAILURE in 6m 20s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e63e72aea8f348ceaaf92d76b61a7aa6 : FAILURE in 4m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/523f718172cf4cf3b8dfc4305e84ebec : SUCCESS in 54m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b4ba28c4edd4d8099e41fc2bd662531 : SUCCESS in 35m 09s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/844297da4674416ea069d970662c1bad : SUCCESS in 58m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d77f3d2f44204b4e8b6327a2a0db69bc : SUCCESS in 39m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7e6b6afe5b6143da84284a03daf0992a : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45159b8ef1f84180832e5e7da87b9dbf : SUCCESS in 35m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05d4655af7ba40978ce3cc1e56618f99 : SUCCESS in 40m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdbcdba48c1e4c368bc7d02c9ed6e764 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3c3c020eda55434eb62a6f4c8d4fa104 : FAILURE in 11m 26s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/30ed2026b5964ec6aaaeafc4a31341ff : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f0507f4a3f9c4ab9a62e510b06a788b2 : SUCCESS in 32m 53s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/faf85325cd7c439daa1e9cd4fe9f3490 : FAILURE in 7m 18s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/872fa62dd39a4ffebbc36bcae9edfcbf : SUCCESS in 1h 07m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35bae48435304273a27ae7d33395973b : SUCCESS in 2h 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"087af6171169090f61654cea5a564a2812650843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-25 18:16:02.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"16e357ff5aec2082d714f57dfd46cedf64d8b69e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 20:19:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0e90c6a7c2b42bd9d9701545cb67379\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a4f1fffeea40c9a9b9c4b6cca3d566 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30a5e42692534a9e8ccc36839a5d5fa4 : SUCCESS in 5m 41s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de74e33df55e4d34b879ed0fc52748a0 : SUCCESS in 8m 10s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/446d396636884174b169c894d678f4ee : FAILURE in 6m 25s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d4aac4c7afe14413b45a7a709a189dee : FAILURE in 6m 22s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29bf95a53afc4cd6af5721b911570b96 : FAILURE in 4m 51s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/41b288969e614761a1ec8a3f3b38b7ce : FAILURE in 4m 05s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e18b21d55f1746838ba55dcdae4de155 : FAILURE in 3m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ec02872829c94fc9b98bab24b0e2819c : SUCCESS in 41m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f02872dd7d514c9bacd3c6aecd16d1f2 : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/29e6a8ece26e410398bab5da0dcb1215 : SUCCESS in 1h 01m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ddcfed66d4084202808a3a20919b2537 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e7fd2f8275f4b188cea2e889c3f8f2a : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf986958d11740c58f4a4e25155b8838 : SUCCESS in 1h 02m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cb12048f0e54fce86815709447c8690 : SUCCESS in 39m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de5cd3a1409544e49fd453b8949c1097 : SUCCESS in 37m 56s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/1458aa4dab034bd686a4203ead004769 : FAILURE in 7m 07s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/781467df59a4474bb9975a28377e58a0 : SUCCESS in 1h 01m 32s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/48b7d4554f0441ec873b2c76961059dd : SUCCESS in 34m 38s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/44b2ec437cc241039d6a7306a3c94e62 : FAILURE in 7m 16s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/668545658ff4466395e0ffc33e102606 : SUCCESS in 36m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76621bc07c9b4db3b5879a49406043ea : SUCCESS in 1h 56m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d397919ed6d1a99476b22d9223e28e0b0ae6db93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-26 07:51:48.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7dec27214293c6c39ff5f3adc6f45e4268e85aab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-26 07:59:35.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"53acee3de125b73efa5590e4b659646b53632b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-26 08:09:23.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"853c93cf98f6e6970954f1ca1e52605cf5ecbb4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 10:32:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1bda267886d40388d9b8c73cb5b54ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf23d94050d3485fa417411a4a247550 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5650d6b3250d4e9e9ef303418aeb8813 : SUCCESS in 3m 21s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc2cb987b2a4b968ce7544264bcbe69 : SUCCESS in 12m 24s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b367c9bd4ba4b4f8b335e0145fe7237 : SUCCESS in 10m 50s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b1baf434a7984bcba0847b68daa81cb8 : SUCCESS in 10m 52s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4ab6f32e9d84f83a844d2e79c171fff : SUCCESS in 12m 11s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/62dd9d27b4ea44a4a8e062ebc5f1ac6d : SUCCESS in 9m 31s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eaf51afb5546457899b6a752d62ec148 : FAILURE in 5m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b4ea4f7bc1bc4e75bbb7291b61122041 : SUCCESS in 43m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/34976a8e7855404898862ddff849c3d0 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/87ac7e4579984c179a3e1e14dd0eff56 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7e399bed8087475496b2dba39a529778 : FAILURE in 41m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c4ae3e1497cd4192b88a083fdef39972 : FAILURE in 41m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f55a86161ea46dfad4c25b24ba385c1 : SUCCESS in 2h 16m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d79955aa1044c582d73afac5165a4f : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a2db3276d6e4c22a33f8e824e719928 : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/897f60633d3849ee94e7fb4aeb7eda16 : FAILURE in 6m 03s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/71ad806202c64aeeafcb3614cd243cc4 : SUCCESS in 1h 04m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/234caa7092504b84b3a7f0e56e95b6b3 : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fe557f9012284fca93de5e99efb497b8 : FAILURE in 6m 19s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/114e905bb99e4635aeb019601e3cd850 : SUCCESS in 2h 06m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f1e7933770842849c209bd77b9b616f : SUCCESS in 1h 49m 31s","accounts_in_message":[],"_revision_number":5},{"id":"37a46d7e4077a2fd914d36d799c702516a583512","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-26 10:34:46.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55da2fa37ada172165ecd76cd2ee0e2416391d46","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 11:10:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c317122f4f018cd05b31e19571e5379b3fb4c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 12:50:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0873f5802eb4d3bb39bc2e107eb760a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15cd5af1cb8c4bdd930d4b92c96e1db0 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cacad58436e74d6d9c8c95ad69a64d17 : SUCCESS in 3m 30s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f3d25aa45cc47c6a81511811723ee5f : SUCCESS in 9m 58s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3637198ed6d34e9cb9c9e6504e6710d3 : SUCCESS in 10m 33s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/58beda3081b140cfbdf49e84e6ac04e8 : SUCCESS in 9m 50s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73496a3f14d444bcb98e105c08d96674 : SUCCESS in 11m 17s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/72843db5e75245c2aab9db54d2f77be2 : SUCCESS in 10m 01s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/35877d9fa0c14a24b70559c3877a6072 : FAILURE in 5m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/8a874d6a748d45f7ad0fb02c3bfbc337 : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/04033643c213466390f907be95825909 : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4d882d6d3d7c4a83989e61694398e2d3 : SUCCESS in 46m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9ad20ba80604a8f81d59567e08eb711 : SUCCESS in 47m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2f2747a147564639885d224080587ff3 : SUCCESS in 48m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfffc98d43fd4b468972552e6d786e4c : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8184ad7492c428a81a2db8775e9203c : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de71338529f4a02ae05a584e64aa1fa : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/ac1638dc38c5473a9c171787a7807bc1 : FAILURE in 9m 38s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/389da48017b94f5d9f77809d401e3c49 : SUCCESS in 43m 36s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1d8da023ac6e446a8fbeddfc00e55d0f : SUCCESS in 41m 44s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6fb1f1c571a344f484a87af40d297705 : FAILURE in 18m 05s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7fe8042ce9e4c79b3c5917244450aad : SUCCESS in 1h 21m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4be80e8be744fe3b3d0857a4190f404 : SUCCESS in 2h 06m 05s","accounts_in_message":[],"_revision_number":5},{"id":"3c656983a7560cce4bc911ed4d7417a4ab6fe688","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 12:54:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1dd29318b21dcdf4e5586a1e1fb9001e78f08337","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 08:58:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* requirements.yml\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d7682a68deffa35ec388cc811c867a55f2d35f3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 08:59:26.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4f67e75cd5d1c99fea2344d2c9876fc5805abd6c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-09 08:59:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"849464688240336d430fb2eec8614bb4c7419a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 11:10:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e73bf23b6f549c7a877db9fd363de7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3a566850cb4124a145f85f37468cc8 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ccbbfae8f94f279a14c2417ce1705b : SUCCESS in 3m 12s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4750e44618ec42d183e57fcfa2e0649d : SUCCESS in 15m 33s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/38e9914f03fa418cbae09c5cacdcf0e1 : SUCCESS in 12m 22s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/19a5de128e1340eaba73594a4cf80218 : SUCCESS in 12m 32s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87b487c1759648058c98630660dc0855 : SUCCESS in 12m 34s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/813f6fd0d84e4da082137c23b528a7fd : SUCCESS in 10m 15s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f2527fb9bc849f1b202d585e785a7b9 : FAILURE in 5m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f2ea95aaaf74836b98b33cee3392810 : SUCCESS in 59m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b6fdcff049b145dc89a44e2e2ae8440d : SUCCESS in 1h 18m 00s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cc6c4627f8604ad3bda3cd3daa4e24f4 : SUCCESS in 52m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b0afc64481e0493a9dbbb0418a9869b7 : SUCCESS in 54m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49f427fa084a4235bd63b28ae8c6981c : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82c85b70c42c4f378422a2bb8f770e39 : SUCCESS in 2h 00m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e17755a436cf438caeb4db5714d0259d : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3b4aa1613914211853fbadc998a023f : SUCCESS in 57m 20s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f2faddfef6f499181662640bc898089 : FAILURE in 10m 07s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6fdc0d803e134f6bb79e13641e79d572 : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/aee9ada3ded14d14a1819f3c8ee1cb93 : SUCCESS in 45m 32s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d329c636d5e64d95b777bf90cfbaa133 : FAILURE in 16m 22s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/025e2a5351884bbb9041841123fffc52 : SUCCESS in 1h 48m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10626a5531764581a2dd645c11142c44 : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":7},{"id":"1f697cb28d56e0f95c07b8123ef81f0a3872ae65","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 11:11:47.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a077e5d2488db0b6c4eaca20dd0b5acc4027f3c3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-09 14:03:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b05a5ab150e40f0afd74973165331232a9e4dcc4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-09 15:06:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3755cc9dba09bc9edc55555f08e7394adc705b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:08:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cf250be141971300ed0b77484d5dd1943dbac1d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:50:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b0f7fddc8e24689a61f63d8a5856048\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f766ec6bf843456e88e2f1c16b05240f : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7006432dcc7a4c42bdfd30ee20d3e24e : SUCCESS in 3m 03s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/029d9819e13048fca901b2e5b1b6430b : SUCCESS in 12m 21s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b435938a39634121a82b226302f645e4 : SUCCESS in 12m 38s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9dd15d3c2567456a8a330fc2dede20b1 : SUCCESS in 8m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9452b48313bd490a88afb6036dcff70c : SUCCESS in 15m 54s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/7029f929fdff4f5dbd6b4ef294e59aff : SUCCESS in 11m 53s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/295c8f80ccc846edaca495c292f469d6 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/84664fc721844f6baa73495ada1b320d : SUCCESS in 1h 25m 13s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b7ada4037cec47de88a629595fc17aa8 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc0c90ea168448218b3aaf5f4b0cecb6 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7d795257a7b47ecadfb836290a6b036 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9469384076074f0984db2b1c20c34c00 : SUCCESS in 1h 00m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b3fb1d906b345d188f20adf2c7d6e56 : SUCCESS in 59m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e26ead3bc65142859110d8a17ffd3b92 : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/94ea3c398260436c86c21a58d4dfb93d : SUCCESS in 1h 05m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e498973452ae4322ac4a7d041494e3cd : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f11fdeba802b4b488c2e17b5f06c68b5 : SUCCESS in 53m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddfd32eeeb4f4e9ea11b3f04a4f351b5 : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":7},{"id":"65a25552382046440d06cce9b9e0fc75467b7df7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:50:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4de201fe2d3392b3c37d4c553288605371c94213","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:51:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5162ec295e2940d585061bbe783c5b95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e91d6c66af424486daa017c7678bdb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8b96d32f6b874ec431d2ca2be02f152b7d9cda23","revisions":{"501393978eac77b58af9fa91f132922ce6b2152c":{"kind":"REWORK","_number":1,"created":"2026-02-25 13:52:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/1"}}},"commit":{"parents":[{"commit":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","subject":"Remove deprecated innodb-file-per-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/501393978eac77b58af9fa91f132922ce6b2152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/501393978eac77b58af9fa91f132922ce6b2152c"}]},"branch":"refs/heads/master"},"121430ecce41f7d1b69680c411a9f5b251a6d40e":{"kind":"REWORK","_number":2,"created":"2026-02-25 14:15:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/2"}}},"commit":{"parents":[{"commit":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","subject":"Remove deprecated innodb-file-per-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 14:15:34.000000000","tz":60},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/121430ecce41f7d1b69680c411a9f5b251a6d40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/121430ecce41f7d1b69680c411a9f5b251a6d40e"}]},"branch":"refs/heads/master"},"fb97e4d6cc188b7bd874a547f696bbf85e98d2d7":{"kind":"REWORK","_number":3,"created":"2026-02-25 18:16:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/3"}}},"commit":{"parents":[{"commit":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","subject":"Remove deprecated innodb-file-per-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 18:16:01.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fb97e4d6cc188b7bd874a547f696bbf85e98d2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fb97e4d6cc188b7bd874a547f696bbf85e98d2d7"}]},"branch":"refs/heads/master"},"a6f016bcaca4fb63773a469728ae4d66dc44dc83":{"kind":"REWORK","_number":4,"created":"2026-02-26 07:51:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/4"}}},"commit":{"parents":[{"commit":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","subject":"Remove deprecated innodb-file-per-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-26 07:51:48.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f016bcaca4fb63773a469728ae4d66dc44dc83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f016bcaca4fb63773a469728ae4d66dc44dc83"}]},"branch":"refs/heads/master"},"4520ed7eae3dbb7555253bfa02aabb6817ccc81c":{"kind":"REWORK","_number":5,"created":"2026-02-26 08:09:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/5"}}},"commit":{"parents":[{"commit":"dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc","subject":"Remove deprecated innodb-file-per-table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd7c28671bb1083c0a9c35115d2abb4ed1bc78dc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-26 08:09:07.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4520ed7eae3dbb7555253bfa02aabb6817ccc81c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4520ed7eae3dbb7555253bfa02aabb6817ccc81c"}]},"branch":"refs/heads/master"},"4835739c92250172fdb1dd4ddc692f9f1ce43dfa":{"kind":"REWORK","_number":6,"created":"2026-03-09 08:58:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/6"}}},"commit":{"parents":[{"commit":"b93f76319ae06058f36331f3ee477b0e891c1a3f","subject":"Symlink the role into tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b93f76319ae06058f36331f3ee477b0e891c1a3f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-09 08:58:47.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4835739c92250172fdb1dd4ddc692f9f1ce43dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4835739c92250172fdb1dd4ddc692f9f1ce43dfa"}]},"branch":"refs/heads/master","description":"Rebase"},"8b96d32f6b874ec431d2ca2be02f152b7d9cda23":{"kind":"REWORK","_number":7,"created":"2026-03-09 08:59:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/977963/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/63/977963/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/63/977963/7"}}},"commit":{"parents":[{"commit":"b93f76319ae06058f36331f3ee477b0e891c1a3f","subject":"Symlink the role into tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b93f76319ae06058f36331f3ee477b0e891c1a3f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-09 08:59:26.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8b96d32f6b874ec431d2ca2be02f152b7d9cda23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8b96d32f6b874ec431d2ca2be02f152b7d9cda23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
