)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~979701","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2025.1~I1dfb829df92de5174e6fd123ed8a4717a6972cd6","project":"openstack/openstack-ansible-galera_server","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-03-13 23:02:08.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-13 23:02:08.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-11 16:23:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-13 23:02:08.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-13 20:35:36.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I1dfb829df92de5174e6fd123ed8a4717a6972cd6","subject":"Use READ-COMMITTED for transaction isolation","status":"MERGED","created":"2026-03-09 16:49:22.000000000","updated":"2026-03-13 23:03:00.000000000","submitted":"2026-03-13 23:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979701","cherry_pick_of_change":977963,"cherry_pick_of_patch_set":7,"meta_rev_id":"9746c0ed19a7e5360ded3bbc6fa3c96312fd52af","_number":979701,"virtual_id_number":979701,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 23:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-11 16:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-13 20:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-13 20:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 16:49:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-03-09 16:52:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-03-09 19:08:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-11 16:53:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-03-11 16:53:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-03-11 16:53:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"9c9c7ed1ee09fa55114979a1245bde6be3d3afbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 16:49:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* templates/my.cnf.j2","accounts_in_message":[],"_revision_number":1},{"id":"b17b3cb57ec49f1f6b0383a41d0d0168914540b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 16:52:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f7f579de2f51c82874995461360f2ec6544534a9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 16:52:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"69a4d6ae9a4d86b7748e92ec7c0b068137e73329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:08:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a94ac65faf1c4c85b16ad0045b364bce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aef898f0ef94179bd8b791d46f62d70 : FAILURE in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/552154ad0238431ca313f0d3ab77ee60 : SUCCESS in 2m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acdbfab0a32a448cbca4fa5e0c2ccd62 : SUCCESS in 6m 34s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/13177494386f4f1ba131636595831592 : SUCCESS in 6m 59s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ce9f6c3981e24d409f5c55c04ccdd087 : SUCCESS in 14m 31s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b5a4cd5434643f69efe61b6e61c02a2 : SUCCESS in 9m 13s (non-voting)\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfaefa3e48b54b0a83d4c456cc080638 : SUCCESS in 13m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/743a590371c1478eb6ee1b9ce1817032 : SUCCESS in 11m 20s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/725b42f92d524a5bbef9fb4fd0dcba02 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f39761b55b584a6a9219172c1cb33052 : SUCCESS in 1h 20m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/885cab1a14924b7191b498c41365452e : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ead8ed0071a240e3a4aa14a5f08ce24d : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a48a26f5832b464298b45f98488cbf27 : SUCCESS in 49m 44s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3c2d60b9a754b2aae28211e387ced91 : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/acaf32af4a5b4b36bb1d49f75e95a4a9 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a2b72c9060dc4f2cba45f2eb85f10386 : SUCCESS in 1h 01m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/feb3dea089764f5a82c60b7e5165baea : SUCCESS in 1h 51m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4071cac29aa74cb5bf3b62347abe8b97 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afeb94ac26cb424f9bb66aa99885dad9 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1c3167c255143debd916a8692628dd2 : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9b242130d234c7fbb017fa89fb55e98 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/925f628987664797b7f12d0c0e5892fe : SUCCESS in 44m 34s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca2110735b2a46d9a4d129e25fde6c46 : SUCCESS in 42m 11s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bc7b6f2dae63404c8c4c15b6c9e7af04 : FAILURE in 10m 52s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d5fa754910b440fe92d035ab476522b3 : SUCCESS in 57m 45s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf28ec2c137d4ff6a36aa8c96e095e6f : FAILURE in 8m 00s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30010d561447482b8a55e052d78cdbee : SUCCESS in 1h 15m 00s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78b83f54d815468cbf23a909bd26aaba : SUCCESS in 1h 06m 45s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/82f0b97fa9c14dc8912f7acdfc7aa135 : SUCCESS in 2h 11m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/462d61c9891c4269b11e4c8aa3157e78 : SUCCESS in 1h 40m 54s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95abcc4b74144785a32eee829b5cde1a : SUCCESS in 1h 46m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0dfd21c8fa557fc7a2a5de395375345f82ed96df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 12:28:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ca972b99e9af2e55deefc8bbc1ed8da8c8f94330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 14:51:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a54b9535327e4a37bedaabd0f6c2d84a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4df335b6f346fc86785e274d45fb22 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a4c6630deab4f838dc825565cccb51a : SUCCESS in 2m 54s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63e1ce33b38c4e4fba617de0b5640cb4 : SUCCESS in 11m 52s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/a221a6cffc6749efbbea34460a42b9d1 : SUCCESS in 11m 25s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c6ace00ab33496f80a3a47ed691211d : SUCCESS in 12m 44s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb68165fa630470bb287c087bb8d4f4f : FAILURE in 4m 25s (non-voting)\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7ed550144e44274beb6a0a0f969de85 : SUCCESS in 11m 27s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc86b66ec9bc4707a47f40713b5f289e : SUCCESS in 12m 45s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/bd7537fb5db84d68b73e008cabd10bda : SUCCESS in 10m 28s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc0f4e91ea6c496099097ecaf7d3a667 : SUCCESS in 1h 21m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/893aa79b9a9f4c1995c88fea1f928d1f : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8648114c9c1422ab47ab44832abad02 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/32df8ab510044a0bb91ca093027b8f8f : SUCCESS in 48m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bf964caf9bee4f42ae77a89e885d5ebb : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7503f1c1de49483e94af972a10a9e5eb : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/88455a2670ff4d0787797d597b0c41bc : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5ce2e0d63e74df6bcfd2162aa806711 : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aec95434c9dc400596cf73eb8b4a3f65 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e8d175c843b4338afe2397488959687 : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db83e68853c4238be8a77612da0e9cb : SUCCESS in 2h 15m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/793ebb6ad6dc4382ab6e2d55503aaac8 : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f12d97f836ae41b48fb3f5ad3ad0c419 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56a1b2818e944e6aa3af2de5b38f1cf5 : SUCCESS in 1h 09m 21s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/eb9b92991c91406b9b50b69a69117b32 : FAILURE in 16m 44s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4e80ec804cd54636a9a74dec43b8359f : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/490a34789b924689b5932e9c24e2e749 : FAILURE in 14m 54s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad227adced004a038b87bdba80fc46c8 : SUCCESS in 1h 35m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81f6fe8b84fc4e629dc807a9bf817a6d : SUCCESS in 2h 05m 50s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/775be02117ec48c8b9520426b76b478d : SUCCESS in 2h 07m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f47e67b83a147eb95264c06536c4ba1 : SUCCESS in 1h 41m 14s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b57ef3f02deb465fa75cd07b8d4e54e9 : SUCCESS in 2h 00m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c70eb91b7f87584cce441934f1df9e102e7286a8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-11 16:23:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d9b1fd93d2b93d3910c5d9527e8f60aefef9ab7a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-13 20:35:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"377b9203de0fc22a9a01677d0ebf2ac441082de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 20:37:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e25e777f892be6066b3badf1c34e17946caa2b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 23:02:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25de192d1ab4a06a3d27619f43247eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46fdef6a8db4f80b7b803ef05390c64 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55fa3768945a42feae20330bb4c4d998 : SUCCESS in 2m 58s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29d59dff7a6f40a8b3b2264ef6a589d7 : SUCCESS in 13m 03s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/219f723359854fd39c54e92025550ea0 : SUCCESS in 9m 02s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/63e2815e41014c25848f191f8f7f32b9 : SUCCESS in 11m 33s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c33027ddfd4424983ca923dfcb2e00 : SUCCESS in 6m 42s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f4eef6408645cb8bbf2b2e2a5398f6 : SUCCESS in 7m 44s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/728f28f9ca5b43418448c713a3840622 : SUCCESS in 10m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c0ad438e69f491e8d05b322b6e4b155 : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63568b65225a4dee867ecb5aaa327796 : SUCCESS in 1h 37m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/396ba26253c44217b33484601cd2ca0b : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f75948f5dcf1484784503a5dee3503c0 : SUCCESS in 52m 51s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/05eaab66d9bb4365b1ed214f86211360 : SUCCESS in 46m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/fe80a2ab818844e885edb55c8b3db9ae : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b63cf44f08264c8fa13a1e7d5df80594 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1d75db7821440a09f176b47f6175155 : SUCCESS in 50m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/572a2b841baf401485531f80fafd00b3 : SUCCESS in 47m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/182f0f721e5849a6a33ad03f109548bc : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07a1c927bbb449a3891638a8ad39abb8 : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb04785b18e9419c86a3c831f4a42285 : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b32bfc982701403ca8869f292d42336f : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/58d3f3bec1ff4d15a058e140e04d604f : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63200948a8a74bdeaa33140d48d2660f : SUCCESS in 1h 06m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/065274ab545b4761b3a4a82e22eebb38 : SUCCESS in 1h 58m 30s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a730f085f8a24fb3bcb1df7211406832 : SUCCESS in 2h 13m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19614a8223fd43408e4d2ede77e87e60 : SUCCESS in 1h 36m 40s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42701d1cb0b042f793da8e3b8df09083 : SUCCESS in 1h 53m 39s","accounts_in_message":[],"_revision_number":3},{"id":"abda2ea308b689add5b0e5c7b4eb4d0754ffbf50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 23:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9746c0ed19a7e5360ded3bbc6fa3c96312fd52af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 23:03:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf80fd8b1f8c407fb4d6d13e3e1ddf86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2a63b628224cf78ee8e781297948b0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91bffa5ed21513c6affd07ed264672a4b5f3242c","revisions":{"9c5d8a651f81699a2a268958c795bf012f845ce3":{"kind":"REWORK","_number":1,"created":"2026-03-09 16:49:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/979701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/979701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/1"}}},"commit":{"parents":[{"commit":"731ace7ab12ad09ef1c0945c3a2719c18ed9e88a","subject":"Add mariadb-client-utils to packages for removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/731ace7ab12ad09ef1c0945c3a2719c18ed9e88a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 16:49:22.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9c5d8a651f81699a2a268958c795bf012f845ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9c5d8a651f81699a2a268958c795bf012f845ce3"}]},"branch":"refs/heads/stable/2025.1"},"6505213388f22ae3422331ce1a6abd99c2a85388":{"kind":"REWORK","_number":2,"created":"2026-03-09 16:52:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/979701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/979701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/2"}}},"commit":{"parents":[{"commit":"731ace7ab12ad09ef1c0945c3a2719c18ed9e88a","subject":"Add mariadb-client-utils to packages for removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/731ace7ab12ad09ef1c0945c3a2719c18ed9e88a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 16:52:31.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6505213388f22ae3422331ce1a6abd99c2a85388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6505213388f22ae3422331ce1a6abd99c2a85388"}]},"branch":"refs/heads/stable/2025.1"},"91bffa5ed21513c6affd07ed264672a4b5f3242c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-10 12:28:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/01/979701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/979701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/979701/3"}}},"commit":{"parents":[{"commit":"775fe8d7fefa8645eebe95e219a3f20a2caaf2f0","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/775fe8d7fefa8645eebe95e219a3f20a2caaf2f0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-25 13:52:49.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-10 12:28:05.000000000","tz":0},"subject":"Use READ-COMMITTED for transaction isolation","message":"Use READ-COMMITTED for transaction isolation\n\nSince MariaDB 11.1.1 an new variable transaction_isolation  [1]\nhas been introduced, with default value of REPEATABLE-READ\n\nHowever, this seemingly creates issues for oslo.db, as in tempest we can\nsee failures on instance deletion like [2]\n\nIt seems that for a quite some time, assume isolation is READ-COMMITTED,\nwith a very good description of the issue available for the Mistral [3]\n\nIt is proposed to explicitly set the transaction isolation to\nREAD-COMMITTED to avoid issues in deployments.\n\n[1] https://mariadb.com/docs/server/server-management/variables-and-modes/server-system-variables#transaction_isolation\n[2] https://paste.openstack.org/show/bkY2vqKLS5HpUDzesZnN/\n[3] https://opendev.org/openstack/mistral/src/branch/master/releasenotes/notes/changing-isolation-level-to-read-committed-7080833ad284b901.yaml\n\nChange-Id: I1dfb829df92de5174e6fd123ed8a4717a6972cd6\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91bffa5ed21513c6affd07ed264672a4b5f3242c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91bffa5ed21513c6affd07ed264672a4b5f3242c"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
