)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~984312","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6d70d1d207a077035befdd5cb3ad5ff886ef1a81","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-06-02 00:41:41.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-08 15:10:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-02 00:41:41.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-20 19:26:55.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I6d70d1d207a077035befdd5cb3ad5ff886ef1a81","subject":"Remove galera_disable_privatedevices","status":"MERGED","created":"2026-04-13 07:39:44.000000000","updated":"2026-06-02 00:43:23.000000000","submitted":"2026-06-02 00:41:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984312","meta_rev_id":"08b5bf80b694746ce864b5bd898e450e717f7eff","_number":984312,"virtual_id_number":984312,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 00:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-21 16:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 22:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-25 09:35:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 10:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 12:54:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-05-08 15:11:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-05-08 15:11:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f32a1167c531a68800a1bf38d351019229310c39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 07:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f48f89bb0af39d78dcc0fa9fdff1cacff28f974","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 07:45:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b921692e418b4c5144e72958920a3ed861bc6fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:50:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f73034384a44ea59c7d619084b1be25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85821b18f6414496901d2e53bb4aebcf : SUCCESS in 6m 30s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a61b01815da4ebab6ee33770612335e : SUCCESS in 13m 42s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3c632a21092b4f49bd4448a179b3cfb4 : SUCCESS in 11m 20s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/122654babebf4d34b1c556858a81624a : SUCCESS in 10m 48s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c25625403a9145caa96b8b078cba0825 : SUCCESS in 12m 18s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d6605b43c1454679ae170f354f641e8f : SUCCESS in 11m 51s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/178682e10a384b558b5a5da88164cf27 : SUCCESS in 11m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4a68382829e24b8e9738c73d249e14b7 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f0fb2fd224647339c25180bc9693999 : SUCCESS in 1h 29m 29s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0cc4d8be2907473b9be0c0290f332687 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a379bfa8757047a3b60254d6ade11e03 : SUCCESS in 1h 51m 20s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bd1e9c10f6344bbabaf820fb627f269b : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac23eebb6f8343e4868a4625eec2d312 : SUCCESS in 1h 45m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/160452b65e7345c1948734ccb1beae3b : SUCCESS in 1h 48m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec25147c48694f7abffb5e48153ef05b : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/efc1a05e1c3846d5ae063dab4f3f67c1 : FAILURE in 12m 45s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9eb5c9a6bf14fdfb5c8a3e7cd13950e : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/15f16c832b9b43a4a7f38dea5c41dff8 : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d170f6c5f90a482491c8b65d4b3977d5 : FAILURE in 20m 47s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/444170981ad3473e8f08241bc97d22d4 : SUCCESS in 1h 14m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0e7e0952dfe40aa86034d2fa7283d6f : SUCCESS in 2h 17m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c59d98dc76ac0c48deccd8c76e88843f62ccaf34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 12:54:14.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"60957e88b6246e5dac073ce452f5bda538c65128","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-13 12:54:34.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"fc57b4c6df2eae8b3a963baaf643f550594604d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:13:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfddb1f30c7744f78342f6160de82444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd513281a98c44709a8bb38ae7ed6e54 : SUCCESS in 7m 13s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d900d57cc2284f189ca0fe6405ecb26e : SUCCESS in 14m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ab3b46dec5124ced966ed307b6b2129e : SUCCESS in 13m 54s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/01ce5709b09e454da116f976d865a575 : SUCCESS in 12m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a725568dc77b4a28aa00c80ffe41cc54 : SUCCESS in 14m 18s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/b9997aa742fd4acfb267c005d7a2adfb : SUCCESS in 10m 43s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bb18e62dbd124fd5a0bdde5a6423aa2f : SUCCESS in 13m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/75364d800daa413b8e307090cbd38669 : SUCCESS in 53m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/27e1898aff2e463ab675cede604c0fae : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2e75ce7fea2e4ccfb2cb7406db85e0ab : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ce3cf4f953d44c7bb0b2b28bf3c3882b : SUCCESS in 1h 34m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5d47f7029a614165bd35f7812eb930b5 : SUCCESS in 1h 39m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f273fef571f04041933647e1cc6e9de1 : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac637eca65a24e24ad398390246032f9 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a62b673b53c4af0ae159882678975db : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c62d77734dae4327936ef9e4b7573814 : FAILURE in 11m 39s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8cceaeb7ff2141f39c2036e797c20a03 : SUCCESS in 1h 03m 08s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2abd4fb81b6749e19003b09e642387f9 : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c44bc4dbebba41e8bdf031fa57b3ce3c : FAILURE in 14m 11s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51321b6d43614836aca8a0f6c099b7df : SUCCESS in 1h 05m 07s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/800e30e92d9242adb9896434fe3ab622 : SUCCESS in 2h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"852a3715d79efa453c10a0e683d0af4f70dc90bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-08 15:10:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f6ff724a02abaf726e71709fe4978cfa75d2a4f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-08 15:11:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a15efcf37f09a0cf95ff717382263ce0915cb443","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-13 08:27:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c6e18fc5a27ff15e1371d5ad9ed8dee8d10c758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:30:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b2dc5ffceba7e543508a17e35b33792b7f967f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:56:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5521fb5a4824dc491e55d35c8a52458\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df07ac33ed574ee99db9be8a8e40b3df : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b52e1fc7d34326a304487c0580462a : SUCCESS in 3m 40s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4a0f5f4f6014644a527f25ed34a9efd : SUCCESS in 15m 22s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/638e3519800848d68c87f146fe0084fa : SUCCESS in 13m 08s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/435d15a2f95a4ad28a8de7c822d5197b : SUCCESS in 14m 53s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8f4ef1512a3432c9af1d56106cd635a : SUCCESS in 10m 22s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2ff9ef272a2640848c26fd3293810917 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c35f29f898a14b80a8fdcfcb40f9d1e1 : FAILURE in 2h 25m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a7a2015401c941bcaeda1acea3ec1657 : SUCCESS in 2h 01m 48s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0bfa09105d0a45faaa7f82051ac161b8 : SUCCESS in 2h 12m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/03ec5e673d6142fc87dd39facaafca96 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2f0cbbcdb2614b7bbc20351fc2f56ef2 : SUCCESS in 2h 15m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a06c57c80aa4331b87a06740fdba886 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daf86932f44848dba1c7c95c86dc1fa3 : SUCCESS in 2h 19m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789bd385365f41ee862f017e3fa93bd1 : SUCCESS in 1h 31m 21s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2aaa35a183df4bac9d9e53d4d7e52bfd : SUCCESS in 1h 54m 21s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f547870f4a5c405abffb8f2fdb033e26 : SUCCESS in 1h 40m 36s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1c878efef8f4af6b4494cda0eaef07f : SUCCESS in 1h 45m 00s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f3213946f3f42bd9512763b4613b621 : SUCCESS in 1h 42m 03s","accounts_in_message":[],"_revision_number":3},{"id":"65548a84e199ba1228ad46f1eb84e3d33c357467","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 13:58:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"571a2d2c3c8c452141e7364f5357a138beaf1990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:02:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08e5ffae648345c794ceb7901734edf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a644b286e39b49a9abf4243dbcbbba58 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9d8f4daca86422694190cb02582b62f : SUCCESS in 3m 34s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc19b8f8ff04ca8b0ba4c59165d0a42 : SUCCESS in 12m 20s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/e981c8ef32a4418a9ac0242d6d1fe2ba : SUCCESS in 12m 38s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c62a2190b66b45bfaa1618f2b00df48f : SUCCESS in 14m 05s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92b150d3b53446259ba184847391ae50 : SUCCESS in 11m 53s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f32f4666e21e45e0be822ee96f470d2c : FAILURE in 6m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/77273658bdbb4385b5818ee555a4da9e : SUCCESS in 10m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6d5d173e793c4f08832021f1268eb0fe : SUCCESS in 14m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a15013ddb95c42a2ada891b842504737 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a69af59189754e638f1e47c24626607d : SUCCESS in 2h 08m 32s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8b0a62d966e5412dafe94cd887fb5bc3 : SUCCESS in 1h 06m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/58c3f87d0c8b487f8757f13162a7ad23 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/859955cad3bd419bbd37aa22512f9379 : FAILURE in 1h 40m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39b9f96b28e148f7a47f494c0e694ebd : SUCCESS in 1h 54m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bedfddbb2b554538b4fa7a087cf49ede : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43d43b4744c244d9b2128d25cfd6d57e : SUCCESS in 1h 58m 09s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0286fc1f017c48ceac709faa2266ad39 : FAILURE in 13m 08s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f0cf457738ac4d7aaa41aeb9ec5df18a : SUCCESS in 2h 01m 05s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/69d88e3f322e42df85b35dca8ed9c25e : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9510fb21e6594316b869876ae314b561 : FAILURE in 10m 17s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dede79eb1974324a67be26e057dca32 : SUCCESS in 1h 53m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b311e0990f8e45039d2c996f28999a80 : SUCCESS in 2h 51m 13s","accounts_in_message":[],"_revision_number":3},{"id":"7a79407a5c6ff7c51d55358b838ce5288b12e557","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-16 20:23:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5a79e26ae9738b117b030e33f8c3d9e12c5d65ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 22:49:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/270f8a8d068342028f45ca5135137b20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bef7375ca1f499a9177e8ce11e35f74 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0830f83b4ac64ddf97b1ef15439e939c : SUCCESS in 4m 57s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6c71a00ae9a413a8014c7ffc540196a : SUCCESS in 8m 57s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/7716eb93f07a4a4eba6ac8c511cd1b11 : SUCCESS in 12m 32s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c3caa0903af248aea6560407470e05f0 : SUCCESS in 9m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe89764455ca4e25b3ec8972bb56e134 : SUCCESS in 9m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2b6bb03b70554530a05caf738167ee06 : FAILURE in 4m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f3f9889ee6be484f8a4d71ba5639b1f8 : SUCCESS in 9m 48s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d413267532de4a26b4c0639bd3933095 : SUCCESS in 7m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/303c49f6b380484185c009df4620a3ee : FAILURE in 1h 41m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/558d2c1d9e974e6298f1c95ac796c543 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9f87d491c52d4469a5668686e32af1e3 : SUCCESS in 1h 07m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74840962c56146c1b3af5b28b7c07024 : SUCCESS in 2h 00m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17586a465d3f4abe80bf1113df8fc902 : SUCCESS in 1h 38m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd229a1c32a483b86387da2079f83da : SUCCESS in 56m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62271f9134604b1a8aedb7d4e4a2b300 : SUCCESS in 1h 50m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af38af2784a1460c8e31643a475aa2df : SUCCESS in 1h 57m 16s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9bc29d86ebd481f9f58ebe462b1287c : FAILURE in 12m 47s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0fee2aa1c7674f3fb7713fe2d6c2d09f : SUCCESS in 54m 50s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/89fe9f25c47846cb969bd9c745f89162 : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b892ecfdf63a47c7b84ae85c6f27df44 : FAILURE in 9m 40s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3071ab40796647798d2cda6b12b7a20a : SUCCESS in 1h 40m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dcbaa810d3247b88a28decf66ce9f91 : SUCCESS in 2h 18m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f9a0602031558a6abcadbf699bd28e63d49c051a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:12:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c361d06313a1064f44214de32ead4ca8d6ad2f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 22:44:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac5a9cb728924bdbaec1266dfd549791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fdd8f4cb5b4e0680653027387f54e7 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ce9d3073db424c8ba82df95a6e366d : SUCCESS in 2m 47s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55e34196a4744c78b08d334223bd173a : SUCCESS in 11m 08s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/56ca219fa46f45fe8bb2a16249380a88 : SUCCESS in 6m 36s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3421c88a123a4bee9bc8c7f996f035fb : SUCCESS in 12m 36s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3babc31fcc445bfb9239d30366a2cdf : SUCCESS in 15m 24s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/aeefe167ece24a0a9653b2e74631213c : FAILURE in 5m 49s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f3531a68b379416fa976f7a45e89a2ab : SUCCESS in 11m 36s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a3541567285747c0b9f4946ad1d8c174 : FAILURE in 5m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b481a05e341946ea8a4227fbc0aecc43 : FAILURE in 2h 13m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f6f5bb7f6bd34f17b24d50e33db2dc2f : SUCCESS in 1h 47m 29s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/653874c675454111853343f1cc126ab8 : SUCCESS in 1h 46m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d4a9e2efcae4022a1b1327acaca6841 : FAILURE in 36m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8289b60b27974ecebad40ca0d7c13844 : SUCCESS in 1h 59m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e09303c928749eea31b1f83fa67a375 : SUCCESS in 1h 51m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0550fe44c8f44fb796ee3fc50ae16fc2 : SUCCESS in 1h 49m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aae75d79d0cf4fbda709ec6bb1b0baec : SUCCESS in 2h 00m 50s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f1f6a1c618743f5861fe76e3a61952a : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/284fefd912224989a82caf6a9cc54941 : SUCCESS in 1h 50m 08s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9ab76a61175e47ffa67d007a77f4e7e9 : SUCCESS in 1h 47m 40s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b545762352ff4e9fa3fe6b4ff276ef4c : FAILURE in 7m 40s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1be57cc9be045108e8e01cf1386bbb5 : SUCCESS in 2h 13m 18s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d2e9366382544e8b8dadfb4b99b208c : SUCCESS in 2h 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2b3ad9d4e3b3d79db69d73a90ddfff535b177879","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 15:59:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db7f5b5a4332df1838cdc2d12cee26f56c673bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 18:34:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47185615b7984133baef2c05d1fc3b7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cb813576744e3ba8f2c17dc5bac6e4 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8bfbe154ed146348c3cd96fb944e0d8 : SUCCESS in 3m 39s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f3b66b019b442dbd764235ef514ec3 : SUCCESS in 14m 13s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/86b5ffa3ae1f45d5a5d9351e4fa6f31f : SUCCESS in 13m 37s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/fe1ae86a76324aa3bfb7012c4fe742d3 : SUCCESS in 14m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fccc0934b4b4874b41526818efa6d5b : SUCCESS in 13m 23s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2fcadf55dfe2435581047f7357a28241 : FAILURE in 6m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/df6859e0d4954cde86a28f6e0c4cdc3d : SUCCESS in 10m 53s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aca2c5be07f34e56b7e8b19da3858802 : SUCCESS in 7m 51s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/73f25bfdba4442aeb823d315b195f336 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/52fec7f025d5492c8075ca0bca2250f5 : SUCCESS in 1h 45m 35s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6598480584a148d293e2edfcfbd0f957 : SUCCESS in 2h 01m 03s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7adff6529bdd4c97a741cddd831633a2 : SUCCESS in 1h 51m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee5cb12500394c888310788e20c7dd94 : SUCCESS in 1h 50m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a34b3fca162145268b88bbf3f4e1007c : SUCCESS in 1h 52m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f5ad6aa352f4c9caca1fb4ee43c45b9 : SUCCESS in 1h 32m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0ff102d891e4053bb954a3d90edc020 : SUCCESS in 1h 59m 46s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f64c87f0b84b403c865011dce58ab634 : FAILURE in 11m 19s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4556557acec44f7ade0426bde97c8ea : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/38ec5b22c2bc43cdbda88c05dd288315 : SUCCESS in 1h 43m 40s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05e412bc63394ce280edbe7f3800fa99 : FAILURE in 7m 29s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb6f55b581aa443488eaed8b47de5db3 : SUCCESS in 1h 29m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8feb36285874eaaa94add0ba8938bec : SUCCESS in 2h 23m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ed772e9c133acf1d9beb5c5fea5bf3cbf1fa03b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 19:26:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1d47759f1f11387abf8afcb1749a5fb36615a21a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 21:54:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f96a48a8e9dc4b539a22ac69cfcdc98d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab509e4e78e4dfab1a0b4c48c169834 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3480a07d419e4ad790d4dd01e252d97b : SUCCESS in 2m 49s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3558675c2b146308691644f165fa957 : SUCCESS in 6m 45s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/967ca92f8687463f86b9496c7c56a0d7 : SUCCESS in 11m 20s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/29823ab522aa4e479bbaa44ad6627f66 : SUCCESS in 11m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5fb8fd1a067418cb793ceb32600106c : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/683dfc5ceac848e29d7bf5486101ff79 : FAILURE in 4m 06s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a9f19708362a43cd992328607b706d41 : SUCCESS in 9m 36s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/08b8808368b54727917e6bc3e75e8564 : FAILURE in 6m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/855c663bd5434ebd8b5233984974200a : FAILURE in 2h 15m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a91385ee73bf4fbc8130a41c0fa44410 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0b4133496f1144948f5758f34fadfe2d : SUCCESS in 1h 34m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/58a6b70a834c4c169dadce3eee158518 : SUCCESS in 1h 33m 32s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/772b014833e54b2589ee2ee36d8f4d66 : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e94c9d78a8f4721a734428cba0847c9 : SUCCESS in 56m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1379babcd49d4e8d89b4c88f2a883982 : SUCCESS in 1h 53m 17s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec336af7448b4dfd9c6380f8b3594bcf : SUCCESS in 1h 53m 22s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/43d01e16aa3f4f9e960cd9558164946b : FAILURE in 12m 06s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f8892cd00fe0426f9ed21cfc47b2f2d3 : SUCCESS in 1h 43m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/db7a5b06551a4b928b592002c5dc17bc : SUCCESS in 1h 49m 15s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bd899a8b63274d9b94ac00eb3caf9df1 : FAILURE in 13m 23s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7cb97b63d054d07a1c8f0f3242c8bea : SUCCESS in 1h 47m 27s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/defeb08592154121988b6e0d19a7f644 : SUCCESS in 2h 21m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a0fa372cc7c3611e4b9d0fb9c0987f4f1e73d222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 22:08:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"db91e7a68cfaf0165d3f554ff210de0f0521552e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 00:38:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65523b17714f403cb2364646a2086184\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33fc05995244e4ea02260f6fcc3d131 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a4d0fbd8181457386cd3d2cb342a716 : SUCCESS in 2m 38s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bac2215c966646c98365b0259c26ff4c : SUCCESS in 6m 27s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/68dc261ba60646509677ff10d3dc9291 : SUCCESS in 9m 28s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/8459e89947004d52bd13c2dbba2a796a : SUCCESS in 12m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/440b0de75eaf4c7492eb8354ea47c608 : SUCCESS in 14m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e69f913297914fe1bbbc7e4299b4e2c4 : SUCCESS in 6m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d6de7b326467496081bf989aa2924541 : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9bddb7b18ab441528dddabdd4b8022ee : SUCCESS in 7m 01s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3c7e0cf31d744a8e919e40ac95a815e8 : FAILURE in 1h 06m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b7b9a4ba5e754cc1aa85869eb877f2e1 : SUCCESS in 56m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2c93773448eb46a1865085486dcf6c04 : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8368b16256144d58b4bb9e1d4bbf121a : SUCCESS in 2h 18m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6916e86478cf45ffac4debb101198096 : SUCCESS in 1h 39m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf3c57fb654e48e8bb86a724330c5e03 : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f17a60009d084dfb89fd66653db44eb8 : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2f7cb8fa8fe4eeabad25303227e685f : SUCCESS in 1h 50m 51s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5aa6829f49374a449f3914c1c68a398f : FAILURE in 8m 56s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/63cb1c3eeb3d445780d09b1ff4597611 : SUCCESS in 1h 51m 24s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/675c6b8f7c674cddb27cf262f5f36931 : SUCCESS in 58m 19s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6937a95d4a9441fb7e9bb9a0d2b97da : FAILURE in 14m 11s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aed440a7ba4f4f57a4fda052a4b0585d : SUCCESS in 54m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a32d910a224a73afcaae732a5a1363 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":5},{"id":"431f58f32e110bb50e7989dd271dba0384e18566","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-21 16:14:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"20805e1c79477e50f934a1c92ef65bc34be3764b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 19:11:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2a18c9bb39f4c13b1fd0a5eb92667d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936c4d1633d446e9bfad48e49147bd1c : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cb6912baa5243c887135e50e48316ce : SUCCESS in 3m 19s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d4e5d456d564ab4abb3e6f5dba6b7fc : SUCCESS in 12m 19s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a9a1212a7ca4ce8b07112f696c4e2c5 : SUCCESS in 13m 44s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ca846ede3a0548bca748b650a2fb7ed6 : SUCCESS in 7m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24b76c48f7b04a6894cc6860d4907030 : SUCCESS in 7m 58s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ceb8e6bf138b41f9b8f26073094d1e23 : SUCCESS in 12m 06s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/3a72391656024eb49836736f65a7712f : SUCCESS in 11m 13s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74e9f382ca26405f9335b7151d5025f7 : FAILURE in 6m 02s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f2e7f423f51c41978c721e57e79bcac8 : FAILURE in 1h 17m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b5cb746cfb89403dbd51f4c193ee53be : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3446045107db45eeae705adf0bf717de : SUCCESS in 2h 00m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b71b7289bc2a42699d6e2007d07ef1d8 : SUCCESS in 1h 25m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aada605dd54341c39fed442f36c28ad3 : FAILURE in 1h 33m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b79210bddd59490fbb8e7d5b0d191a36 : SUCCESS in 1h 58m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/460444ad6c6e4cd0a4c302e1f9a2790e : SUCCESS in 1h 06m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/095ef2fe18e84e72968eafba376b0ada : SUCCESS in 2h 04m 49s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/971426abcfa84716a03f098238b6664d : FAILURE in 12m 47s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/aa47beff3b5a48f9843c44ea63747bf7 : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7fca5e5db41741feb7a307175d4a436a : SUCCESS in 1h 45m 30s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/073d45a0e8334304a268976d6735cbb6 : FAILURE in 10m 07s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d8bf3dcb7e4037b59e5f34d037d3cf : SUCCESS in 1h 50m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4edb12b4d17847ff8037c21e17e25b42 : SUCCESS in 2h 37m 00s","accounts_in_message":[],"_revision_number":5},{"id":"0b332c3d7f92058230fdb2b775d0564d7f4fd004","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 21:10:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"97c074beb3d20e65f077f55ea7bccd0633aba798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 23:56:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc40c6080b843879b1782c698714646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d4c65fc99140f5b16da54ec9c44320 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fd91ed4534047e6b279f294a1698d80 : SUCCESS in 1m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8e6cf94e066416d8f43e078929c9472 : SUCCESS in 11m 04s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/dbc4f881ade04134ab26f5627a0a59a3 : SUCCESS in 11m 41s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ce7cb29f94424803a9033247cc7bccef : SUCCESS in 14m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffcb14cf5002408ba2351d252b34b348 : SUCCESS in 12m 39s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/775b2ae514ac4eb3b7f3a55cd692ff21 : SUCCESS in 9m 24s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2e8ddf460ae647459e5ac4738c875085 : SUCCESS in 9m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1202d04f1ac04ff780ea28779115aa5c : SUCCESS in 7m 22s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/42c544cb5c69456fbaa87cab72f79565 : SUCCESS in 2h 19m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9bfc309cd34f470293600ff9c7b6e24e : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/09179bcb8e16447e8aa331c5ef756a58 : SUCCESS in 1h 54m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3ab744190ca747ce8b6344af9a2c5dcd : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2fbff140319043a889ee4b8c685e3797 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/987f4b6ad683475bbea0e50ff5169d02 : SUCCESS in 1h 30m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69348de3307f4259aa23cf2711f8abb1 : SUCCESS in 1h 52m 51s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f96f563a33874ec59e5d168411b9b8ee : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c2b19e8c5b564255b33e9e7068d1827e : FAILURE in 6m 44s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1bb1fc2763c648d6a40eb1fc00bf8ecd : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/cf32bc05a7504121a4a9940d3c12cafa : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86a80d98985b4a88962240dc14b26f80 : FAILURE in 16m 02s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b5597885fb04cdca15a91ffbcb91b74 : SUCCESS in 1h 47m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a4c8bef586f4f30a6caa3b8caa2d157 : SUCCESS in 2h 38m 02s","accounts_in_message":[],"_revision_number":5},{"id":"49e52e3893c09ba0e2bc253a516827dca254513d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-25 09:35:15.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c54ce3e966049866ad0306ab08198ea1afdb5921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 09:37:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b6656868209a471db82de31ef38604a58e941469","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 11:46:23.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7709735040fe4f4d9536a200cf0cc8b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a8130b881d40d3b99d529e5c9cbb72 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5234dc11701d46ecbc431ff5f2957d1c : SUCCESS in 4m 03s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6b0e7f1975a4c3280709181f8ab2a82 : SUCCESS in 9m 20s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/abf9394085e44909afe3979f9a4a1de6 : SUCCESS in 10m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/cc61d8c643e04a9987ddeb870e9d499f : SUCCESS in 26m 25s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d919fac0efa41bab6ccee553e3b54a4 : SUCCESS in 12m 31s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6e0514612c0d47df818b4fba399c7148 : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/47e21d321fae44e9802c2b192cb3d891 : SUCCESS in 10m 27s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d76610a0690540e2aea53331b0cc6f17 : FAILURE in 2h 02m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/14d4cd3e07894a4a88dd74c27a1ea07e : SUCCESS in 1h 50m 51s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/06ae1bed2ee041c8b37312f86538b77d : SUCCESS in 1h 39m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/608d04c586474818ab9bd0a0ed53abd8 : SUCCESS in 1h 49m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e535f33dff7245c9a2b72d0b34535583 : SUCCESS in 1h 58m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7221f004a81b4a2bacd5bbb1b7cb3175 : SUCCESS in 1h 04m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0d3e76b88ef4559bfb673bce2fd12c7 : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/714521b3c16c4eb590722585c8120c3f : SUCCESS in 1h 56m 34s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cab2d459e7cb41059d337453fed2a46e : SUCCESS in 1h 46m 56s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5bd9103d74444ddfa0b763567b271439 : SUCCESS in 1h 47m 39s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22f8696f30424f3cbbb8a82ca2aaf04e : SUCCESS in 1h 47m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1755909e7a874df38504d80d8e91c14c : FAILURE in 1h 10m 23s","accounts_in_message":[],"_revision_number":5},{"id":"435162d6a71355f01d3fe0548f6e2aa98c0b7ef6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-25 12:30:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"878f17683b375f8ab989d61b11fd19d2a67e6696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 15:00:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bda84712334fcb9029c0eebb3aca73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fcfb052d5b4cbeaba458774e1d6f98 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2508c4ddbc549d0891e8c2ce5a5fcc3 : SUCCESS in 3m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf1e37214a344ac7bcd9cf40da353c8c : SUCCESS in 16m 12s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/85fe3a9bf6b4463aa1fde4f3a7476335 : SUCCESS in 6m 57s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/416517ddd2a44e37a230c17ce260db8f : SUCCESS in 11m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bde723e02da34eaabe56724c60f94513 : SUCCESS in 11m 08s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cb1f608125fd4110b8de4ab5324aa3d2 : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/21d365fdfc15440ba3b29fd2f0750711 : SUCCESS in 10m 49s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2feb18eec4fd4da3b900c9e62bd64ae2 : SUCCESS in 24m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4ec7aa244ed1419fa573d4b73e963e18 : FAILURE in 2h 20m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/0be83ecd8c5c4b4a97b8f60f7352a550 : SUCCESS in 1h 54m 56s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6c8485533aa04261926b36175c714bc9 : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9fc22011df44fbd8fd488912d22c97f : SUCCESS in 1h 13m 27s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc6b86c332cc40f68aa4849dda8293c7 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b76c1676c1e84ea7894daa6aec7a90d8 : SUCCESS in 1h 55m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c70c4d4eb24bc8898cb0fec281656e : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b2f6bfc03f48c9ab5843bc48dcde6d : SUCCESS in 1h 55m 49s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/78169ab4e4b7465da7d38a6532d6dc88 : FAILURE in 14m 42s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d5496242177436d925b8e11be0e4556 : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7107d4562d5742758ec729302f500ba9 : SUCCESS in 1h 47m 45s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c44b7ed69374403d9548fdb5589b0168 : FAILURE in 12m 09s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2333ca8b7f84f1d9aa6bda32a357f5a : SUCCESS in 1h 49m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aeb8c8f00fa46439ad16475272cecc8 : SUCCESS in 2h 14m 30s","accounts_in_message":[],"_revision_number":5},{"id":"5aa996fc750b58495c159007c08df2de300794b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 14:02:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d38dcc742fe39482370779064438f4f8ffe20f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:05:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c8f8038d92c40f4bb35c2b278cab4b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8965140754d144db8ca24b0487319d7a : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4df181814b94b039e7126260461b4c3 : SUCCESS in 3m 40s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57889d8122f4448cb4a0e3e29d3b664b : SUCCESS in 13m 06s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea1e0f534f5044eb823c2df868d8ed65 : SUCCESS in 13m 32s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3a8e6779ea924155afb54083b6806afd : SUCCESS in 11m 45s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0ffa22ee124c998d5ea3bc1318a400 : SUCCESS in 13m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/64969bb361ce4b359ebafe93e8b1aab9 : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/72f5953d4e3f4e62a85fecb8c2a8860c : SUCCESS in 10m 36s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/02ec66707bfe4c619865fc7187489aaf : FAILURE in 5m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4d7fbdea38e04e39ab99c81b85290334 : FAILURE in 2h 56m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bb9db60aa58d4fd88b2456b28a0df7f4 : SUCCESS in 1h 56m 56s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6fc7531592ff4bd88a1dc4f93247ceb9 : SUCCESS in 2h 07m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ff40529a468f47e3b30fba1741c0c3bb : SUCCESS in 1h 53m 03s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/efaf67c559ea43ef83038f6093688a38 : FAILURE in 1h 30m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f141857592a342afb15afb07aea8b204 : SUCCESS in 1h 58m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9a84b9a51004579b43a5fb9a2b2c314 : SUCCESS in 2h 06m 24s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/639fc2af57004733bb50b5adf72ff468 : SUCCESS in 2h 03m 41s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f8c2e3564689470e8b7f14484000d2a1 : SUCCESS in 1h 49m 04s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7add67df6be849238b57c9bbee2e104b : SUCCESS in 1h 57m 21s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b8bcf1ba9b405b8ffec0e8f3194d45 : SUCCESS in 2h 03m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/492a65c2962d4fe4a63e20b727fe46e7 : SUCCESS in 2h 26m 24s","accounts_in_message":[],"_revision_number":5},{"id":"bffb39e25cd6026cbf76b894eacd2f9f995d21cb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-26 18:24:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b88ee47e8dcc5c16d3f495879ccbd03559b3b51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 20:58:02.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bd4490e791a4f358b8c585edaa4e9f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a70aaba84df4126aa4cb758d2406b5c : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/987d0aa685644350a549e6b9c03abf54 : SUCCESS in 3m 53s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b509c6b3fb804254af67ed9b7042a373 : SUCCESS in 15m 23s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/47b024a6465f4726af51930adf23122a : SUCCESS in 12m 35s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/4a8b8dacb721414188d5b2c4c5d4e7fb : SUCCESS in 12m 24s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28ef96513ed14ee3856c949360a3f3e7 : SUCCESS in 12m 49s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f4769b397c2d4518bf55914643eed99f : SUCCESS in 11m 14s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/fd6c4e7053e04385b302846c0a8b9a17 : SUCCESS in 10m 46s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbac4447f48e4b3cb82c888b46bda33b : SUCCESS in 8m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a1abca2b2fe74fb3a7e50e3b916e9578 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9fa5e17d5adc436bbd3fcbc217635b78 : SUCCESS in 1h 50m 08s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/947d81e0b02a4a00bbe66be217e36e06 : SUCCESS in 1h 06m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/397ccda37ecd43629963360d54e0b38f : SUCCESS in 1h 47m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c85b9aa6a2d44a792eca709b33e90b5 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/050e5c2c1a71423fb430a55421fc48a1 : SUCCESS in 1h 53m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a03fe31d4e084cf2be73f1f0c081451f : SUCCESS in 1h 55m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1693125271b5433cb3759b9ebc3e1af8 : SUCCESS in 1h 54m 29s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/95f456e5e3464ab88f3fc11e9b3bec88 : SUCCESS in 1h 44m 12s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b38336f3302d408084b3c5b5983ef56e : SUCCESS in 1h 51m 02s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e75710a1cb141f199f4afcc63e0e064 : SUCCESS in 1h 39m 15s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a06f7900e174b5284cd150e00a9f57c : FAILURE in 1h 09m 57s","accounts_in_message":[],"_revision_number":5},{"id":"0581bad09758c5042fe2563080e98e7da8dae1a1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-28 08:52:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5d6a79b305e5633b345b9be63f901a127891601e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:19:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e839a58cf4484473b964dcdc4005e3a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade10ae1de06470e8db579e186361972 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df3a421d8ac04be781a67e8e53fca842 : SUCCESS in 2m 41s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e789ab2892a44922b6032dcd1ae6962d : SUCCESS in 8m 41s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/68a6797c62a649c7b0c60cc8aaae3aee : SUCCESS in 7m 08s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/862e14aa23ed4cf6be596da5d74a699e : SUCCESS in 12m 52s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d724713655224c37bf8f24290d300698 : SUCCESS in 6m 30s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/be4780cf34fb482ab0a59776ea4c8a9a : SUCCESS in 11m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c3291add4ef4e05b58a66cd32a54c3a : SUCCESS in 16m 36s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04b6f45d115541b2b92e7917b52f4d54 : SUCCESS in 7m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4bc8f2a6c5c443b59c9279199725d3a8 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5ba259815f5421a80b140129a80029d : SUCCESS in 2h 03m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/becb7180efe54b87b360f03dfb126f55 : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/937250d6cead4350bae3f520c513dade : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1c12b7f0845c40c9aeabeb986df2c8d4 : SUCCESS in 1h 23m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be636a647b9b425bb30bd211b8cf6316 : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68eb9b2535ad4b5586c79236a68ae501 : SUCCESS in 1h 52m 32s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5018c944fdb42ad8d9430d803817430 : SUCCESS in 2h 09m 40s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e88e65edfd34d4fafdf6c0416f32220 : SUCCESS in 1h 48m 57s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/5a01f100d3804014a5cfcf6e4db3dd7d : SUCCESS in 1h 51m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f4af93d125e49d1936534a393acfaa3 : SUCCESS in 1h 03m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de5af93892724498b98c8ab1765aa76b : SUCCESS in 2h 19m 34s","accounts_in_message":[],"_revision_number":5},{"id":"05fea55e19e3fe46e5974100db4c5e19dea4a64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:21:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2605cc731ca544b65d39f942ed49a5d2cc049e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:44:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/faf508ba5fa049cfb94fec28758fa861\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4497bcf6e04c2884c419084e54e468 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15e208016cbc45218236238d7b7500fa : SUCCESS in 2m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e424fb348104fc4870083607f55d1e4 : SUCCESS in 7m 47s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/87158854fbc0408ea528935f3e5fe4a5 : SUCCESS in 11m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/15821121d1214dbf995ad5c8259e5d8f : SUCCESS in 11m 36s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1c39d4696fa4e598709f8d976357f59 : SUCCESS in 6m 28s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/425744dfdc40459cae4cdc17e2028f7b : SUCCESS in 7m 11s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc67a5cf6d2b47dc93569917b4390833 : SUCCESS in 10m 26s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3bd7b1b2c2fe439ca30a39146ab65489 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a93315807004c3ea7c2dec2647260aa : SUCCESS in 1h 48m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/638d847eccb84dd5ada5a1607e08eb6f : SUCCESS in 1h 58m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/80e5b038475344e5b5676be99363bb27 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ee35d46673434283bed9444562d41992 : SUCCESS in 2h 17m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/224b63d4b2774aa4ba7d9c9a4fd06784 : SUCCESS in 2h 19m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/931e32a45c7d4a6f9d4880bc9a8803f6 : SUCCESS in 1h 43m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f63c08c3384a4702b5136a0a5403bb85 : SUCCESS in 1h 58m 42s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/254dc5cc5f04445aac7d30a99cb5b18d : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/198a79c2b7e844f3ac27e4b0d3a66c8f : SUCCESS in 1h 43m 32s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26ec685c676046c0908cc0a85d576a22 : SUCCESS in 1h 41m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/445c2b902c9e46dd9d1d30f8dc90a352 : FAILURE in 1h 52m 00s","accounts_in_message":[],"_revision_number":5},{"id":"2e72a9327e1f43585ddbd021e0e2ea74b8b0a565","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 10:13:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0ff0bdd28828b5adf5a81edaa31d9e175456cd0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 11:29:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e108c7283fb456c84080409d5a95810\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197f2a6e7e6648d59c9c20d2370ee180 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484dd266dc554da8a818e5b84562935e : SUCCESS in 3m 20s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3557ce863fb4509affacd924eec8106 : SUCCESS in 14m 50s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a1e9bbb3269f4e098aa99a446c746758 : SUCCESS in 11m 48s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/2a9784fde1e84a10ab36984477ccef15 : FAILURE in 7m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a6d1e27af7d4eaaa0da0614239549ac : SUCCESS in 12m 46s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/165b68aef9814c81ad0464da46893c93 : SUCCESS in 12m 04s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/b3a0227cace143dfb23892623c415367 : SUCCESS in 10m 11s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49a65874f3b74a1198659dfd4bcdd93a : FAILURE in 5m 10s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a00a1e2ee4104584992d321af42fad4c : FAILURE in 1h 04m 49s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/28ef7ed4dee24eeb955c17efaaac75da : FAILURE in 1h 00m 34s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d16d96551fa240e5b59877558d11eba8 : FAILURE in 1h 03m 43s","accounts_in_message":[],"_revision_number":5},{"id":"c0b2ef640e533c454e4011d53c1fa416539aefa4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 21:34:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"764eefde6f4693e30b9bef5d0d45be3a8a0cf048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:13:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/538a56af33134cf9bbeb5f4c4f420c45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46aea162d28a40559bf6066e68789fbc : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b7a106255554b53a076f7692749b8ce : SUCCESS in 1m 53s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1b955818706420fafc1f835aba7edba : SUCCESS in 7m 52s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5f2f577d624542ea8421cff64cdf9a8e : SUCCESS in 12m 36s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c3e4f5018fc644f18ca4d95bf42f6f75 : SUCCESS in 12m 34s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9470b49f173c4f9fa292d20af1f6275a : SUCCESS in 13m 21s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f06d45936f7c4174b3c96000eb86dc63 : SUCCESS in 13m 01s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/76a43062902b46b6b01a8fed6ee5b356 : SUCCESS in 9m 54s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/07762732a83244a5be35453fd002ccd6 : FAILURE in 5m 35s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a421d0aa329143b68a103bb213a87146 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/42b8e418f10a4921bc8513f9364bf256 : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb2ff224ae38497897aef1bdd873b210 : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":5},{"id":"8b5c3c51d8ccde70233064028126b06e133d8359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:16:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ffc25b9e9d73791f3764e7d9b544a4ac13083ca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:41:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d69c0b39cae94877b5683f4a45ad6e17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82653984f3134b50b164e76b435394f0 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1b34be8f91943648dfc1dd36e395800 : SUCCESS in 2m 44s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6204de4da46540a492b43ed53812d9e7 : SUCCESS in 9m 42s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/36e5e1e4bd8543e99293fb5af1df4ced : SUCCESS in 6m 31s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/49b12ae795d34e1c806c9a6f333ba6e3 : SUCCESS in 13m 21s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d13199a06cc462dbc986ebfaf89347c : SUCCESS in 10m 40s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/eab10ec5fe4540e7bfde4ba9222e7a00 : SUCCESS in 6m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4f9f6a74296d4b8cb330fa468166020a : SUCCESS in 9m 55s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3f0899934a6e407da974de1c2102271f : SUCCESS in 46m 45s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e33e1488d4934c12a22806458c2986e9 : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/144720c728ee4d988a70f23992942eb3 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":5},{"id":"46981e31bd4fb25307388d0a64e79d804676c8fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:41:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"08b5bf80b694746ce864b5bd898e450e717f7eff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 00:43:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dda0b625a0f477081e9548f87338717\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a053344cc8442038e25424859b067b0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c88be164c894b416b08ce9b1198704b31832945b","revisions":{"a4291ac052880d877e232090ce9db3b07b2de461":{"kind":"REWORK","_number":1,"created":"2026-04-13 07:39:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/984312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/12/984312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/1"}}},"commit":{"parents":[{"commit":"00256b6cc14738b236610b1203c259bbf753eead","subject":"Add hashi_vault pki backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00256b6cc14738b236610b1203c259bbf753eead"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I6d70d1d207a077035befdd5cb3ad5ff886ef1a81\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a4291ac052880d877e232090ce9db3b07b2de461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a4291ac052880d877e232090ce9db3b07b2de461"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"123d7040344c94804ea5ca52447abbe709169883":{"kind":"REWORK","_number":2,"created":"2026-04-13 07:45:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/984312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/12/984312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/2"}}},"commit":{"parents":[{"commit":"00256b6cc14738b236610b1203c259bbf753eead","subject":"Add hashi_vault pki backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00256b6cc14738b236610b1203c259bbf753eead"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:45:14.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I6d70d1d207a077035befdd5cb3ad5ff886ef1a81\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/123d7040344c94804ea5ca52447abbe709169883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/123d7040344c94804ea5ca52447abbe709169883"}]},"branch":"refs/heads/master"},"5d3f2d73b2dd6d14355c29a7642a3b8a57848ec1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-13 12:54:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/984312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/12/984312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/3"}}},"commit":{"parents":[{"commit":"00256b6cc14738b236610b1203c259bbf753eead","subject":"Add hashi_vault pki backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00256b6cc14738b236610b1203c259bbf753eead"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 12:54:14.000000000","tz":0},"subject":"Remove galera_disable_privatedevices","message":"Remove galera_disable_privatedevices\n\nWe now supported only RHEL10.\n\nChange-Id: I6d70d1d207a077035befdd5cb3ad5ff886ef1a81\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5d3f2d73b2dd6d14355c29a7642a3b8a57848ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5d3f2d73b2dd6d14355c29a7642a3b8a57848ec1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cae1d06f4daf697a0f57381f4e441aed55686457":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-20 19:26:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/984312/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/12/984312/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/4"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-20 19:26:51.000000000","tz":0},"subject":"Remove galera_disable_privatedevices","message":"Remove galera_disable_privatedevices\n\nWe now supported only RHEL10.\n\nChange-Id: I6d70d1d207a077035befdd5cb3ad5ff886ef1a81\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cae1d06f4daf697a0f57381f4e441aed55686457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cae1d06f4daf697a0f57381f4e441aed55686457"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"5d3f2d73b2dd6d14355c29a7642a3b8a57848ec1","theirs":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","contains_conflicts":false}},"c88be164c894b416b08ce9b1198704b31832945b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-20 22:08:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/12/984312/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/12/984312/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/12/984312/5"}}},"commit":{"parents":[{"commit":"d6c4f4332a96273b5d41d4c030823bcfd343e961","subject":"Upgrade MariaDB to 11.8.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d6c4f4332a96273b5d41d4c030823bcfd343e961"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-13 07:39:44.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-20 22:08:20.000000000","tz":0},"subject":"Remove galera_disable_privatedevices","message":"Remove galera_disable_privatedevices\n\nWe now supported only RHEL10.\n\nChange-Id: I6d70d1d207a077035befdd5cb3ad5ff886ef1a81\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c88be164c894b416b08ce9b1198704b31832945b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c88be164c894b416b08ce9b1198704b31832945b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"cae1d06f4daf697a0f57381f4e441aed55686457","theirs":"d6c4f4332a96273b5d41d4c030823bcfd343e961","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
