)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~986573","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-05-24 12:17:05.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-24 08:04:20.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-24 12:17:05.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-24 12:17:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff","subject":"Upgrade MariaDB to 11.8.7","status":"MERGED","created":"2026-04-28 19:18:01.000000000","updated":"2026-05-24 12:18:00.000000000","submitted":"2026-05-24 12:17:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986573","meta_rev_id":"785e5e36e2ecf046f70e84567ca4da7f16eec566","_number":986573,"virtual_id_number":986573,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-24 12:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-05-22 08:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-05-21 07:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-21 22:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-05-22 08:06:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 23:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-20 15:55:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-05-20 19:27:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-05-20 19:27:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-05-21 22:04:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-05-22 08:06:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8032e913d9ea592fd3aca258ae6b2ca55341d725","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2128488fbd0d85e7b4eb7f0ddcaf68ab89799d03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:38:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c0e86a19add65c5fdd0f36ede0a7862e2ccf07f0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:45:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"ce9681d764a5e1262361d8792151b516e5751696","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:47:44.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"0886f51e4db464bce9fc374f9d7d0f7b1ba44eab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:48:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"28b5d7fedeb6d1ee7e739cec8236be5c6391d439","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 19:51:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"34367cb3987bbfe4ff8be2121605c97ed79e277f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 20:04:06.000000000","message":"Abandoned\n\n..","accounts_in_message":[],"_revision_number":4},{"id":"142b71e895e2b9c8c5294b6018c8a3489e166298","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:12:34.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"86420b880f2cb1384a0c2ff586ea7db13d9a4ae5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:13:08.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b88b2a802b65081c8afa85f0bb62b16eae6d38ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:13:18.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"78080a9f9b666ff8e0b871bccd9e32ef9679f77c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:13:45.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b6d89a2a77bcec188dfa2e64e6430f3ef62845fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:14:14.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"89fe048900e90dabb1d82e17419c524fc0fecc35","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:14:19.000000000","message":"Patch Set 8:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"e5960aae39314c848ba098a1d1d4eab3dcfa6a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 20:14:38.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"5f0c5a4b6c726fadb9f8c030576ff2f30e6e657e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 21:08:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":9},{"id":"300e8d49794ba8547e199474d50b560586dbbbd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 21:20:54.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"5972b522a95a9a4e27ff747dfd763794a88332d1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-18 21:21:02.000000000","message":"Patch Set 10:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"41fc6e3716af7328077a4728e748c900a0a96861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:50:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdfddb8d668743fcb06a7561bcb9c884\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9247a83cb82948438352c4b794d8ce43 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcc14cf5ae7c4ba89b237453490aaae7 : SUCCESS in 2m 55s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d524b03a3814075956aee18a16cad8a : SUCCESS in 8m 59s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/dea5aaa7b5d3422c808911db53aa4058 : SUCCESS in 6m 13s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/81039108dc4446f6abb177a84724d85d : SUCCESS in 7m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cbaf94da0fd4bf09a507a4b15813f4f : SUCCESS in 9m 35s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/31750ed684ec4d20bbf72226fbfa398e : SUCCESS in 7m 39s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/463181df223f46d1992c1f7932e4186c : SUCCESS in 10m 37s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05adaac87bd24150a37fc576666985b7 : SUCCESS in 11m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/28173e5d4e064d378adad41367ef7ec7 : FAILURE in 55m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ffb63a201f740f084d87f6229887d5a : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8414f25f13f5428685ce9427231c7d7d : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/47728dca8fce4c3e900675c47f7ce835 : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/22a3defa81834f98ad05a7de2fb6ff5d : SUCCESS in 1h 39m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f100ba14e2647f3b16889b93c3d642f : SUCCESS in 1h 53m 46s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fff53f0969ec4f709b8ec34d4625f17a : SUCCESS in 55m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a20a5e8ead847d38aeddc0d1e95dfb9 : SUCCESS in 56m 06s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/79662b83dafa45a8a7187062503b554c : FAILURE in 9m 07s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9c1c098aa37b41fd8e98338acc364c08 : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d13af269be374e40981c30397b815902 : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e68b255473a4662b66d7683efa966aa : FAILURE in 9m 11s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f04a7b15e124610a1e7519c3928c134 : SUCCESS in 1h 56m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfb60f35e11040b0b674e9836e6aea44 : SUCCESS in 2h 21m 48s","accounts_in_message":[],"_revision_number":10},{"id":"bbe5ebdb46b8140f640ded83878295eb11e4c628","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-20 15:55:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cf1734ebdb3a9d9484c206e27d081c2ef95c939f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-20 15:57:52.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"bdb936283c37780369f570e3fc4b6523b58b73a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 19:03:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f395eac7b469464398de1095529f39aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17009c2837674589a3795683559295a0 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/237cc572a9524abf90446d66c205da43 : SUCCESS in 2m 55s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f69404cb5f46471993f5ac8e7599cecc : SUCCESS in 16m 41s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b4c459502de64446bd9105e43c54c87c : SUCCESS in 10m 23s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/9af63cb0a6134165ac49bbdf0ded5f80 : SUCCESS in 13m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfeac7e14b85476490f4a3cca3c44bc4 : SUCCESS in 14m 42s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ce573666d4914f4fb54ae8fe9d0f7c77 : SUCCESS in 12m 41s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5fb6c28d36554724bac2b2c3259130c5 : SUCCESS in 10m 38s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/718e58474f9446b2bcb358406a95aa1c : SUCCESS in 8m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1f5448fc0af4036bcf27c0afabfeeae : SUCCESS in 2h 20m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/93d62f3382824bf1a87adeb9632f0c27 : SUCCESS in 2h 01m 19s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6d0c4ebd159d461f823c31467d289cec : SUCCESS in 1h 54m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5942d7f56380465ea7d2ccc93030a72d : SUCCESS in 2h 11m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8674d390506b41239fe8b797ca623b02 : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b4e8fce05ea4ac4a86c2f6bdd240fc6 : SUCCESS in 1h 55m 26s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9cb05ee885348528186c46b8da6293c : SUCCESS in 1h 52m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1154d59dbee54e2281314fe660161543 : SUCCESS in 1h 03m 31s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1b018048f4b4b0ba5c3ad1ebf51acdb : FAILURE in 7m 55s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/073af62d41104d6a96c07ec72af4e6fa : SUCCESS in 1h 01m 15s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b004cacd064e49ee9816ecc87c26fba4 : SUCCESS in 1h 40m 39s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9e40d84328af4f8e9a3b0ed1ab4aa380 : FAILURE in 9m 29s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b874858f33b24fa9b333582afda67b74 : SUCCESS in 1h 47m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f3e5fae63e4474dbff72e5d65590be5 : SUCCESS in 2h 54m 39s","accounts_in_message":[],"_revision_number":11},{"id":"bddcea16d1c3324d535757e72ad8c3ebb7ab3287","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-21 07:53:51.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"2aabc82aa36e9603cf04425fe1a6a66e3545249a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-21 22:04:55.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8659136d462dd1703567bf2080d93d2dbe7a1820","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-05-22 08:06:14.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"703821451cc0af3cca10a39484775d351ba8a786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:08:58.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"024d302517a4bf30970c53ea0dd7a3e3ea8f9ba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 10:41:54.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/673fabc0d6734d64809df1e282b41562\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe8725d9e6442de83d75541202aaf92 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/877541f8e0074bbf8c0fd75485941488 : SUCCESS in 3m 51s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fac8367f3b144db5bb1344d41e67833c : SUCCESS in 17m 46s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/46dbacfca83a4652b7120588e580275c : SUCCESS in 11m 12s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/da9abb5c05464777a9600e50375d8616 : SUCCESS in 12m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e2785ccd854498a43fde575a71b2cc : SUCCESS in 15m 18s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/bf837e2535a1494c83333f8f1be05ded : SUCCESS in 10m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ed81209fba2c415196b0980e2700d64e : SUCCESS in 10m 18s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/095b031f39b54d31bc6e8e82f9b847f4 : FAILURE in 1h 58m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/59fa4d24bbf143f69fb8994d571a3299 : SUCCESS in 1h 36m 20s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/39782c630236456885c2f35e26d4a16f : SUCCESS in 2h 00m 28s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2da45363646d44859392182583c43bfc : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05b631c7fd474c88806bd34f45c822ae : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32e75076d6fa4933a0c815cc8b5126f6 : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3e46390a6f64c42a30f978ebe2913a8 : SUCCESS in 2h 10m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc9100043cb94d0ba86f2d96739dcf03 : SUCCESS in 2h 10m 40s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/987edfe5f07e47e2b51e40923fd83b8a : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/68cfd8ebf8eb41a7bd7b9e35c3247ef4 : SUCCESS in 1h 37m 56s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/377b041e9f6b4d6984a6708f927428c7 : SUCCESS in 2h 01m 22s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17cd299aabb242938c14a5c298b3884d : SUCCESS in 2h 28m 22s","accounts_in_message":[],"_revision_number":11},{"id":"8260ee7b5449c8330598838d3fe642e38540cfce","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-22 11:06:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"061365bdc8bd9980f867b276b70642caf00f32ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:33:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83c4d40b71c542599e6066590281c731\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f92007723b4743a409ebfeb8b0f640 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32cc9ef34ccf4d8297ce46fe0f813d26 : SUCCESS in 3m 23s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0ccb168a30244a782bd2ee5f56d04e4 : SUCCESS in 7m 13s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/d6c73287a49c4bc3a07d253e204c8a99 : SUCCESS in 6m 13s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/27e44d83dcfe47ff8ea5adc66bca26f5 : SUCCESS in 11m 19s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9a36a575fc14b2fbee6b2889c1990a8 : SUCCESS in 12m 25s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4e736f4c88774783837ba77c13c9bfa1 : SUCCESS in 13m 13s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/69bff04a5ab7400b8287695caa07a88d : SUCCESS in 12m 17s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0e69f06dbb7248a89313404af0f1270a : FAILURE in 4m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/30f7f162625c4a26bda3a816748700c4 : FAILURE in 2h 10m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3af215b505094e6fbe28ed7a6c830d0b : FAILURE in 1h 09m 40s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8c6405f46a03418bbc90fb24c379a603 : SUCCESS in 1h 45m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/39748249fbdd42c2a090897b818b789b : FAILURE in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f7eb275330ad4d4188926cb2fcf5a8a1 : FAILURE in 1h 21m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5e163a85fb4448a9f8e19b800b8bfa7 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91a0d2d28d1c44ff8e5fabdc8489fa6c : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a92da6b0cee24dc0bb3556a8c36ca6af : SUCCESS in 2h 01m 17s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d8fa71b927a04ad18007cd9ef17c85fe : FAILURE in 9m 50s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a2393813abea411aa788097d3b1b1fc1 : SUCCESS in 1h 46m 28s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ffdc9a9031be4832acadfb16c7f43f61 : FAILURE in 1h 47m 47s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6874a905fc0f499b8cdb5a4fd84bf91d : FAILURE in 10m 47s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43a719eb2581434c906d1a6531326988 : SUCCESS in 1h 41m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98b4a7b407314b449ee1ee7928c3d303 : SUCCESS in 2h 19m 03s","accounts_in_message":[],"_revision_number":11},{"id":"c0b6179aecbed0e0e9e48018dcd86e362e0384e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-24 08:04:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4a62b5ef870b22519264df399862187afee048e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 09:44:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb206efbee846e5a5782c1ac7d275de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f6eb8ee8c6476983e12718fcfa9b50 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3876724077a474ab4f6826330ba662f : SUCCESS in 2m 20s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7785f1c6af0d4967a81680b4749d5c54 : SUCCESS in 8m 14s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/08b1457d16fa46c287e761c28bf28870 : SUCCESS in 9m 03s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3301d5c3d5874f0889af710c95147b3d : SUCCESS in 6m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf29d4f34b3f48509d371fab9ea46359 : SUCCESS in 6m 31s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3553d5f7ec144340910c4d8b0270920b : SUCCESS in 6m 38s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4b191070f9ec428da22151c0967e653f : SUCCESS in 10m 49s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/282cbb26478d406b9c2e41501dd20005 : SUCCESS in 6m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a45ef182e4da4a99ae73892a416637ea : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1424074ebcdf40dcb1b49d9c383dd024 : SUCCESS in 1h 00m 56s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9b54fe202eed43c98981abd73ca64ccd : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/364eb55eba294269a22fc2b8d3549c7e : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8a0204f524646f698f7f64ca73c8db9 : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed4962600abe4977a3035698b88f26f6 : SUCCESS in 58m 41s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc310e176a349ea83c4f0c995b96a91 : SUCCESS in 1h 00m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be6d5e6189a04a408ab366ded2f34e6b : SUCCESS in 1h 00m 01s\n- openstack-ansible-deploy-infra_distro_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f03b450ba6094c3985263183ff1c7e96 : FAILURE in 7m 59s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2da954cfef054cc9a9a7a33ca73f7844 : SUCCESS in 1h 21m 50s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2e52781b87334b68a70a177d3c51d6ba : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-infra_distro_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73a1ea8802cf463081deaed14d3892a7 : FAILURE in 9m 11s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22b4c78bd1b145078f9078473b5e1095 : SUCCESS in 1h 00m 40s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f1d9824ab5140b5a7edcfe0b642046e : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":11},{"id":"810252cc345d3e4cd82916b24efb5fccf02357ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 09:47:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"20b3bad250c3f08c5ecb7751156771725711c066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 12:17:05.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1982732a6c3c4ac4a7457b0dbb98e2dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e557e65ef63b452c9ded2e4f72cb6514 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62b84c07e8834e37b86f7ccf4613d9bf : SUCCESS in 1m 51s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1256a0df422945b5849405e64111a115 : SUCCESS in 8m 16s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/76c9b2130c494b0da4aed35eaf6e4cfe : SUCCESS in 9m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c4589f782f4c470b932d507bde48249c : SUCCESS in 11m 15s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81dbe509f4654a7aac09684b2d9e98cc : SUCCESS in 11m 57s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f463e0c7790f42b9b579a7c21d4d2117 : SUCCESS in 11m 14s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d332a3722b884dfdbbf354bffca8b8f3 : SUCCESS in 7m 17s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/845e3782e9c74702a595b24e90f14ade : SUCCESS in 1h 10m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5440ba5593524bebb185346449fc9c79 : SUCCESS in 58m 38s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ef9224e9a5574f7ca7d41b53fdb19147 : SUCCESS in 1h 52m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c79a220b19a84e56971a029af6e4b897 : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/51ee167d8d9d44a1b12c6bf6d7d0c762 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eafed22d464749f79e4fec3f2473667b : SUCCESS in 1h 53m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e70077d0ee9c411cbe2c5a8261b477fb : SUCCESS in 1h 48m 36s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb2033b833fe44b4af85784676f1c839 : SUCCESS in 1h 53m 46s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/23b91130bd704cabbeda37a3a1b36808 : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3ce3f6cd505440448663468325e4ba35 : SUCCESS in 1h 46m 55s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a1aad07c6234b4898f63488fdd89e29 : SUCCESS in 1h 45m 53s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d2948725acf4881b004ad711bf7aa2f : SUCCESS in 2h 24m 27s","accounts_in_message":[],"_revision_number":11},{"id":"6aee0e9d1777fb25597f5fb385169bae6088b856","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 12:17:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"785e5e36e2ecf046f70e84567ca4da7f16eec566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 12:18:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d90b04a82254f1faa89cbf70c726f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abae8c1bbb44446a21ff47b6595c07b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d6c4f4332a96273b5d41d4c030823bcfd343e961","revisions":{"3142d40bd22a89035647a269f664c0e6cb551ff4":{"kind":"REWORK","_number":1,"created":"2026-04-28 19:18:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/1"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3142d40bd22a89035647a269f664c0e6cb551ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3142d40bd22a89035647a269f664c0e6cb551ff4"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"6bf527da39bc57031635e0c67b06c94b3bf16ba6":{"kind":"REWORK","_number":2,"created":"2026-04-28 19:38:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/2"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:38:36.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6bf527da39bc57031635e0c67b06c94b3bf16ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6bf527da39bc57031635e0c67b06c94b3bf16ba6"}]},"branch":"refs/heads/master"},"c1617829fbe013e71e4df07e862da22376d4c772":{"kind":"REWORK","_number":3,"created":"2026-04-28 19:48:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/3"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:47:59.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c1617829fbe013e71e4df07e862da22376d4c772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c1617829fbe013e71e4df07e862da22376d4c772"}]},"branch":"refs/heads/master"},"4974762996825376e4b561ca4824d9c7d5813b3e":{"kind":"REWORK","_number":4,"created":"2026-04-28 19:51:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/4"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:51:48.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4974762996825376e4b561ca4824d9c7d5813b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4974762996825376e4b561ca4824d9c7d5813b3e"}]},"branch":"refs/heads/master"},"7b0694a5a5adf7d58f742ff48d882fee7759c4ac":{"kind":"REWORK","_number":5,"created":"2026-05-18 20:13:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/5"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 20:13:05.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b0694a5a5adf7d58f742ff48d882fee7759c4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b0694a5a5adf7d58f742ff48d882fee7759c4ac"}]},"branch":"refs/heads/master"},"0a9c0fee46d29e7506cee76420aea8382d233414":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-05-18 20:13:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/6"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 20:13:18.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0a9c0fee46d29e7506cee76420aea8382d233414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0a9c0fee46d29e7506cee76420aea8382d233414"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cc2ff2fa01476df35725cb3225cca51b15c2adf3":{"kind":"REWORK","_number":7,"created":"2026-05-18 20:13:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/7"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 20:13:45.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc2ff2fa01476df35725cb3225cca51b15c2adf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc2ff2fa01476df35725cb3225cca51b15c2adf3"}]},"branch":"refs/heads/master"},"67e7adb7f1cd967c4821ed9286efc5152df23285":{"kind":"REWORK","_number":8,"created":"2026-05-18 20:14:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/8"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 20:14:13.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/67e7adb7f1cd967c4821ed9286efc5152df23285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/67e7adb7f1cd967c4821ed9286efc5152df23285"}]},"branch":"refs/heads/master"},"6db71589ea8b825b7b794cf7c578570ce449e756":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-05-18 20:14:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/9"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 20:14:38.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nhttps://mariadb.com/docs/release-notes/community-server/11.8/11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6db71589ea8b825b7b794cf7c578570ce449e756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6db71589ea8b825b7b794cf7c578570ce449e756"}]},"branch":"refs/heads/master","description":"Edit commit message"},"76ae0301c8193d93d01885bd646eabd187c4c9c4":{"kind":"REWORK","_number":10,"created":"2026-05-18 21:20:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/986573/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/10"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-18 21:20:44.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nhttps://mariadb.com/docs/release-notes/community-server/11.8/11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/76ae0301c8193d93d01885bd646eabd187c4c9c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/76ae0301c8193d93d01885bd646eabd187c4c9c4"}]},"branch":"refs/heads/master"},"d6c4f4332a96273b5d41d4c030823bcfd343e961":{"kind":"REWORK","_number":11,"created":"2026-05-20 15:57:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/73/986573/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/73/986573/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/73/986573/11"}}},"commit":{"parents":[{"commit":"8ce0f88e394ab93a0c0b0885f470f3447d3f275d","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8ce0f88e394ab93a0c0b0885f470f3447d3f275d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-28 19:18:01.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-05-20 15:57:48.000000000","tz":0},"subject":"Upgrade MariaDB to 11.8.7","message":"Upgrade MariaDB to 11.8.7\n\nhttps://mariadb.com/docs/release-notes/community-server/11.8/11.8.7\n\nChange-Id: Ic6955dcd5f83e8b64114f3292028b3b8799ef3ff\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d6c4f4332a96273b5d41d4c030823bcfd343e961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d6c4f4332a96273b5d41d4c030823bcfd343e961"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
