)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~990472","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ie0432080357861427941ccf2d9952387ac5309dc","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/core-2.19","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-05-31 23:31:14.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-30 18:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ie0432080357861427941ccf2d9952387ac5309dc","subject":"Ensure galera_server_bootstrap_node is deifned","status":"MERGED","created":"2026-05-28 13:49:54.000000000","updated":"2026-05-31 23:32:03.000000000","submitted":"2026-05-31 23:31:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990472-osa/core-2.19","meta_rev_id":"3518cb2a64efbe9320aac7b95c46880c48a19293","_number":990472,"virtual_id_number":990472,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-31 23:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-05-31 19:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-31 19:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-05-31 19:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 16:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-29 07:15:04.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-05-29 15:02:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d9537463c19c730c9ccab8909ff20618e79709cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-28 13:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"126e494e840a654479a304098ed9f40a7f29485b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 16:25:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6735775b94e45dea2fb09af87076f79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0efa6e6c31c644eea9a7fc3b6c7d827e : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc9777d11ff24f7fa013cb38d3501d01 : SUCCESS in 1m 53s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8ffad2f11e1462486d9ca7b3447382f : SUCCESS in 9m 49s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/9d05878136864440833a83826150bbad : SUCCESS in 12m 24s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f0bfd9272abb4b6485a0c0e7c6c7f959 : SUCCESS in 6m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4d9583120e142d39c2fd8a755529432 : SUCCESS in 13m 59s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c1a7a7d32e984cc1932f6d88fb051284 : SUCCESS in 7m 11s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/33d5c36969354e5286c79f8ba67290a5 : SUCCESS in 10m 39s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6d898e2a41104f47a6e6044b947c0b5c : SUCCESS in 6m 15s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/3e0a688b7fbc411cb0b02f614fe9ead5 : FAILURE in 2h 12m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/bb7ec0eefdfa480fb8bdeb895d347f44 : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0a20b5e93b334facb36e5ea973bea76e : SUCCESS in 1h 50m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/09a8e4e992bd4dbdaf98e706f3cd2f57 : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b86984cfb5be4c83be02f2a39e0e2063 : FAILURE in 1h 42m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72dd57b964474f888880d0f29f01ccec : SUCCESS in 1h 55m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a16b33b8a204137906981e139e7d121 : SUCCESS in 1h 53m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a6929c5138243e0b5fc3d27c3e7d958 : SUCCESS in 2h 03m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c7b728d1e97a4bdca2e0c9deac275ca5 : FAILURE in 1h 29m 48s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/748d463828814473b003577a4a208c23 : SUCCESS in 1h 39m 19s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b2ee75aa3bd4be58844617424fd6d07 : SUCCESS in 1h 45m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/878317152ae24d26aeae178ff4427c06 : SUCCESS in 2h 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9cc05671096926bd268d5492d3c2d20e9eaac791","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 07:13:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c97ad5800f3c6d017b31ab7e46585796b3ec825e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 09:32:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa29546a7bff4ef9bad603ac086f7a10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3a07f0362c4230b49d2442711302dc : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7ea67e76d2b441da47687c0ad17670f : SUCCESS in 3m 02s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0198240217c846b98129f238ca7b26e5 : SUCCESS in 12m 45s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0efe397e57cd49a2a630ee8d6fc0c1f4 : SUCCESS in 10m 30s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/be6f95e23c7742ca81b56cc00d5c8f9e : SUCCESS in 10m 39s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94fd157bcb1b4215a9a6f229f25a049d : SUCCESS in 7m 28s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1fe277d510804ed5981c4c689406ba7a : SUCCESS in 11m 59s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e009c4e6d29140238a9775f253d4dd42 : SUCCESS in 10m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e6cb4d5f127d4fdfb96a9f83af0af458 : SUCCESS in 29m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e709d803fd1e4429a5065c1f5b7b288c : FAILURE in 1h 52m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/835971af98d04d1c802a97b241bf256c : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f59b02face9c414984714f709b191f25 : SUCCESS in 1h 55m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/702e6c182f0d434db614ac8fb890e6d6 : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/187e60aad3e747f686a53e284ab29fcb : SUCCESS in 1h 55m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fed4fd418bc4a1aa4e4b561c291e987 : SUCCESS in 1h 52m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc4dd3d6e7794759bc2096216302a6fc : SUCCESS in 1h 58m 10s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef7df091ee343afbd4e030c02ac3c66 : SUCCESS in 1h 54m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2d4c0e6d58f74f2583c759a81a3e7450 : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1978d4a76d474c01b5d0948cbc41b99a : SUCCESS in 1h 50m 23s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/880ec7aac2964abdb7f1082086874521 : SUCCESS in 1h 45m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb40a7450ea74d7cbfa952e7ba6671fd : SUCCESS in 2h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dac689327dd3f410460fa13bf48e335ec5330799","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 09:54:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e9bfc34f79b8f866f5b2f092cf03fcefe55ffa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 12:39:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d755624532c4111949e14dc0c84c934\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ac2e095a7044a697e7ce71da417a91 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bf61660975c4c7cbdf97447414b4182 : SUCCESS in 3m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/601212af58954c8abbbfe3f163a83f7e : SUCCESS in 11m 46s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/83bdf158dcf04d2e8a23778f9cbdc6a2 : SUCCESS in 11m 38s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3ca2740c38c145e8878cfc3a100442c9 : SUCCESS in 11m 38s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac30d56dca34f96914212c8b6b7cb4f : SUCCESS in 11m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/480a3fc9b6df4541b68c9a80f42346be : SUCCESS in 14m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/dfd12d4e80994632b34df8678b702eb2 : SUCCESS in 10m 42s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d27ac8c2b5f4b99baf5310179bdb5ba : SUCCESS in 6m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6fd95a58d63a4a448136c8d2d3768d53 : FAILURE in 2h 28m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9cec7f0460eb48ebb3b1b7f9e3772c83 : SUCCESS in 1h 56m 55s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a0f906be727246e19428c37a84764141 : SUCCESS in 1h 57m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/39bf96de54b140f19c474836433c4a50 : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/99f7236603804d70980f13954b557dbc : SUCCESS in 2h 14m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/431cde602a2e404ea96436d443b73ce9 : SUCCESS in 1h 51m 49s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc58a75a899646728d55ad90de9a4a21 : SUCCESS in 1h 02m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22760be147b342469a60ebff4e40f66d : SUCCESS in 1h 57m 39s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b4f9bbc0f7514fa7a22d3f261a248d25 : SUCCESS in 1h 48m 27s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d85d99570c66429a9e11d2d2aa5bfec7 : SUCCESS in 1h 45m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91a26a6ecc754683962727c1640d1865 : SUCCESS in 1h 34m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f68949db64044ee79af510f4f0bbdbe5 : SUCCESS in 2h 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8ccb8a716b311f3b87b5b650957ff34fc956cede","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 13:23:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6192a5a9653202fb0188eace9a3ab6ecaf4f4b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 13:27:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"65e274e62b2eb9b92477b6690a00f342545276f5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-29 15:02:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1384ee7145d39cb492380a47e4fd5fe97a1466e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:51:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f598e694254f8e89570eace739a91a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff036ee988645c5ab2681ca02ec9d53 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf01fb366364430bbc637b11156c775 : SUCCESS in 3m 38s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f665eb91be44e5bac6eb30a76907f8b : SUCCESS in 11m 14s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d77b3a444054aecb423c9d82f95d1fd : SUCCESS in 14m 33s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/8ceaec6fa3fd48998b72c6691f569dc0 : SUCCESS in 11m 56s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d5cbb5071a44dea8d864df0aca93ff7 : SUCCESS in 13m 50s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f8d8a2ace71b41659d4f76675625381b : SUCCESS in 14m 10s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/194f3193b5fd442b9e447dc5c47436c0 : SUCCESS in 11m 15s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f4fea893b1d42969d70102ee6cbd23c : FAILURE in 5m 22s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e9287fe0747b486b8ab1bb63cb61a51b : SUCCESS in 1h 58m 43s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a5cb2623626c416481d3233cc2174718 : SUCCESS in 2h 13m 59s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d66d6714dbb4a02a8b32db2921139f6 : SUCCESS in 1h 41m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aa9757bf9ff4f029374431cae48772a : SUCCESS in 2h 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f4deee2b13e9209498f63ea86cc6014ed3a28ca6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-05-29 16:11:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8206aaae782894855c3bdfa9ed2b83949ba5a5fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:13:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"97f9d82c187178ec3c63034c8cb4b821cc5bad02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 17:35:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d97552b9cb342c5f9d12c5ce9810dcfd01ecbbbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 18:15:38.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/990645 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"90b40ab88e1fc29197bcbc31b56d2a9b7861900b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 19:39:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48570c2c77f342848095eb2a067b6c4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7047d9de26404f6ab1ae1bc32072d0c1 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc4a885aa3744d3b80824aab85a9f31c : SUCCESS in 3m 09s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3dbdb4d2c5041e0bbf6b45902520693 : SUCCESS in 8m 53s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c790603692f648d8b98be6bd9f2b709a : SUCCESS in 12m 47s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c5d5266d1d3b4d4f83b289f2db409ca5 : SUCCESS in 12m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/976d4a36413f4abb85cd2779e43f9b45 : SUCCESS in 7m 57s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f9f19b6f1f174fca8e6ffedb3396a4d4 : SUCCESS in 13m 46s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/55353176d38e4adbb5febc7785553fe4 : SUCCESS in 10m 51s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/037041ed8b624aacbb0f56d9fd8813c3 : SUCCESS in 11m 32s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f732aa451bff47b6a57fa31626ac1cdd : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/e9c519f2743544d7a6cd52fc3fa48a25 : SUCCESS in 1h 39m 42s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a1368d86a3645dabf6f49431bac64fa : SUCCESS in 1h 37m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8bf519b5d544a0ab1f84642bc2fea41 : SUCCESS in 2h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cbe3100d37f43d3f4aedf5e50489a22de02f01d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 22:41:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16ea3c45583a92f674ef38426ec4b29e257b0836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 00:31:50.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/990645 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"705f78abc3be04df590d7ef4c795f4c22b9d89cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-30 18:48:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"277a95522d05d903966deb045a3c13963dabbc2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 20:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad8759a74164f8fb780a0152222799a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04160d45c83e436e843a0fe76b88b2da : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6bfc8e7814b4d69ac22144e1a8bac80 : SUCCESS in 1m 57s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97869b82103942518f5e9d0c84093b97 : SUCCESS in 6m 51s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/eae2ac20184749a8a63cca306c780e8f : SUCCESS in 7m 11s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/bcfc4ba3f44341cca154e8e4c4d81d3d : SUCCESS in 11m 20s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81b7c951f34f4e8082768ada41c31a04 : SUCCESS in 6m 48s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7a35ec5e563f4409bc25c7c04ae19aaa : SUCCESS in 10m 37s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d24e34470c44d6d9b2fd91661437777 : SUCCESS in 10m 06s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8ef3bdc29e7a4ca1913a11da842c1e66 : SUCCESS in 6m 20s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/62b6ebc3f3414fdba863d4f13bfb5a21 : SUCCESS in 1h 46m 28s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/bca8876e0e314dbeb3fddbefa5f30280 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9f0feecf63e4e37bcac4b89d36d2cf0 : SUCCESS in 1h 42m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09f3e7fd7bab46fa9df3484f9238a2e5 : SUCCESS in 1h 41m 54s","accounts_in_message":[],"_revision_number":2},{"id":"03d1512974965249be9332496ea894cb40df8f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-31 19:37:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c8a38861279753dca9c44c8ae4c3727a1cd9d3a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-31 19:38:05.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"085883530d674f7b3b824a3ce0650097b6723fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 21:42:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6532a99a5d44df8d7233f5469bffb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27a861dd65f44479790a7394af2d5b8 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f64dc29fa337481d9591365ef4705d4f : SUCCESS in 4m 07s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/500f16faf4cc40778fad8db5a78fa5b1 : SUCCESS in 11m 51s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b19c89a528d74bd18319ed12d23281ff : SUCCESS in 13m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c58eb5799de24822ad0ce7bd2f2db294 : SUCCESS in 13m 37s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d86f319a3a441ca6577a4874d61558 : SUCCESS in 14m 11s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a4154b9e41d040e0bb3cf6b2aac81803 : SUCCESS in 14m 30s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a379335293e14a1a9267f362cf7bfd8b : SUCCESS in 9m 19s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a26a490d5555481d8d2bdbf03af74753 : FAILURE in 5m 19s (non-voting)\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/304cad2ae0a5431089881ce2e91c13e9 : SUCCESS in 1h 45m 07s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/4518f3b3c2514394a114bd5515408a03 : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/021eb8a7e72642eabebd2728018437b4 : SUCCESS in 1h 46m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4e191cb8cc2a16caf05c00d427c187c0addc0501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 21:45:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"491c2d52378135752d7b0b5547076f96b6a631f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 23:31:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/090337bab8a041ee82a9c64eb84a8249\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fe5fa8206144fb904a8e05c37b037e : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a7e182cb874a73a7694e1b6c6a4eef : SUCCESS in 2m 10s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ec804781f5435db298754b608f1696 : SUCCESS in 8m 08s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e528d359be7494fb1217956b2695feb : SUCCESS in 13m 34s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a1ba75deb3004390a5fee9a45162efa2 : SUCCESS in 6m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d1b6e375a8c49b0a925eaf076d6937d : SUCCESS in 7m 09s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6ef420149a7247c58c6822af3f2a2b2e : SUCCESS in 6m 57s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/04a22578283c4416884afb7661aad12b : SUCCESS in 10m 30s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a058dcd9e6ca4070a9e404a38ea85c5f : SUCCESS in 1h 41m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/66b58d810cc84c9e96e17467c5c807bc : SUCCESS in 1h 02m 37s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecca6320d2704896999875fb355535b2 : SUCCESS in 1h 44m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6843ddcb7c0bd4ab7e9af885b4287f9fd1e0a76e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 23:31:14.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"3518cb2a64efbe9320aac7b95c46880c48a19293","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 23:32:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3de67d791e604afd86a45282d179fed3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c042f5d9ed49ef9493db12605cb107 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"28ac0fbcd95088b06ca98a2360a1aacd093211e6","revisions":{"5e99eaea036a2cf8afd9dfcab5359305a7d3352f":{"kind":"REWORK","_number":1,"created":"2026-05-28 13:49:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/990472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/990472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/1"}}},"commit":{"parents":[{"commit":"d6c4f4332a96273b5d41d4c030823bcfd343e961","subject":"Upgrade MariaDB to 11.8.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d6c4f4332a96273b5d41d4c030823bcfd343e961"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-28 13:49:45.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-28 13:49:45.000000000","tz":120},"subject":"Ensure galera_server_bootstrap_node is deifned","message":"Ensure galera_server_bootstrap_node is deifned\n\nAs code is being shared between client and server,\nrunning code against client will trigger handlers evaluation for server.\n\nDespite conditions on handler execution will not pass, their evaluation\nresults in a role failure, as delegate_to is now evaluated before\nconditions.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: Ie0432080357861427941ccf2d9952387ac5309dc\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e99eaea036a2cf8afd9dfcab5359305a7d3352f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e99eaea036a2cf8afd9dfcab5359305a7d3352f"}]},"branch":"refs/heads/master"},"adca7e9f5f809004d893dfec1ee6b18eadf82d0c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-29 13:27:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/990472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/990472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/2"}}},"commit":{"parents":[{"commit":"d5c5f584f1e505027c029cb448ab359ec2ff622b","subject":"Disable CI jobs for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d5c5f584f1e505027c029cb448ab359ec2ff622b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-28 13:49:45.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-29 13:27:53.000000000","tz":0},"subject":"Ensure galera_server_bootstrap_node is deifned","message":"Ensure galera_server_bootstrap_node is deifned\n\nAs code is being shared between client and server,\nrunning code against client will trigger handlers evaluation for server.\n\nDespite conditions on handler execution will not pass, their evaluation\nresults in a role failure, as delegate_to is now evaluated before\nconditions.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: Ie0432080357861427941ccf2d9952387ac5309dc\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/adca7e9f5f809004d893dfec1ee6b18eadf82d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/adca7e9f5f809004d893dfec1ee6b18eadf82d0c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"5e99eaea036a2cf8afd9dfcab5359305a7d3352f","theirs":"d5c5f584f1e505027c029cb448ab359ec2ff622b","contains_conflicts":false}},"28ac0fbcd95088b06ca98a2360a1aacd093211e6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-31 19:37:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/990472/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/990472/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/990472/3"}}},"commit":{"parents":[{"commit":"2d8237fcc78cd5f1e5141ed8617cced4354e7eea","subject":"Disable CI jobs for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2d8237fcc78cd5f1e5141ed8617cced4354e7eea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-28 13:49:45.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-31 19:37:58.000000000","tz":0},"subject":"Ensure galera_server_bootstrap_node is deifned","message":"Ensure galera_server_bootstrap_node is deifned\n\nAs code is being shared between client and server,\nrunning code against client will trigger handlers evaluation for server.\n\nDespite conditions on handler execution will not pass, their evaluation\nresults in a role failure, as delegate_to is now evaluated before\nconditions.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: Ie0432080357861427941ccf2d9952387ac5309dc\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac0fbcd95088b06ca98a2360a1aacd093211e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac0fbcd95088b06ca98a2360a1aacd093211e6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"adca7e9f5f809004d893dfec1ee6b18eadf82d0c","theirs":"2d8237fcc78cd5f1e5141ed8617cced4354e7eea","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
