)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~990646","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I8fa1fd69c34d0f676fa916957a41ae27b9a634f0","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/core-2.19","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-06-03 09:20:22.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-03 11:32:45.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-03 11:32:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fa1fd69c34d0f676fa916957a41ae27b9a634f0","subject":"Revert \"Disable CI jobs for the role\"","status":"MERGED","created":"2026-05-29 13:28:56.000000000","updated":"2026-06-03 11:34:58.000000000","submitted":"2026-06-03 11:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990646-osa/core-2.19","meta_rev_id":"03c49320172d60200cdf3cd6badafa7c528ba816","_number":990646,"virtual_id_number":990646,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 11:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-06-03 09:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2026-06-01 21:32:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 21:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-06-03 09:20:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 15:02:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-05-29 15:34:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-29 16:11:14.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-05-31 12:24:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-01 21:48:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"0ed08e301035b2ffbe65c05c6424e70cc89b5ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-29 13:28:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10649daa461ae78f878bad4754d31c2a7d1c3527","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-29 15:02:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a384a64a60b18d1ed96b28b700f9fcc5816b4e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:34:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9a290614fc4979aaf4e8e8712a4abf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76fe55a04cb04828b7f232efa02731b2 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/382a3d7002cc4335b348ad8488b9427a : SUCCESS in 4m 18s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12b73892fd5c40afab13c21463aedc15 : SUCCESS in 10m 00s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/ccced3777a23448688ce53f6cce28d62 : SUCCESS in 12m 21s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0ee9b2a821d749c1870c768735ebd725 : SUCCESS in 11m 37s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7abc3305dbd14795a239f594950f5ebb : SUCCESS in 12m 27s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c90606e287c64aaab98ffef3c09cdb4b : SUCCESS in 12m 15s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/dd2ed9b1b9d743ba9599c479ec8a555d : SUCCESS in 10m 40s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/14c4c9fcead9436ebd43886ebbe102aa : FAILURE in 5m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f0e8e0b74cbc4ad8b53725cff021aa87 : SUCCESS in 1h 39m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/b8f778af2a3047909c57155851dd2fe8 : SUCCESS in 49m 35s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f4587fa8489d4a01a3285ce382bb2e1d : SUCCESS in 52m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/06db5c2c7b774e09b058e89afbb2238e : SUCCESS in 1h 00m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d746b4ddf3bc486e825bccf5e0c3266b : SUCCESS in 1h 52m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c562cf69856b48619d6223869d6998ef : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/869061f444b3435cae1ca67b6189011e : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73b5aa43b76e42cca806355310f68c64 : SUCCESS in 1h 33m 29s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/08f7b798c8024fababd35dd7f45eedb8 : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/7b6f3982ea0b48ed808d6cc82ad28fe0 : SUCCESS in 48m 22s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0639c3ee23794b7ead4b45613b419850 : SUCCESS in 48m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e2e89acf4214c7aae5dbda66507d602 : SUCCESS in 1h 56m 11s","accounts_in_message":[],"_revision_number":1},{"id":"881e08c25385dc8a7abea54dec35faa867856b7f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-05-29 16:11:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"34e9b9081407508dce124dda95bfa876a9944435","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 12:24:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b197682b85def802792a7f379bf48d18602e9f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-31 19:38:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cb94fb0492e5003d03ec12c36e13ff4cf9054a1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-31 19:38:28.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7cac70916f4d668caf3879b999ac32305894df02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 21:51:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c981a86b41fa4d908a277b43838fbd01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7273e4a2e674fb7bc3215dd46e12653 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e10babce430d4693af66449c2b2a8873 : SUCCESS in 3m 35s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34a78c2e41784fa79b59f67c055b7a5b : SUCCESS in 10m 10s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f7258e92f2346e4bc79dafc0dc93f47 : SUCCESS in 12m 16s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0268908ed2c04b0fa87960b366c041d5 : SUCCESS in 12m 20s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86356581b5b446fea78ffc806ed9d584 : SUCCESS in 12m 55s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7101825f718840c9be3448f135ff150c : SUCCESS in 12m 28s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/56fc731514ad4f5aba36a6ab1ba270a5 : SUCCESS in 9m 26s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2156a906db214f1abc92b3d677f80c6a : FAILURE in 5m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0db8fcf4760042328d9569134e96cf7e : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/891f322d9e56481898d4b59714a4e8e9 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9bbba9b9f5614a4f84412e7ea28339e5 : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e19c2a9e97134560b7597ff8bc908bac : FAILURE in 1h 41m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53d202707862411faee8082c69aa0fb8 : SUCCESS in 1h 44m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02897a7989d14596a73bb67ae9ba72d4 : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cda6d01facca43d19582d0f14b81deb8 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/308c4e804c684a9290c13799f68f8eea : SUCCESS in 1h 35m 51s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/84e7a7b32b914002b1fccf2ba90680be : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/61a1221a3e0b4dfc96f2539cb4c37c82 : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca91c3b1fec2495a82391b2d13f99de8 : SUCCESS in 1h 16m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f7e7c071af44e68f9b5049092c9dcc : SUCCESS in 1h 57m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5305e274a04985c095b992dbc8efcc47a382c451","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 06:52:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7c7cbb9c1000485acfa3e130ce508bc051b2f8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 08:56:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26940216aae541cc8d3f5b472d34c90d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe041c97455d403fbe219607362bdaa9 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d2927e812f34de5b9846f7ebec77cb8 : SUCCESS in 4m 23s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12bb2c2e14314173bbcd11130da1ab6d : SUCCESS in 9m 36s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/313b96e278494116be864dec024780a9 : SUCCESS in 10m 07s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/15ea8af1f6a2485d99f6d7c8e7257586 : SUCCESS in 13m 05s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e4b2b381b04427e8842aafd80ed0772 : SUCCESS in 7m 53s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e8181eebb1eb49a4ac1c95089fd893f9 : SUCCESS in 11m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/6938666b32bf4f478eeb2d5c30a1b25b : SUCCESS in 11m 12s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e01ae673cc5441e886cb45448561ec43 : FAILURE in 5m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9c262407b63b4f63a0c5e9b962c962d3 : SUCCESS in 1h 36m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ca70fb9d0bf4c02ab8df32543008013 : SUCCESS in 48m 20s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c6c0e0afebec47c783f4ade1bacc751c : SUCCESS in 52m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7714161eebfc49bd909e924779adb4e8 : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b980713754f44a084d0e685a58b755b : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab9028481a6647ccb9474e5265112d83 : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e66cafb3f9b2441997f76bfb83020a0b : SUCCESS in 1h 35m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0830441dfba548308c2259d985e304da : SUCCESS in 1h 28m 23s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/701fde6b75224a32a1057bdef769eb0b : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/040caf84a97c41258ab592f594c9365e : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/388b1ce32e074fba8532294a5b1afa53 : SUCCESS in 1h 24m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/043b10f15ba44859a146f4108cccd436 : SUCCESS in 1h 55m 03s","accounts_in_message":[],"_revision_number":3},{"id":"fc27b4b98a79b28262b25aaefca4bfab352ee673","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 21:32:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cfcd220d81a928257f7fba83c216c0a0fc86dc8e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-01 21:48:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2107650d4f73af89410402ad0ff10ddab860b56","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-06-03 09:20:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"579547fe7c7364aec9123fad10f111db538c727c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 09:25:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"809f72217218ec3a8a46b53c130ff9c82e30a280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:32:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8431452c74f34236bb6e11f7511ff96c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c31ed2114d34af9b97d2325ceebf363 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d812c95edffc43a59547970fa52005db : SUCCESS in 4m 40s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a9535ee7ef40cf8d17cd44a7a04d11 : SUCCESS in 13m 50s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2668f93036584b5fbe325312c308a611 : SUCCESS in 11m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/7686936b0d094a6fa32b249f7a950f58 : SUCCESS in 11m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/021339e6ee7c4612bb7dc0112a5a2483 : SUCCESS in 12m 47s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6ef0b4e72b444dbebb048a03d3e31c38 : SUCCESS in 11m 52s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2ba3c2639d1d4bffba7b3ae0e1151866 : SUCCESS in 11m 26s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/790902ca02a542f5b446fc975e40cfa3 : SUCCESS in 1h 48m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6299c894a0ce4e9a86f75c6ef4bfa4f3 : SUCCESS in 1h 04m 38s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/446ea850c7a84cd6846ea45730042bcf : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/40bc4d17119c4df5a8fd17792c7b338b : SUCCESS in 1h 52m 33s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbc64dc000724de897596d52b459eb03 : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40b5af8bfbf14be49274e0011bffcffe : SUCCESS in 1h 04m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c620a1aa5ee47b7af385a0e2a3af039 : SUCCESS in 1h 49m 31s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a594fbe91e5447e2b4fb9b5a2df98571 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-infra_distro_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9e91dd3c6cd04ccc8323fed77a59999f : SUCCESS in 1h 36m 05s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c141850b507443d6963edb26655c0aca : SUCCESS in 1h 24m 50s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eec1a5fa2285429999e21a6472e9dc19 : SUCCESS in 1h 36m 47s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b31e8bdcdc4d5aab1c5f535c4941bf : SUCCESS in 2h 06m 02s","accounts_in_message":[],"_revision_number":3},{"id":"3032efbd59dc6014674ec0e0af29efb4941fb619","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:32:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"03c49320172d60200cdf3cd6badafa7c528ba816","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:34:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7332033d0f5b4560a88cffec4a1b9441\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899d159b22d24b858b1d08ef9b834686 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d55c48d3dd0d791b1cda9e2065b51f1d3e1fe06","revisions":{"5cf3423b114ff73cc1860b0a5fc80935e243f58a":{"kind":"REWORK","_number":1,"created":"2026-05-29 13:28:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/990646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/46/990646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/1"}}},"commit":{"parents":[{"commit":"adca7e9f5f809004d893dfec1ee6b18eadf82d0c","subject":"Ensure galera_server_bootstrap_node is deifned","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/adca7e9f5f809004d893dfec1ee6b18eadf82d0c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-29 13:28:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-29 13:28:49.000000000","tz":120},"subject":"Revert \"Disable CI jobs for the role\"","message":"Revert \"Disable CI jobs for the role\"\n\nThis reverts commit d5c5f584f1e505027c029cb448ab359ec2ff622b.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: I8fa1fd69c34d0f676fa916957a41ae27b9a634f0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5cf3423b114ff73cc1860b0a5fc80935e243f58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5cf3423b114ff73cc1860b0a5fc80935e243f58a"}]},"branch":"refs/heads/master"},"f9ad1536a4dbe29b0bf491ecbe3359dfd710e54b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-31 19:38:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/990646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/46/990646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/2"}}},"commit":{"parents":[{"commit":"28ac0fbcd95088b06ca98a2360a1aacd093211e6","subject":"Ensure galera_server_bootstrap_node is deifned","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac0fbcd95088b06ca98a2360a1aacd093211e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-29 13:28:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-31 19:38:14.000000000","tz":0},"subject":"Revert \"Disable CI jobs for the role\"","message":"Revert \"Disable CI jobs for the role\"\n\nThis reverts commit d5c5f584f1e505027c029cb448ab359ec2ff622b.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: I8fa1fd69c34d0f676fa916957a41ae27b9a634f0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f9ad1536a4dbe29b0bf491ecbe3359dfd710e54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f9ad1536a4dbe29b0bf491ecbe3359dfd710e54b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"5cf3423b114ff73cc1860b0a5fc80935e243f58a","theirs":"28ac0fbcd95088b06ca98a2360a1aacd093211e6","contains_conflicts":false}},"4d55c48d3dd0d791b1cda9e2065b51f1d3e1fe06":{"kind":"REWORK","_number":3,"created":"2026-05-31 19:38:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/990646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/46/990646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/46/990646/3"}}},"commit":{"parents":[{"commit":"28ac0fbcd95088b06ca98a2360a1aacd093211e6","subject":"Ensure galera_server_bootstrap_node is deifned","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac0fbcd95088b06ca98a2360a1aacd093211e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-29 13:28:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-31 19:38:28.000000000","tz":0},"subject":"Revert \"Disable CI jobs for the role\"","message":"Revert \"Disable CI jobs for the role\"\n\nThis reverts commit d5c5f584f1e505027c029cb448ab359ec2ff622b.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/990456\nChange-Id: I8fa1fd69c34d0f676fa916957a41ae27b9a634f0\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4d55c48d3dd0d791b1cda9e2065b51f1d3e1fe06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4d55c48d3dd0d791b1cda9e2065b51f1d3e1fe06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
