)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~998695","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24","subject":"wip","status":"ABANDONED","created":"2026-07-25 07:24:13.000000000","updated":"2026-09-02 11:35:28.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"10d3fa5fb6a4beb6f70be248dc7c8677820bbd87","_number":998695,"virtual_id_number":998695,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-27 17:34:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-07-25 12:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68095a69b45aca9455c21c30da157d94c5cfbedc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 07:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ed91248506f13ce0f324ac72a6c53bbd50ebf79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 07:37:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"26b58e9153db3a8de7b31dc182cbd6aa48d7cf1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 07:38:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94d60a1373582df4bfe6b644607507d7b81eee97","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 08:38:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"666fca3006dfdcced42244c0ec5011435b35277f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 12:03:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd8a5fb89abf4c26bcd22135ea119b1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aadb8d02d6014e73ae3de3c3a435b79f : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fc2e16974c24ba9bb6c60f12b9fedab : SUCCESS in 3m 07s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73342698c2744d4c8262bfccbdee5cce : SUCCESS in 14m 59s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f0db50c61aed4e338b6f5b777316d43d : FAILURE in 6m 15s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee141e52f04e4c838b9a1a3fd5a5d6d2 : FAILURE in 8m 13s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e9b3f88b97a54058be7c047b942c6249 : FAILURE in 5m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1531a87899154a309eacfe832da44f6f : FAILURE in 6m 07s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f6a66c8995a496ea668f85c992d8278 : FAILURE in 4m 15s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/beab841534624785bfcdfff62e25d547 : TIMED_OUT in 3h 16m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/469b42718f264880aa1065f16b63d2c6 : FAILURE in 28m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b883d7d423e546cd834eedbc54a33f3e : SUCCESS in 1h 36m 42s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fea4eeae040d472788a06109086fad56 : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5edc5c5ed13349d4a320f488cb40c9fe : FAILURE in 33m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f9119bef03cd46b99e227c8fd7f61f5d : FAILURE in 33m 12s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/800cc06bad9644b0acb7af1b06842c21 : FAILURE in 42m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b06b17e3c4c24686a512479b996581f7 : FAILURE in 35m 46s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/89194296e43646099ce3facefc0d45d1 : SUCCESS in 1h 12m 06s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44b682a7f68b402dbf174af356d74a58 : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/27137144ba8542c0b5d265f0aa419874 : SUCCESS in 1h 31m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a29e35f92094d17900e56dacb18fa2b : SUCCESS in 2h 24m 58s","accounts_in_message":[],"_revision_number":4},{"id":"fad15c5048dbf3848a1621fd517257cf489b48a2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-28 19:06:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c782547337a559f93b17b7d74c75234b7e715a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 22:30:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50555fe1848c42959181d60e1edb3b87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fb3798cc7c44588565f39cbcc60cbe : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51b4014acf7c4f5b85856e3619ff9063 : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d1a64b179049f68bf68117353eab7e : SUCCESS in 8m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/87a4f41e6f30438c9757f97582929462 : FAILURE in 8m 02s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca62f49f7af34aa5be64fd90e96534e1 : FAILURE in 4m 10s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/11768d608eb8434ea03160d5f033b395 : FAILURE in 3m 39s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/31b54bf54d4a452691745fec1552e12e : FAILURE in 6m 51s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e029ff8b703c489db18205c381a31fad : FAILURE in 5m 11s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc3e354d9d744024883f85338a797e0b : TIMED_OUT in 3h 14m 06s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b9ea2fe4e80e4723ac3b48f1bdcefceb : FAILURE in 40m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/61ee2066a47844f59c30be285a866d8c : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e03245506a84436b4ce68401a11a232 : SUCCESS in 1h 01m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0078372889c400db3d203923dd9be98 : FAILURE in 24m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/63bb302ef36d49e49bcb7dff0044067b : FAILURE in 40m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fa5fe715cef490fae90e7f9f5e9fbed : FAILURE in 40m 20s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f57902942ff442993f22a4808c7fc7a : FAILURE in 41m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/06873d46c6124e9da03a93fd57a12653 : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d658d6d16e84f5e85d6f00bd168ebf0 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c1f2c7c35ce24013960469a03b5d5b81 : SUCCESS in 1h 21m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ede8700c0434fc49e1d4778d34539e3 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f89d033676d1747672e3f4db6270e22c570829ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 06:41:09.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7f33f8f24a594c8804818d450e758109d44e0f4f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 06:41:21.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9a4b4066370e29ddc29d54ed144699585fec747b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 09:55:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/964461c3d2974d91af7ab087d71ca1d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552eda9314d74187b456c0a1003be6fe : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b259a54a634763acaac118f952e104 : SUCCESS in 3m 02s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/427acec33a0f4d76a9e83ddaa556b0a2 : SUCCESS in 8m 17s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f59e9d95ca9546619557f7bf14f00144 : FAILURE in 6m 58s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004fd8f871e147698373b3874b9563bd : FAILURE in 7m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e688674bb4744aed8fc57def3788f40f : FAILURE in 7m 21s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4339d2de8b43489d9034b840f5ca8e1b : FAILURE in 6m 30s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3ff55efc3404380bd18f447f9130b1e : FAILURE in 4m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c2e4921330f44631bafab0dac0fe836a : SUCCESS in 3h 06m 07s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9c24f2b4e179445db20bbbf7f4baf632 : FAILURE in 1h 05m 12s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8a3e5dc3a2284d58be2c0c1a0359b9c4 : SUCCESS in 1h 48m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b2e2feb20c44c5e9a97637516cc4d55 : SUCCESS in 1h 00m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d848e2b0dd3a4584ba99d638ac581f54 : SUCCESS in 1h 38m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f2666a0a2c414a5e8da06c33cf01eeb5 : SUCCESS in 1h 36m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1795016bfc71438faafcb2df9e45c49a : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d4863094dcc437f9fbea933971afe90 : SUCCESS in 1h 47m 35s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/95338188a34545c3903f50e900f6ab3e : TIMED_OUT in 3h 05m 09s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ae3d373339f46dea1b42c2116b0c76a : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f03b78e1a0b645fd94be29a12e7ab5b8 : SUCCESS in 1h 27m 02s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e21ebe4cb94745968022645ed618cf14 : SUCCESS in 2h 10m 48s","accounts_in_message":[],"_revision_number":6},{"id":"46b7467b25fc870b41269b8a69d0f0364fe187f6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-20 15:59:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3307ad9ec26aee0e16b3b7bfe36b64da51e22f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 18:35:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6871a9230af4749b0999c36aa569f2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89128eefe7d043c589156b401cc9e8d1 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c53da82b6d3a43908dabec67ea0e5a90 : SUCCESS in 2m 18s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6d8b283a9114fab9a215b5cd7d643fb : SUCCESS in 8m 30s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b585e5988a7143b0940db59b14b45448 : FAILURE in 4m 37s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac7c5657f67444a5893cb9b99d10130a : FAILURE in 8m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6ab06c2a9e084a83a5c5f4c0f09ccd27 : FAILURE in 7m 24s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c697c4feeca446e6a1f57640f1b18f53 : FAILURE in 6m 43s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a6c3a0fed6444e28973975947961ad9 : FAILURE in 4m 56s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d3f1876b87d4467b9e8f3bc601c8717d : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/1db8ef2e6f6e4d8a87110c5975bb1772 : SUCCESS in 1h 01m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e89b48f81be4877b91a5d3663f91046 : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/764048e3c7a34969a05e8100522eadcd : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e4107e3237942d59ebb864ed43f6ea7 : SUCCESS in 1h 40m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ea26375b68d24410997e501d7b50da54 : SUCCESS in 58m 54s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/942bac6aff554544bde3a28ea3e8575f : SUCCESS in 54m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a030b6f98034088a32f17a78950696f : SUCCESS in 57m 49s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/820a645f38554ce7b604295402561ee7 : SUCCESS in 1h 31m 27s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f980f8afd1d84c91be3e1d2d9ffd3b97 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7e8fc8c29c2c4ac295bcb499f584f509 : SUCCESS in 1h 33m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39925410eaf34b58bfeba7923affb7bc : SUCCESS in 2h 21m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e8ded8b84e318a062d7103713dd910b597178f12","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-27 14:54:10.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2742c213d9687f063f5936405b3808746ed37b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 17:34:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37743d2c51ae4f6c8ff7dac3aef0a366\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e8976ca61a444e8837bcf674f8e5dc : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6cb14c3a74488288caa955c2d64eaf : SUCCESS in 3m 34s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0cbf861470542c5bdd2e90000269fae : SUCCESS in 14m 48s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/cb5e47b7d7674935b465a007c4e1f071 : FAILURE in 8m 22s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c445037c2654fa4bf5e3312e285199b : FAILURE in 5m 34s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/056897796e094df7955fd3cc86fda3d3 : FAILURE in 7m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/616cad71232348239764e5ecd26c9aad : FAILURE in 11m 11s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/94a9739bac444989a3739829b15a1a90 : FAILURE in 5m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a8c1759e5ca94475842f8f8bfce4b965 : SUCCESS in 1h 48m 21s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d22de9aa74a74d8c8805ede1142faa76 : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43e49ad5cdb0478ea300b48851910ee1 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2f253bd670645a0b2acfe27d9d1518d : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ef542b49f674533b24eeb6da7e1e542 : SUCCESS in 1h 30m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/860c36aea62544ed93dc4cf81a301e5c : SUCCESS in 55m 12s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc29ec99385b4f8eb8f75f2ff15d2955 : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d6b1b1c7aaf44bfb9c07f599e906805 : SUCCESS in 1h 00m 02s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6fb40623ab5b4d569774bb02d16f7589 : SUCCESS in 1h 29m 10s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b024b3b41074422a81ead69b7b12159 : SUCCESS in 2h 17m 22s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/dbc61845c7a4438b8ddbbb85b3614fc0 : SUCCESS in 1h 30m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72b27bc5277148eaa89ab7a05b7229e3 : SUCCESS in 2h 17m 01s","accounts_in_message":[],"_revision_number":7},{"id":"10d3fa5fb6a4beb6f70be248dc7c8677820bbd87","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-02 11:35:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a1bda4767bd9d3117fc7268db21f700dec09555e","revisions":{"df6e0e67ae6e1099fedbaeffdfba794535974f7b":{"kind":"REWORK","_number":1,"created":"2026-07-25 07:24:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/1"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/df6e0e67ae6e1099fedbaeffdfba794535974f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/df6e0e67ae6e1099fedbaeffdfba794535974f7b"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"53bd100962e99554b253741ab295cff54c9ae16b":{"kind":"REWORK","_number":2,"created":"2026-07-25 07:37:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/2"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:37:36.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/53bd100962e99554b253741ab295cff54c9ae16b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/53bd100962e99554b253741ab295cff54c9ae16b"}]},"branch":"refs/heads/master"},"473ecd20c369042e8bd8a3933909c3f64efe5ae7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-07-25 07:38:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/3"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:38:17.000000000","tz":0},"subject":"wip","message":"wip\n\nno-handler: Tasks that run when changed should likely be handlers. (warning)\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_install_apt.yml:81:9 Task/Handler: Update apt repositories when config is changed\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_backups.yml:91:3 Task/Handler: Grant access to the database for the backup service\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_setup.yml:35:3 Task/Handler: Create galera users\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/473ecd20c369042e8bd8a3933909c3f64efe5ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/473ecd20c369042e8bd8a3933909c3f64efe5ae7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"84ebe1c7aacda56f7e98fc5e5c3ba882f546e770":{"kind":"REWORK","_number":4,"created":"2026-07-25 08:38:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/4"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 08:38:53.000000000","tz":0},"subject":"wip","message":"wip\n\nno-handler: Tasks that run when changed should likely be handlers. (warning)\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_install_apt.yml:81:9 Task/Handler: Update apt repositories when config is changed\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_backups.yml:91:3 Task/Handler: Grant access to the database for the backup service\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_setup.yml:35:3 Task/Handler: Create galera users\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/84ebe1c7aacda56f7e98fc5e5c3ba882f546e770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/84ebe1c7aacda56f7e98fc5e5c3ba882f546e770"}]},"branch":"refs/heads/master"},"2826781cb31c1c301c76c2edc26f9b7a7e8dc8da":{"kind":"REWORK","_number":5,"created":"2026-07-29 06:41:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/5"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 06:41:08.000000000","tz":0},"subject":"wip","message":"wip\n\nno-handler: Tasks that run when changed should likely be handlers. (warning)\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_install_apt.yml:81:9 Task/Handler: Update apt repositories when config is changed\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_backups.yml:91:3 Task/Handler: Grant access to the database for the backup service\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_setup.yml:35:3 Task/Handler: Create galera users\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2826781cb31c1c301c76c2edc26f9b7a7e8dc8da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2826781cb31c1c301c76c2edc26f9b7a7e8dc8da"}]},"branch":"refs/heads/master"},"e36200bc534556dc71c51c9aae537bf74dee08ca":{"kind":"REWORK","_number":6,"created":"2026-07-29 06:41:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/6"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 06:41:20.000000000","tz":0},"subject":"wip","message":"wip\n\nno-handler: Tasks that run when changed should likely be handlers. (warning)\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_install_apt.yml:81:9 Task/Handler: Update apt repositories when config is changed\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_backups.yml:91:3 Task/Handler: Grant access to the database for the backup service\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_setup.yml:35:3 Task/Handler: Create galera users\n\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e36200bc534556dc71c51c9aae537bf74dee08ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e36200bc534556dc71c51c9aae537bf74dee08ca"}]},"branch":"refs/heads/master"},"a1bda4767bd9d3117fc7268db21f700dec09555e":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-08-27 14:54:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/998695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/95/998695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/95/998695/7"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:24:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-27 14:54:10.000000000","tz":0},"subject":"wip","message":"wip\n\nno-handler: Tasks that run when changed should likely be handlers. (warning)\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_install_apt.yml:81:9 Task/Handler: Update apt repositories when config is changed\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_backups.yml:91:3 Task/Handler: Grant access to the database for the backup service\n\nfqcn[canonical]: You should use canonical module name `ansible.mysql.mysql_user` instead of `community.mysql.mysql_user`.\n/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/tasks/galera_server_setup.yml:35:3 Task/Handler: Create galera users\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/995339\nChange-Id: Iea0c6986f212f8ad95ba01ac1bd6280d40cc4a24\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a1bda4767bd9d3117fc7268db21f700dec09555e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a1bda4767bd9d3117fc7268db21f700dec09555e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
