)]}'
{"id":"openstack%2Fopenstack-ansible-haproxy_server~1006904","triplet_id":"openstack%2Fopenstack-ansible-haproxy_server~master~I1ff9f038e4411811df1d1d890f302299d8c02744","project":"openstack/openstack-ansible-haproxy_server","branch":"master","attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-23 08:29:53.000000000","reason":"Reviewer was added"}},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 09:29:55.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-24 01:35:20.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I1ff9f038e4411811df1d1d890f302299d8c02744","subject":"Revert \"Remove support option httpchk\"","status":"NEW","created":"2026-09-23 08:29:53.000000000","updated":"2026-09-24 11:09:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"revert_of":981846,"meta_rev_id":"d9df51fef1d9725e2be2ecffe77a80338e242150","_number":1006904,"virtual_id_number":1006904,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 11:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-24 08:35:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":1,"date":"2026-09-24 09:29:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-24 08:35:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 08:29:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-09-23 08:29:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-23 08:29:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df74a31a286629442c0bcba683b5a54320ade991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:29:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d85dfbb93d12821baa3d4189709d4b7e238c7569","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:30:43.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cddaf006fb8bb0da0f5ca55fc0ce949271a92272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 11:12:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc170470173c4b5093ce8606003b19a8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421b1e69e80a4e4fa0a8cdff9e90c8ae : SUCCESS in 12m 04s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/6c749f0ba63f47a6a0227be8494d128e : SUCCESS in 10m 56s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bf4e6789994b35aa7a279ea42013fb : SUCCESS in 13m 40s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/35566ccae6e54074b47c80344450a5cf : SUCCESS in 6m 11s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ae018fc237314913aca13a1e12a978d0 : FAILURE in 4m 26s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e97264b7bfc4c69b3fae2716329ec07 : SUCCESS in 7m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/756bb53f458146c89319d85ec0326204 : SUCCESS in 1h 47m 24s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/87869274a4cc4b69af31db649b95e85f : SUCCESS in 1h 41m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fd19567474a4972b7d5072d80afae9d : SUCCESS in 1h 38m 39s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2f66546d242481fa1e13ef147b634f4 : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0600defa0064ca0bfb071be9a3551c5 : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4eb9ebd367404c048d0ebca6c87e923f : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d1f918f32494568bce004fa7b57eaa5 : SUCCESS in 1h 32m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f2f14975e440f589606a4acb58ec03 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/032bc045049d46e7a74e664d31f1da80 : FAILURE in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7aa2cf1602bf43d1bf12fd44eb138205 : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/856e4b37248a4ae6940a7568be3fa455 : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e2e235772184c60a5f901ed6e9ee9f2 : SUCCESS in 1h 42m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8505d1e27e474dd0b3d6866ef510fcae : SUCCESS in 57m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f032097bba40482d9d24ec150fddaf4e : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e06c6237c5164f48acb103fa1d748b21 : SUCCESS in 1h 34m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cacbcb1b33ce4b69b8176baafacde9a1 : SUCCESS in 2h 24m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8b8e7f43d89483c9301ff4f3a87d5fd : SUCCESS in 2h 00m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a94e4a6129c74eccbd06fc4d7dd99f6e : SUCCESS in 1h 39m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91d54bb353e14ebbab19581b898be545 : FAILURE in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56ecac3a1be45caba22db1a70a1538e : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdd5dbdafcbf496bba081ba9ac8484ae : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"842495e655f2ca95be5b70e02102f3381e60d1f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 18:13:19.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"03b1a015c83a7ce5ab0299557e72e26b3098418d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 18:13:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6a615da2a32f93f341d63f6a08c1bec9928367f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 20:55:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2d0bc19267040c3b7fd885ecfa17c47\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a8edd4376b464195f19b9b7a4021db : SUCCESS in 14m 41s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b01772ecddc7447b81d5c96acd8b6143 : SUCCESS in 6m 59s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c3cae7eef264bbea1ae20c46144dcea : SUCCESS in 14m 02s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4355964a8d3148849c56ea80cb75c593 : SUCCESS in 10m 34s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/de7bea7ecc694f2982a7cfce9ddd5a34 : FAILURE in 6m 14s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fba714812b74ba295aa4e49503a0eac : FAILURE in 4m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c24b40bfc131422da2a21f38eff5eb89 : SUCCESS in 1h 42m 44s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/03500505920f4cf385f1ea6fd451dd43 : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3b3a76ae61d4be69f1fd75aef1747da : SUCCESS in 1h 33m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7128bb5726654c9d83f49961d56b5a98 : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f28f4f79d5cf48b2aeaafd0e5a261c73 : SUCCESS in 1h 33m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/32c4c4f038b842cea84000fa25481a93 : SUCCESS in 52m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/239940c63310414f8b5601ef0c9e4262 : SUCCESS in 1h 48m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8bff83fdddb467fafe8bfbecdcedd43 : SUCCESS in 1h 35m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/966370b4113c476b972d6161c5e16044 : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8c122cbd29e445039ecac43c6ec3d600 : SUCCESS in 45m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/57e2cb0f945c43109dd3b8b5604b8bdc : SUCCESS in 52m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e58d1a02959472899536ce407e255b3 : SUCCESS in 1h 02m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f189038da5474b2a84efbba909e09e49 : SUCCESS in 1h 25m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5448b1fdd55444ba89793f12dcdd0cab : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd9be7c32d04b57ba3d7228adabe72d : SUCCESS in 1h 33m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74733ca198da4ae59827fb1c11f7b828 : SUCCESS in 2h 27m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73bc929466ac4ae7a5d637a170c52e6e : SUCCESS in 2h 23m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e30a0510fd19495db4b94e3256bc7005 : SUCCESS in 1h 27m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c3856ae3d5b42b68341813e21a78e85 : SUCCESS in 2h 23m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41569355ba24a9fbc4cfaedaa788332 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eefb4ad22c144178a16b03effd64ed1 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":3},{"id":"74ae8d3895e77bdaa936a850f59fd094b2f572b2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-24 01:35:20.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd652383be6b423e2769e4f00b035325d4858b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 08:35:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f02f9015132424931f1f5b8aef02db8e3191bf20","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 09:29:55.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9df51fef1d9725e2be2ecffe77a80338e242150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 11:09:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b164891d224cd0bcf16afe56a1896d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acf468325f454d72814b45244f4b3b59 : SUCCESS in 16m 05s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/6b7a98273b4b44b0b881116f0dc2e244 : SUCCESS in 10m 04s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86dc634d0c284daea969f3e88050cc09 : SUCCESS in 10m 20s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/faf863e4311844339742d00d6888834e : SUCCESS in 11m 09s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/57dc2830a73b4323a1b4557771d1bb58 : SUCCESS in 10m 35s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/722c08d79d2e43d481b40a0396ccafc9 : SUCCESS in 9m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a3223876d164460a875ad19127aef5f9 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ea75b3c5abe841b691b24b3b500bdb91 : SUCCESS in 1h 45m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ccfdad98eba24153bfdbe9414ba133d8 : SUCCESS in 1h 45m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fd326b58a86d42049e36481226a760da : SUCCESS in 1h 42m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac67e0bc53a40b7995437c5c643fcf9 : SUCCESS in 1h 48m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7a258991fca14816833af85c6d791f0e : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a289bcb22c84fb9a23981d9af1ef99c : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3a27e8df2f743de844cdb6ed38ad344 : SUCCESS in 1h 49m 32s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3cfd1231db93400194a840a23e9b3028 : SUCCESS in 52m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/8d319cabe7b24f119731af1155259530 : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ec96bf0756f14c7ca1316da1052bc649 : SUCCESS in 51m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f1680c35b41d4d3d8e7c91220686ffc3 : SUCCESS in 51m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/722b8469cc014e809857c18258d317ec : SUCCESS in 1h 35m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e9aaa770c2ae466990750412ce57e345 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/991e6cba88c3423d86b2956236132543 : SUCCESS in 1h 31m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e121b116b3244a2e9957de12e3901aa3 : SUCCESS in 1h 24m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56d3e22f5827440897a05553e739a11b : SUCCESS in 2h 07m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8883526a750043b1b3251e88b4208d8c : SUCCESS in 2h 27m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc290d9a0514449ba7624b37b95a800c : SUCCESS in 1h 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3dbda91f1da4d8d8d9f54fdddd8aec4 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/980088adab1d4ead825fd590b4da550c : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b10b06a6d8b46d9b7ef6920078fab44cda7e6784","revisions":{"ee78bb663cf1f1dc7fce72ec7ddababd15ef4569":{"kind":"REWORK","_number":1,"created":"2026-09-23 08:29:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/1006904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-haproxy_server","ref":"refs/changes/04/1006904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/1"}}},"commit":{"parents":[{"commit":"89909c5609073451b0d4b4a6810d26e4f9b49856","subject":"Remove support option httpchk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/89909c5609073451b0d4b4a6810d26e4f9b49856"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 08:29:53.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2026-09-23 08:29:53.000000000","tz":0},"subject":"Revert \"Remove support option httpchk\"","message":"Revert \"Remove support option httpchk\"\n\nThis reverts commit 89909c5609073451b0d4b4a6810d26e4f9b49856.\n\nReason for revert: Option httpchk is supported and required [1] for\nhttp-check to work. Removing it results in a regression of checks.\n\n[1] https://docs.haproxy.org/3.4/configuration.html#4.2-option%20httpchk\n\nChange-Id: I1ff9f038e4411811df1d1d890f302299d8c02744\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/ee78bb663cf1f1dc7fce72ec7ddababd15ef4569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/ee78bb663cf1f1dc7fce72ec7ddababd15ef4569"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"4acd5d603e939a536e8b9174bb86c4dc01d8ce5e":{"kind":"REWORK","_number":2,"created":"2026-09-23 18:13:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/1006904/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-haproxy_server","ref":"refs/changes/04/1006904/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/2"}}},"commit":{"parents":[{"commit":"89909c5609073451b0d4b4a6810d26e4f9b49856","subject":"Remove support option httpchk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/89909c5609073451b0d4b4a6810d26e4f9b49856"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 08:29:53.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 18:13:11.000000000","tz":120},"subject":"Revert \"Remove support option httpchk\"","message":"Revert \"Remove support option httpchk\"\n\nThis reverts commit 89909c5609073451b0d4b4a6810d26e4f9b49856.\n\nReason for revert: Option httpchk is supported and required [1] for\nhttp-check to work. Removing it results in a regression of checks.\n\n[1] https://docs.haproxy.org/3.4/configuration.html#4.2-option%20httpchk\n\nChange-Id: I1ff9f038e4411811df1d1d890f302299d8c02744\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/4acd5d603e939a536e8b9174bb86c4dc01d8ce5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/4acd5d603e939a536e8b9174bb86c4dc01d8ce5e"}]},"branch":"refs/heads/master"},"63beaed1441a09980136dfa917b1e9ce12457798":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-23 18:13:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/1006904/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-haproxy_server","ref":"refs/changes/04/1006904/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/3"}}},"commit":{"parents":[{"commit":"1aa960f906c812492e39ad9d43b0c20f46cbcf8c","subject":"Merge \"Ensure GPG key is distributed before importing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/1aa960f906c812492e39ad9d43b0c20f46cbcf8c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 08:29:53.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 18:13:29.000000000","tz":0},"subject":"Revert \"Remove support option httpchk\"","message":"Revert \"Remove support option httpchk\"\n\nThis reverts commit 89909c5609073451b0d4b4a6810d26e4f9b49856.\n\nReason for revert: Option httpchk is supported and required [1] for\nhttp-check to work. Removing it results in a regression of checks.\n\n[1] https://docs.haproxy.org/3.4/configuration.html#4.2-option%20httpchk\n\nChange-Id: I1ff9f038e4411811df1d1d890f302299d8c02744\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/63beaed1441a09980136dfa917b1e9ce12457798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/63beaed1441a09980136dfa917b1e9ce12457798"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"89909c5609073451b0d4b4a6810d26e4f9b49856","ours":"4acd5d603e939a536e8b9174bb86c4dc01d8ce5e","theirs":"1aa960f906c812492e39ad9d43b0c20f46cbcf8c","merge_strategy":"recursive","contains_conflicts":false}},"b10b06a6d8b46d9b7ef6920078fab44cda7e6784":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-24 08:35:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/1006904/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-haproxy_server","ref":"refs/changes/04/1006904/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/04/1006904/4"}}},"commit":{"parents":[{"commit":"ef73c658cdd073a74ba1449b9911a98b454fe7a2","subject":"Address ansible-galaxy fetching issues.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/ef73c658cdd073a74ba1449b9911a98b454fe7a2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-09-23 08:29:53.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-24 08:35:10.000000000","tz":0},"subject":"Revert \"Remove support option httpchk\"","message":"Revert \"Remove support option httpchk\"\n\nThis reverts commit 89909c5609073451b0d4b4a6810d26e4f9b49856.\n\nReason for revert: Option httpchk is supported and required [1] for\nhttp-check to work. Removing it results in a regression of checks.\n\n[1] https://docs.haproxy.org/3.4/configuration.html#4.2-option%20httpchk\n\nChange-Id: I1ff9f038e4411811df1d1d890f302299d8c02744\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/b10b06a6d8b46d9b7ef6920078fab44cda7e6784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/b10b06a6d8b46d9b7ef6920078fab44cda7e6784"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1aa960f906c812492e39ad9d43b0c20f46cbcf8c","ours":"63beaed1441a09980136dfa917b1e9ce12457798","theirs":"ef73c658cdd073a74ba1449b9911a98b454fe7a2","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
