)]}'
{"id":"openstack%2Fopenstack-ansible-haproxy_server~1006907","triplet_id":"openstack%2Fopenstack-ansible-haproxy_server~stable%2F2026.1~I60e2f9a544120f9a1646776f870715607c32388e","project":"openstack/openstack-ansible-haproxy_server","branch":"stable/2026.1","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-23 12:23:50.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I60e2f9a544120f9a1646776f870715607c32388e","subject":"Ensure GPG key is distributed before importing","status":"NEW","created":"2026-09-23 08:53:55.000000000","updated":"2026-09-23 12:23:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":1005798,"cherry_pick_of_patch_set":1,"meta_rev_id":"ed970948b2cfd49f5ce62d9a98774ecb03a5cec4","_number":1006907,"virtual_id_number":1006907,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-23 12:23:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-23 08:54:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 12:23:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfa16775bc140ffa8a8b247a878fe5a543099f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:53:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"965676fda47be0f08e43d06f8a6d07d9d30dd745","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:54:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed970948b2cfd49f5ce62d9a98774ecb03a5cec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:23:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c3cc36ec51248bd8bcd8c70231c9ffa\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67eeef4edcee44bbb4e14f48094270e3 : SUCCESS in 16m 54s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/c124892bbc2c4e4fb0ab97923521dd42 : SUCCESS in 11m 22s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/f5a8a3fb5951495482014a99097b42d8 : SUCCESS in 11m 19s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f36cada43a94b3380637c935374c506 : SUCCESS in 12m 33s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/81923e0154f24e34a8b1a7b36ce8edf6 : SUCCESS in 12m 19s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/309660d40eba43808b4720b0884bddbb : SUCCESS in 10m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2a4108b171694c37ba5e742ed9dd50ee : FAILURE in 5m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b940d36f82c341fd972340a5c2df17de : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2060bcc8b49248c2b7fcfcca11bdb290 : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3cc432f99c9f48589d08c12da7bcd162 : SUCCESS in 1h 37m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/badbc52d34c545458c4c05a6f372aff7 : SUCCESS in 1h 48m 37s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f238d0432303470cb79d13d3c03f989a : SUCCESS in 1h 54m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f01a01f749f44888abf97eefd368425 : RETRY_LIMIT in 9m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b49c269bdb134de8990349dad1471454 : SUCCESS in 1h 42m 50s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497d0e1771f74195b3a100d4d34fd39d : SUCCESS in 1h 37m 45s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f034053c6a84d629bc0431f873bcaab : SUCCESS in 1h 49m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/09f1b4d4ef444e95a34b05ac0a7beee7 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb36ee5cac5e4751bf9a231cd9ef347b : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7569e075f7144017873a250236dadae4 : SUCCESS in 1h 37m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0aa2dbd095a04d389d69bae4da6af508 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c0524108b9e44c6b3657c88a1cd0406 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e02becbb7c6441f3a1214a258922d08b : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/04a561614f66404294156a70f3142585 : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaa69f38e5ab477e9386b3179a0edc38 : SUCCESS in 1h 23m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b5b4e5de9444e62ae00436db7eb0fee : SUCCESS in 2h 03m 00s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8afcde9a93ee4ec59b191976a879c0dd : SUCCESS in 1h 58m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f292d4bb594d4975a496a141f8c3e015 : FAILURE in 11m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93aeaa93a6e94ce9803509c9988c6880 : SUCCESS in 2h 21m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/470516d71dc449a5b008f1fd96c137c4 : FAILURE in 12m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f84df5eeb2c646ddb1f11d681d942779 : FAILURE in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fd67d9610b40e3b0a6767a663ab8b9 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a7f344d8d1474687b1cbfe32a76094 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"623ea659c32d6017e018f086891f39266866db04","revisions":{"623ea659c32d6017e018f086891f39266866db04":{"kind":"REWORK","_number":1,"created":"2026-09-23 08:53:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/1006907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-haproxy_server","ref":"refs/changes/07/1006907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/07/1006907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/07/1006907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/07/1006907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-haproxy_server refs/changes/07/1006907/1"}}},"commit":{"parents":[{"commit":"60e89b664b32e01409cce85ffea6794ef833d816","subject":"haproxy_server: set X-Forwarded-Proto instead of appending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/60e89b664b32e01409cce85ffea6794ef833d816"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-15 20:24:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 08:53:55.000000000","tz":0},"subject":"Ensure GPG key is distributed before importing","message":"Ensure GPG key is distributed before importing\n\nCentOS Stream does not go with pre-shipped EPEL GPG key in images,\nwhich will cause rpm_key to abruptly fail on import attempt.\nWe need to distribute the key to the managed node first.\n\nThis behavior was first introduced by [1]\n\n[1] https://review.opendev.org/#/q/I0f3393581b94aede31dfc98c5ac5edf5a668d283\n\nChange-Id: I60e2f9a544120f9a1646776f870715607c32388e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n(cherry picked from commit cfc70fc008c143277ae2069b57adbeda5932b30b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/623ea659c32d6017e018f086891f39266866db04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-haproxy_server/commit/623ea659c32d6017e018f086891f39266866db04"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"58909b016c717b3b79716ed275c51576c5731050","ours":"60e89b664b32e01409cce85ffea6794ef833d816","theirs":"cfc70fc008c143277ae2069b57adbeda5932b30b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
