)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~613078","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"reduce-packages","hashtags":[],"change_id":"I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a","subject":"Minimise distro packages installed into the lxc image","status":"MERGED","created":"2018-10-24 15:31:02.000000000","updated":"2020-10-06 15:31:14.000000000","submitted":"2020-10-06 15:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"613078-1601998127106-f61217dc","meta_rev_id":"b52119a0aa2e76d0350be046066fde33d4775872","_number":613078,"virtual_id_number":613078,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 15:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-06 08:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":2,"date":"2020-10-06 15:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":1,"date":"2020-10-06 15:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-24 15:32:58.000000000","updated_by":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"reviewer":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"state":"REVIEWER"},{"updated":"2019-01-14 16:00:35.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-09-24 13:20:29.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-10-06 08:44:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-06 15:02:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-06 15:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53261c80a238f4ba48e9e3ce266c7108353939f6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 15:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79efd40156e8b5bda5e1db49cb648987eb219404","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-10-24 15:32:58.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks like a good cleanup!","accounts_in_message":[],"_revision_number":1},{"id":"8b3b3a1f2f64826066402cfa0c1e7ff9c4f7ac79","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 15:57:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1900a28591bf31a3e4f615be4837a54c0c453c16","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 16:13:03.000000000","message":"Topic set to reduce-packages","accounts_in_message":[],"_revision_number":2},{"id":"583104be7785cce7438254194884595988271465","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 16:14:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"317ad952e5d8e0c2a29da4ea68e117cb5c168426","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 16:18:01.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9ccf72bf3ba75ae6078e90c3f3c1aef3ef4a9fd9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-24 16:23:31.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"956563e8583a8eb84fe9ee517c3038deb6c90795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 19:16:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/5/check/openstack-ansible-linters/b83c907/ : SUCCESS in 6m 27s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/5/check/openstack-ansible-functional-centos-7/4fa7931/ : FAILURE in 11m 14s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/5/check/openstack-ansible-functional-opensuse-150/9ca9378/ : SUCCESS in 8m 50s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/5/check/openstack-ansible-functional-opensuse-423/12a74ae/ : SUCCESS in 8m 26s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-functional-ubuntu-bionic/63036dc/ : SUCCESS in 7m 53s\n- openstack-tox-docs http://logs.openstack.org/78/613078/5/check/openstack-tox-docs/db603d4/html/ : SUCCESS in 2m 41s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-dir-centos-7/ec68ea5/ : FAILURE in 10m 41s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-dir-opensuse-423/060b396/ : FAILURE in 18m 14s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-dir-opensuse-150/fcfc6f5/ : FAILURE in 16m 56s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-dir-ubuntu-bionic/a14b225/ : FAILURE in 12m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/85b40d4/ : FAILURE in 9m 16s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-btrfs-centos-7/f04e212/ : FAILURE in 12m 52s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-btrfs-opensuse-423/117201f/ : FAILURE in 11m 58s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-btrfs-opensuse-150/d996128/ : FAILURE in 16m 18s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/9653043/ : FAILURE in 11m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-machinectl-centos-7/4bec16b/ : FAILURE in 9m 22s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-machinectl-opensuse-423/25bdb1f/ : FAILURE in 11m 38s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-machinectl-opensuse-150/d8aeee7/ : FAILURE in 15m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/7d5e78c/ : FAILURE in 10m 34s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-overlayfs-centos-7/106da1a/ : FAILURE in 11m 24s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-overlayfs-opensuse-423/acb1b99/ : FAILURE in 13m 21s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-overlayfs-opensuse-150/257d5e6/ : FAILURE in 17m 14s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/5/check/openstack-ansible-lxc-zfs-ubuntu-bionic/caa1e48/ : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":5},{"id":"fed167f3d2db1624457dc5bcfc60c44bddce42a7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-10-25 21:34:08.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"2092fc8a5d6967c458660dee716bb320d2a1c162","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-26 02:51:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/6/check/openstack-ansible-linters/4717a6b/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/6/check/openstack-ansible-functional-centos-7/706374c/ : FAILURE in 12m 15s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/6/check/openstack-ansible-functional-opensuse-150/eb16a0c/ : SUCCESS in 8m 53s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/6/check/openstack-ansible-functional-opensuse-423/c59fe70/ : SUCCESS in 8m 58s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-functional-ubuntu-bionic/3db7ee8/ : SUCCESS in 15m 39s\n- openstack-tox-docs http://logs.openstack.org/78/613078/6/check/openstack-tox-docs/8bfbf22/html/ : SUCCESS in 2m 48s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-dir-centos-7/433564c/ : FAILURE in 9m 44s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-dir-opensuse-423/80d1419/ : FAILURE in 11m 13s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-dir-opensuse-150/785bae0/ : FAILURE in 16m 07s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-dir-ubuntu-bionic/50620cd/ : FAILURE in 13m 33s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/5db7d55/ : FAILURE in 9m 00s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-btrfs-centos-7/67e7237/ : FAILURE in 13m 42s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-btrfs-opensuse-423/e936c36/ : FAILURE in 12m 38s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-btrfs-opensuse-150/9df93b3/ : FAILURE in 14m 37s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/5fe65b4/ : FAILURE in 15m 11s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-machinectl-centos-7/fb5ea1a/ : FAILURE in 14m 35s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-machinectl-opensuse-423/7eaea65/ : FAILURE in 11m 32s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-machinectl-opensuse-150/57f520d/ : FAILURE in 11m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6f8ab92/ : FAILURE in 16m 59s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-overlayfs-centos-7/2e3d572/ : FAILURE in 10m 37s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-overlayfs-opensuse-423/7e16b29/ : FAILURE in 12m 18s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-overlayfs-opensuse-150/f4e3324/ : FAILURE in 15m 53s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/6/check/openstack-ansible-lxc-zfs-ubuntu-bionic/9d0ec69/ : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":6},{"id":"ddf75114de62daf01eda0d43d10dfdc488c57efd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-01 15:26:13.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"fd41247ddac185a2cadf18c9ff1ecd3a50c27806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-01 15:58:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/7/check/openstack-ansible-linters/c49fb5b/ : SUCCESS in 7m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-centos-7/5a6794b/ : FAILURE in 10m 12s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-150/38d6815/ : SUCCESS in 9m 52s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-423/893d6f2/ : SUCCESS in 9m 53s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-ubuntu-bionic/fb67ac4/ : SUCCESS in 8m 39s\n- openstack-tox-docs http://logs.openstack.org/78/613078/7/check/openstack-tox-docs/8c93c45/html/ : SUCCESS in 2m 27s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-centos-7/205366f/ : FAILURE in 11m 15s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-423/8cf8a88/ : FAILURE in 12m 30s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-150/61d7a73/ : FAILURE in 15m 13s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-ubuntu-bionic/3f1a517/ : FAILURE in 12m 35s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/b99c234/ : FAILURE in 8m 57s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-centos-7/9b76972/ : FAILURE in 12m 38s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-423/36f8177/ : FAILURE in 12m 14s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-150/2dd4569/ : FAILURE in 16m 06s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/d66450a/ : FAILURE in 12m 18s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-centos-7/328cc77/ : FAILURE in 11m 50s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-423/ecec3e7/ : FAILURE in 11m 31s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-150/c9648fa/ : FAILURE in 14m 34s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/706e67e/ : FAILURE in 11m 34s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-centos-7/7b3c1a4/ : FAILURE in 9m 37s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-423/ca8c5a8/ : FAILURE in 12m 35s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-150/50b211c/ : FAILURE in 15m 48s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-zfs-ubuntu-bionic/87752f1/ : FAILURE in 12m 27s","accounts_in_message":[],"_revision_number":7},{"id":"7cad03d80de3d86288bc61a1be1c49c629cbc7f1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-05 16:31:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1165dcbe40692f2c7b431d882cd582106bbccd7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 17:52:22.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/7/check/openstack-ansible-linters/cc3b6b2/ : SUCCESS in 6m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-centos-7/3a2b3fa/ : FAILURE in 11m 04s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-150/ef73a13/ : SUCCESS in 9m 57s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-423/84f2035/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-ubuntu-bionic/7452aa8/ : SUCCESS in 9m 51s\n- openstack-tox-docs http://logs.openstack.org/78/613078/7/check/openstack-tox-docs/001766d/html/ : SUCCESS in 2m 50s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-centos-7/06f64e3/ : FAILURE in 11m 14s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-423/0541975/ : FAILURE in 13m 35s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-150/7c311c7/ : FAILURE in 12m 15s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-ubuntu-bionic/95520c5/ : FAILURE in 12m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/8d256a7/ : FAILURE in 13m 26s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-centos-7/45f9138/ : FAILURE in 15m 13s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-423/5a7c9bb/ : FAILURE in 12m 14s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-150/8896d23/ : FAILURE in 16m 13s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/8127ade/ : FAILURE in 18m 57s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-centos-7/fd5093f/ : FAILURE in 11m 35s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-423/1e77d56/ : FAILURE in 15m 58s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-150/b8d0120/ : FAILURE in 14m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/a8a9731/ : FAILURE in 11m 57s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-centos-7/024bac5/ : FAILURE in 11m 46s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-423/9e53a0f/ : FAILURE in 12m 16s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-150/c1584d8/ : FAILURE in 15m 41s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-zfs-ubuntu-bionic/e051656/ : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":7},{"id":"19418632e2a7ee3bb2b78f09c849f1ed2bd4ce4e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-07 09:23:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"35c25e083becf64f166a7eb3d317523ee380306d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 09:47:27.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/7/check/openstack-ansible-linters/25da777/ : SUCCESS in 6m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-centos-7/17dbc80/ : FAILURE in 13m 44s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-150/e3b2274/ : SUCCESS in 9m 40s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-opensuse-423/0c59b04/ : SUCCESS in 8m 57s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-functional-ubuntu-bionic/3d8f05f/ : SUCCESS in 9m 27s\n- openstack-tox-docs http://logs.openstack.org/78/613078/7/check/openstack-tox-docs/bba1acf/html/ : SUCCESS in 5m 28s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-centos-7/7903226/ : FAILURE in 13m 26s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-423/d704fab/ : FAILURE in 12m 25s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-opensuse-150/107fe81/ : FAILURE in 17m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-dir-ubuntu-bionic/f73f993/ : FAILURE in 15m 17s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/2c809cf/ : FAILURE in 11m 26s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-centos-7/47e4f32/ : FAILURE in 11m 55s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-423/d2522de/ : FAILURE in 13m 59s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-btrfs-opensuse-150/2381429/ : FAILURE in 14m 29s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/628fa07/ : FAILURE in 12m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-centos-7/2f2a299/ : FAILURE in 12m 59s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-423/8c34d60/ : FAILURE in 12m 09s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-machinectl-opensuse-150/c7aaa60/ : FAILURE in 17m 08s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/3ae1b30/ : FAILURE in 15m 04s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-centos-7/ce2fea9/ : FAILURE in 13m 10s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-423/340cc42/ : FAILURE in 13m 36s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-overlayfs-opensuse-150/b750e4b/ : FAILURE in 13m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/7/check/openstack-ansible-lxc-zfs-ubuntu-bionic/0f4d0df/ : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":7},{"id":"1b9cbd5966cd644f12f2b609eb4b9234e1d11de4","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-11-07 12:36:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"c7b9c04923cc6b1a986a5af9bb1e6b6b3b12db8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 12:59:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/8/check/openstack-ansible-linters/787a935/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-centos-7/c7ae479/ : FAILURE in 12m 59s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-opensuse-150/31370f9/ : SUCCESS in 8m 23s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-opensuse-423/b42b551/ : SUCCESS in 9m 14s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-ubuntu-bionic/2c27e60/ : SUCCESS in 13m 18s\n- openstack-tox-docs http://logs.openstack.org/78/613078/8/check/openstack-tox-docs/de8b633/html/ : SUCCESS in 3m 03s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-centos-7/d6af477/ : FAILURE in 8m 55s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-opensuse-423/eeee3df/ : FAILURE in 11m 52s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-opensuse-150/44ea017/ : FAILURE in 15m 34s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-ubuntu-bionic/528c260/ : FAILURE in 17m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/429d368/ : FAILURE in 13m 40s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-centos-7/64b1174/ : FAILURE in 12m 45s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-opensuse-423/83568e7/ : FAILURE in 11m 12s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-opensuse-150/2e69ba9/ : FAILURE in 16m 03s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/e9a51e9/ : FAILURE in 17m 09s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-centos-7/7e6e213/ : FAILURE in 12m 12s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-opensuse-423/44530a4/ : FAILURE in 10m 52s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-opensuse-150/85969ba/ : FAILURE in 13m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/e67a5a2/ : FAILURE in 12m 59s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-centos-7/b51dc33/ : FAILURE in 10m 44s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-opensuse-423/b5390cd/ : FAILURE in 15m 04s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-opensuse-150/17aeb01/ : FAILURE in 13m 55s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-zfs-ubuntu-bionic/9cf1eda/ : FAILURE in 10m 48s","accounts_in_message":[],"_revision_number":8},{"id":"0d5e627bdbbb944fdfccf7c4913dad873b8a2187","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-10 15:08:00.000000000","message":"Patch Set 8:\n\nrecheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"eeb020aba591bd4726da624c899ad0ae20719e38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-10 15:39:24.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/8/check/openstack-ansible-linters/9e879db/ : SUCCESS in 5m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-centos-7/ac6f05b/ : FAILURE in 11m 00s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-opensuse-150/8a38eff/ : SUCCESS in 10m 37s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-opensuse-423/3e806a9/ : SUCCESS in 7m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-functional-ubuntu-bionic/f47130d/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/78/613078/8/check/openstack-tox-docs/acfa018/html/ : SUCCESS in 2m 23s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-centos-7/986f96b/ : FAILURE in 10m 23s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-opensuse-423/2aeecf3/ : FAILURE in 12m 32s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-opensuse-150/57dde82/ : FAILURE in 10m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-dir-ubuntu-bionic/8c0b2da/ : FAILURE in 12m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/71e8a71/ : FAILURE in 8m 55s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-centos-7/a5f5161/ : FAILURE in 13m 04s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-opensuse-423/a730003/ : FAILURE in 15m 21s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-btrfs-opensuse-150/2b42ec0/ : FAILURE in 12m 56s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/92acdd3/ : FAILURE in 13m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-centos-7/5dcaae0/ : FAILURE in 9m 11s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-opensuse-423/a54bab6/ : FAILURE in 13m 18s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-machinectl-opensuse-150/dc69e45/ : FAILURE in 14m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/b37f399/ : FAILURE in 9m 48s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-centos-7/40f90be/ : FAILURE in 10m 53s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-opensuse-423/a694b16/ : FAILURE in 10m 41s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-overlayfs-opensuse-150/2b032dc/ : FAILURE in 11m 29s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/8/check/openstack-ansible-lxc-zfs-ubuntu-bionic/13aef8a/ : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":8},{"id":"09a78b8c2fbe3c77baa89019cb35abc2f6736ca2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 07:06:34.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5f462a45fb7c61f2752efd30a2b26e3cdc0569f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 07:28:36.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/9/check/openstack-ansible-linters/b10b013/ : SUCCESS in 6m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/9/check/openstack-ansible-functional-centos-7/3a3c845/ : FAILURE in 9m 57s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/9/check/openstack-ansible-functional-opensuse-150/956985b/ : SUCCESS in 10m 24s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/9/check/openstack-ansible-functional-opensuse-423/1bd230f/ : SUCCESS in 9m 31s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-functional-ubuntu-bionic/76be68e/ : SUCCESS in 9m 20s\n- openstack-tox-docs http://logs.openstack.org/78/613078/9/check/openstack-tox-docs/3845880/html/ : SUCCESS in 2m 46s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-dir-centos-7/54d18cf/ : FAILURE in 11m 17s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-dir-opensuse-423/9e2b195/ : FAILURE in 12m 26s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-dir-opensuse-150/c0f59e5/ : FAILURE in 16m 13s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-dir-ubuntu-bionic/051a330/ : FAILURE in 12m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/6b164c8/ : FAILURE in 13m 40s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-btrfs-centos-7/dbb14ac/ : FAILURE in 10m 25s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-btrfs-opensuse-423/34f8202/ : FAILURE in 12m 23s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-btrfs-opensuse-150/ee036c8/ : FAILURE in 15m 01s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/52879fd/ : FAILURE in 11m 37s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-machinectl-centos-7/f259ab3/ : FAILURE in 11m 34s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-machinectl-opensuse-423/74c5fa6/ : FAILURE in 12m 54s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-machinectl-opensuse-150/564e90f/ : FAILURE in 14m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6a20bed/ : FAILURE in 11m 43s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-overlayfs-centos-7/1fc1a8c/ : FAILURE in 12m 11s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-overlayfs-opensuse-423/6580654/ : FAILURE in 12m 30s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-overlayfs-opensuse-150/19efc2a/ : FAILURE in 16m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/9/check/openstack-ansible-lxc-zfs-ubuntu-bionic/a0af051/ : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":9},{"id":"f61d1f3bed362dda013ac40137f0fa19a40dce57","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 08:55:02.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"0a0fc3e0d2180b50afe3e62d22dab7ed085c8a9b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 09:23:17.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"759fb8f1c676187cd8fe42b81f591057cc1ba060","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 09:54:31.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"6c83bebd7e86c1aaf66f2c646e8649efc7c64873","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 10:21:10.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e50ce0258f5f26fafe27a338fab6213f74c89fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 10:59:49.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/13/check/openstack-ansible-linters/9b750e7/ : SUCCESS in 6m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/13/check/openstack-ansible-functional-centos-7/fb80022/ : FAILURE in 13m 02s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/13/check/openstack-ansible-functional-opensuse-150/ceb4b8c/ : SUCCESS in 10m 28s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/13/check/openstack-ansible-functional-opensuse-423/0874cca/ : SUCCESS in 14m 51s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-functional-ubuntu-bionic/0785a5a/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/78/613078/13/check/openstack-tox-docs/5bc66b7/html/ : SUCCESS in 3m 14s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-dir-centos-7/4d9388f/ : FAILURE in 13m 57s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-dir-opensuse-423/6a1ead9/ : FAILURE in 14m 45s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-dir-opensuse-150/67f5a8a/ : FAILURE in 18m 55s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-dir-ubuntu-bionic/24ad9f6/ : SUCCESS in 19m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/976ede9/ : FAILURE in 19m 23s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-btrfs-centos-7/8464662/ : FAILURE in 11m 44s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-btrfs-opensuse-423/db8c117/ : FAILURE in 16m 54s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-btrfs-opensuse-150/9ce4296/ : FAILURE in 15m 58s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/6a9b5eb/ : SUCCESS in 12m 14s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-machinectl-centos-7/83a4f1f/ : FAILURE in 15m 32s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-machinectl-opensuse-423/2d0d6e8/ : FAILURE in 11m 41s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-machinectl-opensuse-150/7b0791b/ : FAILURE in 16m 25s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/ae3734c/ : SUCCESS in 12m 42s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-overlayfs-centos-7/185dfb1/ : FAILURE in 13m 09s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-overlayfs-opensuse-423/cdb6cf6/ : FAILURE in 12m 01s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-overlayfs-opensuse-150/8bc8778/ : FAILURE in 16m 22s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/13/check/openstack-ansible-lxc-zfs-ubuntu-bionic/be290ab/ : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":13},{"id":"efab351eb5368c7e5ad9689894c9724d297feae4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 11:17:28.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"e03602c9a6463999b5914fe81db26b680cedcfba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 12:00:11.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/14/check/openstack-ansible-linters/060e185/ : SUCCESS in 5m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/14/check/openstack-ansible-functional-centos-7/4376ff9/ : FAILURE in 10m 29s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/14/check/openstack-ansible-functional-opensuse-150/e0ccf3d/ : SUCCESS in 10m 19s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/14/check/openstack-ansible-functional-opensuse-423/fa3590e/ : SUCCESS in 9m 43s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-functional-ubuntu-bionic/27b4b87/ : SUCCESS in 8m 26s\n- openstack-tox-docs http://logs.openstack.org/78/613078/14/check/openstack-tox-docs/fe23d16/html/ : SUCCESS in 3m 34s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-dir-centos-7/4d72f62/ : FAILURE in 14m 49s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-dir-opensuse-423/a035a21/ : SUCCESS in 13m 36s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-dir-opensuse-150/9871834/ : SUCCESS in 16m 53s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-dir-ubuntu-bionic/1593a24/ : SUCCESS in 12m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/eaa5652/ : FAILURE in 16m 19s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-btrfs-centos-7/fa87ece/ : FAILURE in 12m 48s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-btrfs-opensuse-423/d188944/ : SUCCESS in 13m 21s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-btrfs-opensuse-150/6314fc0/ : SUCCESS in 17m 52s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/743ccd8/ : SUCCESS in 24m 57s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-machinectl-centos-7/02f0593/ : FAILURE in 12m 32s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-machinectl-opensuse-423/69601a8/ : SUCCESS in 12m 23s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-machinectl-opensuse-150/82daf6e/ : SUCCESS in 16m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/f500717/ : SUCCESS in 11m 05s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-overlayfs-centos-7/212b1e3/ : FAILURE in 12m 56s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-overlayfs-opensuse-423/2fceeea/ : SUCCESS in 19m 02s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-overlayfs-opensuse-150/3ccadcf/ : SUCCESS in 18m 44s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/14/check/openstack-ansible-lxc-zfs-ubuntu-bionic/b2bc3bb/ : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":14},{"id":"c36ba9182df15d4915d9aec7ecc60980f8ef2cff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 12:03:55.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"2bb3fd286109184df809040a68daa784446c7882","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 12:30:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/15/check/openstack-ansible-linters/d5305e8/ : SUCCESS in 6m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/15/check/openstack-ansible-functional-centos-7/09f0097/ : SUCCESS in 9m 34s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/15/check/openstack-ansible-functional-opensuse-150/21075c0/ : FAILURE in 11m 43s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/15/check/openstack-ansible-functional-opensuse-423/c0264ec/ : SUCCESS in 8m 34s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-functional-ubuntu-bionic/828358f/ : SUCCESS in 10m 01s\n- openstack-tox-docs http://logs.openstack.org/78/613078/15/check/openstack-tox-docs/ac4177a/html/ : SUCCESS in 3m 02s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-dir-centos-7/83b54f7/ : SUCCESS in 16m 14s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-dir-opensuse-423/c02b0e4/ : SUCCESS in 12m 57s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-dir-opensuse-150/7945d7a/ : SUCCESS in 16m 39s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-dir-ubuntu-bionic/b3170f1/ : SUCCESS in 18m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/44b7c5e/ : FAILURE in 11m 21s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-btrfs-centos-7/98363a9/ : SUCCESS in 14m 33s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-btrfs-opensuse-423/078cc03/ : SUCCESS in 16m 39s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-btrfs-opensuse-150/275c818/ : SUCCESS in 15m 21s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/253524d/ : SUCCESS in 12m 18s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-machinectl-centos-7/b55d393/ : SUCCESS in 16m 16s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-machinectl-opensuse-423/1760d3f/ : SUCCESS in 11m 08s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-machinectl-opensuse-150/45846ed/ : SUCCESS in 15m 00s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/13051b3/ : SUCCESS in 18m 22s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-overlayfs-centos-7/ecdeaa7/ : SUCCESS in 14m 58s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-overlayfs-opensuse-423/b9de2b0/ : SUCCESS in 13m 15s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-overlayfs-opensuse-150/17a00a1/ : SUCCESS in 16m 44s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/15/check/openstack-ansible-lxc-zfs-ubuntu-bionic/90c8e6a/ : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":15},{"id":"d47e66e60bc7d5c6f69b8d25c17a3d63bc82961a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-11-20 13:01:56.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"85b4c022d3710db1425d6ea71723c9edc957593e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-11-20 13:02:19.000000000","message":"Patch Set 15: Code-Review+1\n\nI am positive, but would like to see the impact on timing.\nAdding more things into the containers later might take more time if all the services are using it.\n\nBut in that case, it might be better to assign it to openstack_hosts nowadays, as it would be common with nspawn.","accounts_in_message":[],"_revision_number":15},{"id":"792710e7b015424408abe4d40c557d60a65f3c4f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 13:38:26.000000000","message":"Patch Set 15:\n\n\u003e I am positive, but would like to see the impact on timing.\n \u003e Adding more things into the containers later might take more time\n \u003e if all the services are using it.\n \u003e \n \u003e But in that case, it might be better to assign it to\n \u003e openstack_hosts nowadays, as it would be common with nspawn.\n\nAgreed that it isn\u0027t totally clear what the correct strategy should be here. The patch currently installs the minimum amount possible to achieve a CI pass, in particular some of the yum/apt plugins aren\u0027t obvious and there is a further philosophical question of if our roles should be \"self sufficient\" in terms ensuring the packages they need are present.\n\nThe excercise of stripping excess global packages out has revealed a bunch of accidental cross-dependencies on package install lists in the roles.","accounts_in_message":[],"_revision_number":15},{"id":"c07b6862cdc20c9819c47c92320ef75eb794bfee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 13:56:11.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- openstack-ansible-integrated-deploy-aio http://logs.openstack.org/78/613078/15/experimental/openstack-ansible-integrated-deploy-aio/7404f80/ : FAILURE in 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a58382bbecf6fd4eba902eb729c26d0e40f0a2d5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-20 17:27:07.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"2b99c88c0ea037cf906d8a1b706fc4eab934952e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 18:46:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/16/check/openstack-ansible-linters/2139e3f/ : SUCCESS in 8m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/16/check/openstack-ansible-functional-centos-7/eefd744/ : SUCCESS in 11m 45s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/16/check/openstack-ansible-functional-opensuse-150/7f2b9ac/ : SUCCESS in 9m 43s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/16/check/openstack-ansible-functional-opensuse-423/9ddf384/ : SUCCESS in 10m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-functional-ubuntu-bionic/81eeeda/ : SUCCESS in 10m 42s\n- openstack-tox-docs http://logs.openstack.org/78/613078/16/check/openstack-tox-docs/80754ae/html/ : SUCCESS in 3m 06s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-dir-centos-7/ac3faf3/ : SUCCESS in 22m 11s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-dir-opensuse-423/6fe38bf/ : SUCCESS in 14m 02s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-dir-opensuse-150/da051a0/ : SUCCESS in 48m 59s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-dir-ubuntu-bionic/bae2e0a/ : SUCCESS in 23m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/b306acd/ : FAILURE in 21m 41s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-btrfs-centos-7/fe1dca8/ : SUCCESS in 17m 04s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-btrfs-opensuse-423/789e39a/ : SUCCESS in 15m 08s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-btrfs-opensuse-150/8e2789f/ : SUCCESS in 17m 11s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/2393229/ : SUCCESS in 17m 48s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-machinectl-centos-7/048aeba/ : SUCCESS in 19m 01s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-machinectl-opensuse-423/62b8d7b/ : SUCCESS in 12m 30s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-machinectl-opensuse-150/147d3f6/ : SUCCESS in 16m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/416549d/ : SUCCESS in 12m 27s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-overlayfs-centos-7/4806910/ : SUCCESS in 14m 40s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-overlayfs-opensuse-423/26a467a/ : SUCCESS in 45m 34s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-overlayfs-opensuse-150/9d3f5c6/ : SUCCESS in 20m 08s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/16/check/openstack-ansible-lxc-zfs-ubuntu-bionic/fd3c2d3/ : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":16},{"id":"c661613f35a0163eab4e04bb68d8f6f5deafbf73","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-21 12:04:28.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"9f291168aff60a495015c94a3e465c966de6b609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 12:37:54.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/17/check/openstack-ansible-linters/8592d1e/ : SUCCESS in 7m 47s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-centos-7/13d1d22/ : SUCCESS in 10m 32s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-150/5acf3ee/ : SUCCESS in 8m 29s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-423/2e62982/ : SUCCESS in 10m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-ubuntu-bionic/a03844b/ : SUCCESS in 15m 25s\n- openstack-tox-docs http://logs.openstack.org/78/613078/17/check/openstack-tox-docs/2879ebc/html/ : SUCCESS in 3m 58s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-centos-7/080adbc/ : SUCCESS in 16m 27s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-423/b101fea/ : SUCCESS in 12m 54s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-150/9821d48/ : SUCCESS in 16m 36s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-ubuntu-bionic/7195921/ : SUCCESS in 13m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/70d5a46/ : FAILURE in 11m 01s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-centos-7/cfdb4b8/ : SUCCESS in 18m 49s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-423/271fb2b/ : SUCCESS in 14m 11s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-150/1fa7cc8/ : SUCCESS in 17m 20s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/a90a3b5/ : SUCCESS in 12m 30s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-centos-7/cca32ae/ : SUCCESS in 18m 25s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-423/f4529dc/ : SUCCESS in 12m 56s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-150/948a845/ : SUCCESS in 15m 43s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/4f2e32d/ : SUCCESS in 12m 20s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-centos-7/61b75f4/ : SUCCESS in 18m 47s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-423/35cb03a/ : SUCCESS in 13m 45s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-150/2b07b36/ : SUCCESS in 24m 54s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-zfs-ubuntu-bionic/45e67e9/ : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":17},{"id":"6e4124de9cc0b146953c9d88e0157b04b3f5c8a8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-11-21 13:42:27.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"cd394e2370fdb86464cbc841e7177f1bdbdb10b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-21 15:34:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- openstack-ansible-integrated-deploy-aio http://logs.openstack.org/78/613078/17/experimental/openstack-ansible-integrated-deploy-aio/49a21cb/ : FAILURE in 1h 49m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"aa805b84188158de402a99dc8b5f9b2460649e63","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-11-25 15:43:28.000000000","message":"Patch Set 17: Code-Review-1\n\nthis seems problematic for packages we don\u0027t build wheels for,  like the tempest plugins. \n\neither this needs a depends on with a fix of the building of tempest plugins,  or needs to take care of that fact by keeping the packages necessary for building those plugins. (error on integrated is command \u0027x86_64-linux-gnu-gcc\u0027 failed with exit status 1 ).\n\nFor the latter,  a \u0027todo\u0027 should be added,  to explain the future removal when the tempest plugins issue is addressed.","accounts_in_message":[],"_revision_number":17},{"id":"b4dd58e13a111f86aa1a4c2094911b859dfd9890","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-12-01 11:58:24.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"b1e46e003fb73ec5275e3df8ce3e3e36607ec5ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-01 13:33:24.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/17/check/openstack-ansible-linters/19548e4/ : SUCCESS in 4m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-centos-7/955cf59/ : SUCCESS in 9m 23s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-150/7c8c30d/ : SUCCESS in 9m 43s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-423/41aea8f/ : SUCCESS in 8m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-ubuntu-bionic/f17020b/ : SUCCESS in 18m 22s\n- openstack-tox-docs http://logs.openstack.org/78/613078/17/check/openstack-tox-docs/3386d83/html/ : SUCCESS in 2m 59s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-centos-7/051e61a/ : SUCCESS in 14m 49s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-423/d954e96/ : SUCCESS in 14m 31s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-150/ac3e24e/ : SUCCESS in 14m 00s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-ubuntu-bionic/4a6e7e8/ : SUCCESS in 11m 29s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/3d93397/ : FAILURE in 17m 53s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-centos-7/2f00459/ : SUCCESS in 13m 55s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-423/e6a2e71/ : TIMED_OUT in 1h 32m 05s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-150/ac3f161/ : SUCCESS in 18m 50s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/5b3a59c/ : SUCCESS in 11m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-centos-7/ee85787/ : SUCCESS in 14m 09s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-423/e227be1/ : SUCCESS in 13m 01s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-150/d690f45/ : SUCCESS in 17m 30s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/352fcdb/ : SUCCESS in 12m 24s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-centos-7/2eb1cbf/ : SUCCESS in 15m 08s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-423/5f47b0f/ : SUCCESS in 12m 51s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-150/c4036b5/ : SUCCESS in 19m 04s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-zfs-ubuntu-bionic/df03c91/ : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":17},{"id":"e0761e5a176616b9235deea37c73655ac1f59b94","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-12-01 20:45:02.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"a4d72f83f3abe89c9e4f384ee78afbbd9b72b3a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-01 21:07:26.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/17/check/openstack-ansible-linters/04141eb/ : SUCCESS in 4m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-centos-7/790a502/ : SUCCESS in 9m 51s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-150/1e6cb71/ : SUCCESS in 8m 53s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-opensuse-423/2669630/ : SUCCESS in 7m 55s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-functional-ubuntu-bionic/7a7d741/ : SUCCESS in 8m 55s\n- openstack-tox-docs http://logs.openstack.org/78/613078/17/check/openstack-tox-docs/f56efc3/html/ : SUCCESS in 2m 26s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-centos-7/8dceea0/ : SUCCESS in 13m 35s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-423/5ddb37f/ : SUCCESS in 12m 12s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-opensuse-150/b155302/ : SUCCESS in 15m 18s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-dir-ubuntu-bionic/5ec6a22/ : SUCCESS in 10m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/0d8a082/ : FAILURE in 9m 04s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-centos-7/1317394/ : SUCCESS in 18m 58s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-423/d235302/ : SUCCESS in 12m 53s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-btrfs-opensuse-150/b4a663d/ : SUCCESS in 13m 33s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/97545c7/ : SUCCESS in 18m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-centos-7/14a1a3e/ : SUCCESS in 13m 29s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-423/9ae870e/ : SUCCESS in 10m 07s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-machinectl-opensuse-150/0fb8f5b/ : SUCCESS in 14m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/4fd984a/ : SUCCESS in 19m 50s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-centos-7/ae33421/ : SUCCESS in 13m 09s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-423/c65d6cf/ : SUCCESS in 12m 38s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-overlayfs-opensuse-150/aed0f75/ : SUCCESS in 12m 26s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/17/check/openstack-ansible-lxc-zfs-ubuntu-bionic/a9e573e/ : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":17},{"id":"485ca9a7318538246c95324b3c04f15542931547","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2018-12-01 22:01:37.000000000","message":"Patch Set 17:\n\ncheck experimental","accounts_in_message":[],"_revision_number":17},{"id":"00060837130b53621df5b8735360b426237a18cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-01 23:35:48.000000000","message":"Patch Set 17:\n\nBuild succeeded (experimental pipeline).\n\n- openstack-ansible-integrated-deploy-aio http://logs.openstack.org/78/613078/17/experimental/openstack-ansible-integrated-deploy-aio/2028bc4/ : FAILURE in 1h 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d00ab9488989d31333b6b477bceeeffe972507ef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-03 09:32:59.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"62a64a471456a7838e7acc6e2ec99a3ccca2cc72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-03 09:54:12.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/18/check/openstack-ansible-linters/0d8ebf3/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-centos-7/f2c75a3/ : SUCCESS in 8m 47s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-opensuse-150/de0c5d8/ : SUCCESS in 9m 56s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-opensuse-423/fccdfe0/ : SUCCESS in 7m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-ubuntu-bionic/fac26ae/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/78/613078/18/check/openstack-tox-docs/96f056b/html/ : SUCCESS in 3m 39s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-centos-7/3a2c3ef/ : FAILURE in 12m 01s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-opensuse-423/3122378/ : SUCCESS in 12m 13s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-opensuse-150/2aa0636/ : SUCCESS in 15m 49s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-ubuntu-bionic/5cef185/ : FAILURE in 12m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/75de433/ : FAILURE in 11m 36s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-centos-7/8e740b2/ : FAILURE in 13m 34s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-opensuse-423/d7cae91/ : SUCCESS in 12m 38s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-opensuse-150/dbf848e/ : SUCCESS in 16m 20s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/432db4d/ : FAILURE in 10m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-centos-7/ec40bcc/ : FAILURE in 10m 58s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-opensuse-423/f9800da/ : SUCCESS in 12m 33s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-opensuse-150/43f5032/ : SUCCESS in 14m 41s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/3d8290f/ : FAILURE in 11m 03s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-centos-7/d58d722/ : FAILURE in 14m 27s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-opensuse-423/a712c35/ : SUCCESS in 12m 37s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-opensuse-150/db02498/ : SUCCESS in 15m 58s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-zfs-ubuntu-bionic/b28cc3c/ : FAILURE in 10m 31s","accounts_in_message":[],"_revision_number":18},{"id":"3813f2ad1c3f7774fd1f673b9824c570d4e16d75","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-07 12:06:36.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"b1cb0ee76e152f11e62c39b83bf631fcd8a1b020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 12:33:11.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/18/check/openstack-ansible-linters/68219c7/ : SUCCESS in 6m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-centos-7/ec00ef7/ : SUCCESS in 8m 40s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-opensuse-150/e0ada01/ : SUCCESS in 9m 20s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-opensuse-423/fe96a50/ : SUCCESS in 8m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-functional-ubuntu-bionic/1257562/ : SUCCESS in 8m 28s\n- openstack-tox-docs http://logs.openstack.org/78/613078/18/check/openstack-tox-docs/bd79fd3/html/ : SUCCESS in 3m 08s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-centos-7/5a3e8e7/ : FAILURE in 11m 47s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-opensuse-423/30e6cd4/ : SUCCESS in 12m 37s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-opensuse-150/6309b2e/ : SUCCESS in 18m 20s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-dir-ubuntu-bionic/4bfbed9/ : FAILURE in 11m 40s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/a679e93/ : FAILURE in 11m 13s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-centos-7/955a894/ : FAILURE in 13m 14s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-opensuse-423/db43bb3/ : SUCCESS in 13m 37s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-btrfs-opensuse-150/48a62ee/ : SUCCESS in 13m 59s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/fada91d/ : FAILURE in 11m 13s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-centos-7/f3fd014/ : FAILURE in 12m 01s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-opensuse-423/05392fd/ : SUCCESS in 12m 29s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-machinectl-opensuse-150/89940f5/ : SUCCESS in 17m 19s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/3a67576/ : FAILURE in 10m 15s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-centos-7/d6e716b/ : FAILURE in 13m 48s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-opensuse-423/d784f49/ : SUCCESS in 13m 19s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-overlayfs-opensuse-150/4457a68/ : SUCCESS in 16m 21s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/18/check/openstack-ansible-lxc-zfs-ubuntu-bionic/b8f37aa/ : FAILURE in 10m 34s","accounts_in_message":[],"_revision_number":18},{"id":"37ad1ab580b56571112ede8dca829fee10c7f6ae","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-07 12:44:33.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"91dc51e991c05a72e5c263c31d43f03363e6ece7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 13:06:03.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/19/check/openstack-ansible-linters/d591c4f/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-centos-7/f96ed9b/ : SUCCESS in 8m 43s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-150/3da554c/ : SUCCESS in 8m 51s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-423/5bb6c5f/ : SUCCESS in 8m 38s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-ubuntu-bionic/8974985/ : SUCCESS in 7m 35s\n- openstack-tox-docs http://logs.openstack.org/78/613078/19/check/openstack-tox-docs/4e29709/html/ : SUCCESS in 2m 33s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-centos-7/3d9d80e/ : SUCCESS in 12m 49s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-423/d8e0cec/ : SUCCESS in 11m 26s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-150/c0c9952/ : SUCCESS in 14m 49s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-ubuntu-bionic/a16988e/ : SUCCESS in 12m 42s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/ffd21ff/ : SUCCESS in 13m 33s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-centos-7/15f7a6b/ : FAILURE in 15m 21s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-423/2bacc35/ : SUCCESS in 12m 07s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-150/b20b889/ : SUCCESS in 16m 31s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/f91ed57/ : SUCCESS in 10m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-centos-7/6b021ed/ : SUCCESS in 12m 36s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-423/3db6e3c/ : SUCCESS in 12m 37s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-150/adc76b6/ : SUCCESS in 15m 27s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/245ad60/ : SUCCESS in 13m 27s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-centos-7/94f19ac/ : FAILURE in 14m 15s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-423/5c4ab31/ : SUCCESS in 14m 06s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-150/97f5421/ : SUCCESS in 17m 34s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-zfs-ubuntu-bionic/00588d6/ : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":19},{"id":"c99d6e0ee202b95e8fcf9e739a029b30d23576cd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-07 21:32:46.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"3395b3d13890e0eadfecc0a56272689807ec2102","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 21:56:13.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/19/check/openstack-ansible-linters/774d033/ : SUCCESS in 4m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-centos-7/9d926c3/ : SUCCESS in 9m 07s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-150/fcf403e/ : SUCCESS in 8m 40s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-423/9a19e9e/ : SUCCESS in 8m 17s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-ubuntu-bionic/5de1d1a/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/78/613078/19/check/openstack-tox-docs/0fb0a21/html/ : SUCCESS in 3m 01s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-centos-7/9a66ad1/ : SUCCESS in 12m 56s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-423/e07a36f/ : SUCCESS in 12m 35s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-150/9bf27da/ : SUCCESS in 15m 51s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-ubuntu-bionic/8aaa0fb/ : SUCCESS in 10m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/15028fb/ : SUCCESS in 11m 28s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-centos-7/7782654/ : FAILURE in 14m 14s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-423/977fbd3/ : SUCCESS in 12m 19s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-150/b3011cf/ : SUCCESS in 17m 05s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/258dcb8/ : SUCCESS in 11m 55s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-centos-7/662d0fe/ : SUCCESS in 13m 39s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-423/0c0df77/ : SUCCESS in 12m 07s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-150/3d8201e/ : SUCCESS in 15m 16s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/e9088ab/ : SUCCESS in 14m 50s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-centos-7/dc8f90c/ : FAILURE in 15m 35s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-423/d2c544b/ : SUCCESS in 12m 15s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-150/bcf9378/ : SUCCESS in 15m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-zfs-ubuntu-bionic/f91a5c1/ : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":19},{"id":"f7f2edc48da7bb79678a7a30c73ee28ea2da1527","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-11 14:59:02.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"ccf08d0049a8e101ffd3a5c8b7b293aaac419efe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 15:36:50.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/19/check/openstack-ansible-linters/7d91111/ : SUCCESS in 6m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-centos-7/8f1f545/ : SUCCESS in 9m 18s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-150/341f04d/ : SUCCESS in 10m 08s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-423/f88c5b8/ : SUCCESS in 8m 53s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-ubuntu-bionic/d04a6a0/ : SUCCESS in 19m 51s\n- openstack-tox-docs http://logs.openstack.org/78/613078/19/check/openstack-tox-docs/ebd8935/html/ : SUCCESS in 2m 55s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-centos-7/fafec76/ : SUCCESS in 13m 40s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-423/402ee84/ : SUCCESS in 12m 41s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-150/3a77c4f/ : SUCCESS in 16m 28s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-ubuntu-bionic/5e249b4/ : SUCCESS in 12m 09s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/63e507c/ : SUCCESS in 12m 46s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-centos-7/ec80544/ : FAILURE in 16m 17s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-423/3e3b328/ : SUCCESS in 13m 26s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-150/2f71b45/ : SUCCESS in 14m 41s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/96992d8/ : SUCCESS in 12m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-centos-7/7ccf6c2/ : SUCCESS in 14m 09s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-423/b4b360f/ : SUCCESS in 13m 06s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-150/92ddb89/ : SUCCESS in 15m 15s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/ce02a15/ : SUCCESS in 18m 51s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-centos-7/c93c973/ : FAILURE in 14m 20s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-423/dd89c78/ : SUCCESS in 14m 15s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-150/05c62cd/ : SUCCESS in 16m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-zfs-ubuntu-bionic/748a31f/ : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":19},{"id":"ad65a1ebc6358116ab00b320686d8b12c467d3e7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-01-14 10:31:57.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"ca440ccaba5b9502ee1b722220559ca7a6f6941d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 11:04:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/613078/19/check/openstack-ansible-linters/f1934b9/ : SUCCESS in 14m 40s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-centos-7/d481935/ : SUCCESS in 14m 21s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-150/63d0995/ : SUCCESS in 15m 05s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-opensuse-423/174e31b/ : SUCCESS in 12m 12s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-functional-ubuntu-bionic/c020dbc/ : SUCCESS in 20m 11s\n- openstack-tox-docs http://logs.openstack.org/78/613078/19/check/openstack-tox-docs/7cb8b1e/html/ : SUCCESS in 9m 59s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-centos-7/9637fa4/ : SUCCESS in 18m 59s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-423/9858c81/ : SUCCESS in 15m 47s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-opensuse-150/4a3207b/ : SUCCESS in 19m 40s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-dir-ubuntu-bionic/938e381/ : SUCCESS in 25m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/9a097d2/ : SUCCESS in 22m 17s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-423/49c9487/ : SUCCESS in 21m 08s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-btrfs-opensuse-150/97a9e1c/ : SUCCESS in 20m 36s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/d609564/ : SUCCESS in 25m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-centos-7/a0f3928/ : SUCCESS in 16m 33s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-423/0f98343/ : SUCCESS in 23m 44s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-machinectl-opensuse-150/bc6a2fe/ : SUCCESS in 21m 26s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6f3e8cb/ : SUCCESS in 20m 36s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-423/adafd8f/ : SUCCESS in 22m 38s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-overlayfs-opensuse-150/b052e53/ : SUCCESS in 18m 27s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/613078/19/check/openstack-ansible-lxc-zfs-ubuntu-bionic/a7351b8/ : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":19},{"id":"7df03019ee7f0b43fb4eae567c2cc47cbea24f03","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-14 11:07:23.000000000","message":"Patch Set 19:\n\ncheck experimental","accounts_in_message":[],"_revision_number":19},{"id":"c0eb3b3c221ce921f9362580ec5a256ba4609080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 12:36:30.000000000","message":"Patch Set 19:\n\nBuild succeeded (experimental pipeline).\n\n- openstack-ansible-integrated-deploy-aio http://logs.openstack.org/78/613078/19/experimental/openstack-ansible-integrated-deploy-aio/1a1f750/ : FAILURE in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7420e62a97059bde17249d5bc1d57df1cd04d6be","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-01-14 16:00:35.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"b9d3c00b6aa6bf2c8d69bdf5ee9ba45de49e077f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-01-14 16:05:38.000000000","message":"Patch Set 19:\n\n\u003e Patch Set 19: Code-Review+2\n\nNote that this will fail the integrtated build because there is no longer enough stuff installed for the tempest role to build the plugins from source","accounts_in_message":[],"_revision_number":19},{"id":"1b183d61005d5f20b849c980a05f966132f25c67","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-01-14 16:23:22.000000000","message":"Patch Set 19: Workflow-1\n\nLet\u0027s get this working with the integrated build first.","accounts_in_message":[],"_revision_number":19},{"id":"5e5c63aaa7c3516f1a05009b59b3c5d401e14417","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-03 07:38:52.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a4c83064379660cbef5399800daec4e3fdc128d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 08:03:43.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce476e1294b746949c85051ef89ffdc5 : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4c340cb24e4044a8af705d1ed1c0a77c : SUCCESS in 12m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7586f7e990bd4134a712f20e23ec4cff : SUCCESS in 12m 08s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d9ee22bbe1304c8ab1c38076b61c8bd0 : FAILURE in 8m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/558a5a93a7b0468897861b11a39fa668 : SUCCESS in 13m 41s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f990d0520a341da83659a75c6538537 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaef55ea9f544be814de0e9008811a4 : SUCCESS in 5m 57s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/72e1615b3f0f48ea820f0f503fcd2c58 : SUCCESS in 23m 23s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18ad4b0d80f94b9282db647430432c8c : SUCCESS in 13m 27s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84d57c7fef0148ae9cf38fd650ab748c : SUCCESS in 14m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad6b746b8f6f42b8b84f6760ecbf92d9 : SUCCESS in 17m 33s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 https://zuul.opendev.org/t/openstack/build/3aeedef07348465bb6abb896cf196895 : SUCCESS in 22m 33s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8485144a193144579441f3117f0de471 : SUCCESS in 16m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fcbd418367b4843ac7827df3b37f74c : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":20},{"id":"bd94d99de947781cae147d75b42ccebb59040d2e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 16:16:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"ea2bd19d5f7d94c70e492ebdb3423fccd24448e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:39:10.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/00d6626d5a674059bc06567553639a60 : SUCCESS in 19m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4af909ccef1d496f8476e081f0b69731 : SUCCESS in 9m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/49e9c1308db64cf99d9ed95e8c761715 : SUCCESS in 12m 39s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/cf0bb83d23cd461d85b810388199a174 : SUCCESS in 15m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d3ebac753d14e99aea3121d9ce8138d : SUCCESS in 13m 22s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/007f43a3720e447fad67ff4e21dc9525 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc0366e7cc24a9baa828c3323092da5 : SUCCESS in 7m 17s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/e079693e7968406dad44d9bd0542b012 : SUCCESS in 16m 40s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8733b724edce4fe5ac97604cf86d1a8c : SUCCESS in 14m 02s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2149bca4fc340cc9fe73bfbb6c13043 : SUCCESS in 14m 30s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fddcc0561c943f59bf228c3ca4f82fb : SUCCESS in 14m 41s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b344872184f4ebfac21976a41bb730f : SUCCESS in 16m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adf79136869746499cd6214323e4bedd : SUCCESS in 18m 28s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f4e80f4b02453d88eeb8c169356155 : SUCCESS in 16m 01s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd88bc11543b423fabc55fb478a2d34f : SUCCESS in 15m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad381a9e25e0473f8a7c68eb664c7bac : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":21},{"id":"ef75a48077f14960ef277abb950a58068825f7ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-23 11:29:34.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"87d8a7199c0e41b87e4c1ed398971999ee28825c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 09:08:47.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"b9d9c48b23c14c94c2c66b7652506679cc5e1958","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 09:08:52.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"4b2ba9b0e857128ced4a684d0e4207ee0112464f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 09:40:08.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dba9e01528ce4ce182239e735df0abcc : SUCCESS in 6m 40s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2a8d9caf6c8049b1b054b7e2bca5e22d : SUCCESS in 9m 10s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8196e974018a48ea9c70c2c69752dd5a : SUCCESS in 10m 50s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0200b313adc8421d944651e4b4862a86 : SUCCESS in 11m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/624ab78c1dc447d6ab96e52281a6ff38 : SUCCESS in 8m 45s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85a3577e66314a8b960844d2b0925ea3 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0283ac38da94488a362bb931f4defa0 : SUCCESS in 5m 14s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/6ab10a61721e45a79004f3e8b6ef2594 : SUCCESS in 16m 05s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e260543bae0a493fb4890f754fb25f37 : SUCCESS in 13m 59s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640f1c7478f44f5f8f168ab1bf849bde : SUCCESS in 15m 31s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/351186b2924a4e2e971c8eba9a55e018 : SUCCESS in 14m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1f41f286aef4cee869d38831f2b0615 : SUCCESS in 15m 24s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da9abbd0f9d34011aeb15d903da915d0 : SUCCESS in 13m 36s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f876d56a241c44aba95e03d7fe231224 : SUCCESS in 14m 45s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5358bb14e0a34d178302a74f7e446cfe : SUCCESS in 14m 16s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f39839b66e5a496ea0e5ebdf22981db1 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":23},{"id":"d704d645dd7b51ed2ebf7321919e08d30d97a9a0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-09-24 13:20:29.000000000","message":"Patch Set 23:\n\nA 2018 patch! Wow! :)\n\nI can\u0027t judge nowadays on this, but thanks for reviving this work if it\u0027s necessary!","accounts_in_message":[],"_revision_number":23},{"id":"e62c35ad40f5e86afc68190f9041e1a7917b7658","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-24 13:28:53.000000000","message":"Patch Set 23: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"36e84b4e5d3360abcc7e841cbb103bf5581fdb39","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 13:45:29.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"57e86cad6df6b0019fb91db1782fdca1b7e31dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 14:17:01.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f5ba14a7072842ada3cb37a452a025ab : SUCCESS in 5m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4deb1935abde42b39774da3a6c5fccd7 : SUCCESS in 9m 45s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/dc41c8d3656342eda4b2e2544685b35d : SUCCESS in 10m 07s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8ef09409f59e48a1b3a996c5feef3e3f : SUCCESS in 12m 32s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12c6af7ead294e299a0dc288ddfcb373 : SUCCESS in 8m 50s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6894a39c619f4a40a1c7c7ab9d2c3e16 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe05b0443fb44d4b6c28042720ca6f7 : SUCCESS in 6m 25s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/571255c4cbbc4ce79d58d6718516549f : SUCCESS in 16m 32s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77c3b1a1d5b5464592466e5c589e9dad : SUCCESS in 14m 50s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9efffdc1a9ae46be9bbf222e9f6967c0 : SUCCESS in 14m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c03b4466fc5e429997ea09529e2d404a : SUCCESS in 14m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57767dfad00344e1b3d7991de8ddf055 : SUCCESS in 15m 35s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57df0887077642f088126b6d3ebe7a26 : SUCCESS in 16m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd15f8e82386457fb32b06419fba5ed5 : SUCCESS in 16m 09s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15b82052aa3149c99cbb6b1902ee1f9d : SUCCESS in 15m 46s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bb6d864661c4152b412f4df5ce3b2b9 : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":24},{"id":"4955d5e19e1ba75c4536cb216f7869b0660f3d5d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 14:27:18.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"01d788768aa7cbb3c2503a4fe04f76e26be536d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:01:01.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3f38a9e5b6f54c10a1a6e0bb39dd9c53 : SUCCESS in 7m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/06a371f33c7c45a48d288b2dbefade02 : SUCCESS in 11m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7301006d0e5c49efaa8701fa6014032e : SUCCESS in 10m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/92d233f35cf8470986fcf48334a4d26b : SUCCESS in 10m 46s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8db12d8204ac4f7d8c6b0ac8964042c5 : SUCCESS in 10m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e724c35b121c4f44b5741bd24abc9588 : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f717ddc1f9b0450db50570af0e2d8454 : SUCCESS in 5m 49s\n- openstack-ansible-lxc-dir-centos-7 https://zuul.opendev.org/t/openstack/build/705d6899a31b48f18dab9e4a2819f150 : SUCCESS in 16m 57s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f80db8623db48299287615b62838d83 : SUCCESS in 19m 53s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e94d919dab945ceafcbd3e56dec023b : SUCCESS in 16m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bf9111cc363429880fb4e3c1f378df3 : SUCCESS in 15m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34cf7a4eefcd4c2191612f0e1e0d2563 : SUCCESS in 23m 06s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6faecec2535f4d19a077ad1593a123db : SUCCESS in 21m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9033cfe36b294190a47e85d98f83a665 : SUCCESS in 21m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eea1ca8bf60e48c8a504b95abb4aa89e : SUCCESS in 18m 22s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb3a1d8431874aacb64baee7e3591234 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":24},{"id":"0267279c5f63efc501e17ed2c590dceeb5647608","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-24 15:59:10.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"dc48f6a484dfe0b288336a6ec0ce3fa9aa72e244","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-06 06:35:08.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"1852381428dc8664351096795dbda82b9c82aa21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 08:08:00.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e833c5b1b5334837b8c508e6f95a902a : SUCCESS in 8m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b593b1b2b0ca4743894745404fae4b5d : SUCCESS in 9m 48s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7efe3fa08d9640bb8c3ac49204160f92 : SUCCESS in 14m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2c9afdfbd0f4c32b1d0ac0669546853 : SUCCESS in 11m 36s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8687f6842604d4b8c57d728f42911fb : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37165398cce4afc9e25fa1ad6b7ae32 : SUCCESS in 6m 22s\n- openstack-ansible-lxc-dir-centos-8 https://zuul.opendev.org/t/openstack/build/0c5aa762ed694201a43b73427a901f55 : SUCCESS in 21m 13s\n- openstack-ansible-lxc-dir-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40fa48da84304881924a3b936bed1f7e : SUCCESS in 16m 56s\n- openstack-ansible-lxc-dir-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aa1bcdfed704eff87b32908250e9022 : SUCCESS in 17m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/372de15ca2784611a0f0358d8e18f214 : SUCCESS in 16m 51s\n- openstack-ansible-lxc-btrfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b92ea23ae07e437397b96d0e0bdaf46c : SUCCESS in 19m 48s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/842fb389435a4aaa89d2e83c9e49196b : SUCCESS in 17m 50s\n- openstack-ansible-lxc-overlayfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/018b150ad6b241a6a604c8054430beb9 : SUCCESS in 17m 16s\n- openstack-ansible-lxc-zfs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbd7eb2eeaa74fd3a11dcf408076f57a : SUCCESS in 17m 58s\n- openstack-ansible-lxc-zfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9186c481be4f436a8a72629457122bbf : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":25},{"id":"1ad5c417aaee483c5f80d97fcba78f7a556a885c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-06 08:44:44.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"43b5d5c862e7a931e3b0eb0aaa22945aa24b4db3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 15:02:04.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"0cd36d3dfa4749cc1bff58ae57d9acf986c208bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:02:41.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"ec4dcf721e74e1d740fc841faed7b2990eddd12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:28:46.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/266a702b7bd048dcb908ed79779c4422 : SUCCESS in 9m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d8cfb11fe29a4d1581b6a9c71a816eeb : SUCCESS in 11m 18s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db230fd311584a5ca4c5ef5782d9636e : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642a8ed145e04a4cb8fbcef0dc3f19f4 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":25},{"id":"1bb4e811a8034972e0c53b6ae16488b861baa2cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:28:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"b52119a0aa2e76d0350be046066fde33d4775872","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 15:31:14.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a768eb18164be5ba5a087c860bb07e : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"cdd6315a8b486d13e13fa7b9783000f32aa22294","revisions":{"5684ab97a8e79a42543c5f2614f91661beeeb61a":{"kind":"REWORK","_number":1,"created":"2018-10-24 15:31:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/1"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:30:22.000000000","tz":60},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613069\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5684ab97a8e79a42543c5f2614f91661beeeb61a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5684ab97a8e79a42543c5f2614f91661beeeb61a"}]},"branch":"refs/heads/master"},"8773e081b4ca12d7cc6757f32ea54eb86cc1dbca":{"kind":"REWORK","_number":2,"created":"2018-10-24 15:57:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/2"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:55:02.000000000","tz":60},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613069\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8773e081b4ca12d7cc6757f32ea54eb86cc1dbca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8773e081b4ca12d7cc6757f32ea54eb86cc1dbca"}]},"branch":"refs/heads/master"},"dd6896a33957f54d2b7d58227b7e6ecf8d3c8fec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-10-24 16:14:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/3"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 16:14:27.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613099\nDepends-On: https://review.openstack.org/613069\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dd6896a33957f54d2b7d58227b7e6ecf8d3c8fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/dd6896a33957f54d2b7d58227b7e6ecf8d3c8fec"}]},"branch":"refs/heads/master"},"009d4f11d76674235d920bfdf9d12c0632c250ff":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2018-10-24 16:18:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/4"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 16:17:59.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613102\nDepends-On: https://review.openstack.org/613099\nDepends-On: https://review.openstack.org/613069\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/009d4f11d76674235d920bfdf9d12c0632c250ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/009d4f11d76674235d920bfdf9d12c0632c250ff"}]},"branch":"refs/heads/master"},"84d8379cc24437b97715f72cfefc3d5bfcaf8009":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2018-10-24 16:23:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/5"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 16:23:29.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613103\nDepends-On: https://review.openstack.org/613102\nDepends-On: https://review.openstack.org/613099\nDepends-On: https://review.openstack.org/613069\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/84d8379cc24437b97715f72cfefc3d5bfcaf8009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/84d8379cc24437b97715f72cfefc3d5bfcaf8009"}]},"branch":"refs/heads/master"},"5ba00e5065691ec6fbe4eb7b75ca58495db0fc01":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2018-10-25 21:34:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/6"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-25 21:34:05.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nDepends-On: https://review.openstack.org/613375\nDepends-On: https://review.openstack.org/613428\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ba00e5065691ec6fbe4eb7b75ca58495db0fc01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5ba00e5065691ec6fbe4eb7b75ca58495db0fc01"}]},"branch":"refs/heads/master"},"cb97e8190fa85b61857162a3d3b1196b34e54916":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2018-11-01 15:26:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/7"}}},"commit":{"parents":[{"commit":"7125cde0f6df1b34fdc4913236cef187ce1b48e0","subject":"Change the default option name to reduce confusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7125cde0f6df1b34fdc4913236cef187ce1b48e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-01 15:26:11.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cb97e8190fa85b61857162a3d3b1196b34e54916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cb97e8190fa85b61857162a3d3b1196b34e54916"}]},"branch":"refs/heads/master"},"435f33344503cc37fa7dea19bc867e59154e7679":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-11-07 12:36:34.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/78/613078/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/8"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse.pretorius@rackspace.co.uk","date":"2018-11-07 12:36:34.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/435f33344503cc37fa7dea19bc867e59154e7679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/435f33344503cc37fa7dea19bc867e59154e7679"}]},"branch":"refs/heads/master"},"93fa27f49e86707660634b8fc4b169bcb04cbe5f":{"kind":"REWORK","_number":9,"created":"2018-11-20 07:06:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/9"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 07:06:29.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/93fa27f49e86707660634b8fc4b169bcb04cbe5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/93fa27f49e86707660634b8fc4b169bcb04cbe5f"}]},"branch":"refs/heads/master"},"a5c400ab98652aacf4cd21ab1a70cdc6da295bdd":{"kind":"REWORK","_number":10,"created":"2018-11-20 08:55:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/10"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 08:55:01.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5c400ab98652aacf4cd21ab1a70cdc6da295bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a5c400ab98652aacf4cd21ab1a70cdc6da295bdd"}]},"branch":"refs/heads/master"},"c8277a46db8b8b32b8ca4cca77808c2a5fee0f6a":{"kind":"REWORK","_number":11,"created":"2018-11-20 09:23:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/11"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 09:23:12.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8277a46db8b8b32b8ca4cca77808c2a5fee0f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c8277a46db8b8b32b8ca4cca77808c2a5fee0f6a"}]},"branch":"refs/heads/master"},"060762f116cd4468a315e16554b53a1abcdb39df":{"kind":"REWORK","_number":12,"created":"2018-11-20 09:54:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/12"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 09:54:24.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/060762f116cd4468a315e16554b53a1abcdb39df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/060762f116cd4468a315e16554b53a1abcdb39df"}]},"branch":"refs/heads/master"},"1c1b40f16d343223d01e05f4f62cdedcc6205ef0":{"kind":"REWORK","_number":13,"created":"2018-11-20 10:21:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/13"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 10:20:50.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, with the\nexpection of \u0027sudo\u0027 which is used extensively via ansible become_user\nin the \u0027setup-openstack\u0027 set of playbooks.\n\nFurther packages should be installed by the ansible roles which further\ncustomise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1c1b40f16d343223d01e05f4f62cdedcc6205ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1c1b40f16d343223d01e05f4f62cdedcc6205ef0"}]},"branch":"refs/heads/master"},"84d77143cf16f47c7d95af6417afec34e2dabb9f":{"kind":"REWORK","_number":14,"created":"2018-11-20 11:17:28.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/14"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 11:17:14.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/84d77143cf16f47c7d95af6417afec34e2dabb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/84d77143cf16f47c7d95af6417afec34e2dabb9f"}]},"branch":"refs/heads/master"},"6479bd4e32ec792817886e1d440dc8512140f973":{"kind":"REWORK","_number":15,"created":"2018-11-20 12:03:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/15"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 12:03:23.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6479bd4e32ec792817886e1d440dc8512140f973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6479bd4e32ec792817886e1d440dc8512140f973"}]},"branch":"refs/heads/master"},"1f63e34071e711d2bcd65220eccab9db708b8bff":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2018-11-20 17:27:07.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/16"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-20 17:27:00.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1f63e34071e711d2bcd65220eccab9db708b8bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1f63e34071e711d2bcd65220eccab9db708b8bff"}]},"branch":"refs/heads/master"},"818086c01f45b9d9870ae82c32efe0c59ab93a43":{"kind":"REWORK","_number":17,"created":"2018-11-21 12:04:28.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/17"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-11-21 12:04:25.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/818086c01f45b9d9870ae82c32efe0c59ab93a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/818086c01f45b9d9870ae82c32efe0c59ab93a43"}]},"branch":"refs/heads/master"},"8efc6b2310ebf4a9084d509e93eda4fb4382b26b":{"kind":"REWORK","_number":18,"created":"2019-01-03 09:32:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/18"}}},"commit":{"parents":[{"commit":"0766e893ec1091a1837b858701483a9381377455","subject":"Merge \"Correct spelling mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0766e893ec1091a1837b858701483a9381377455"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-01-03 09:31:57.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8efc6b2310ebf4a9084d509e93eda4fb4382b26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8efc6b2310ebf4a9084d509e93eda4fb4382b26b"}]},"branch":"refs/heads/master"},"f7db1ad68c15b859e16f846e5e3986fd526e183a":{"kind":"REWORK","_number":19,"created":"2019-01-07 12:44:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/19"}}},"commit":{"parents":[{"commit":"0766e893ec1091a1837b858701483a9381377455","subject":"Merge \"Correct spelling mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/0766e893ec1091a1837b858701483a9381377455"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-01-07 12:44:30.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7db1ad68c15b859e16f846e5e3986fd526e183a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f7db1ad68c15b859e16f846e5e3986fd526e183a"}]},"branch":"refs/heads/master"},"6b3e6eb713f51a0cbedc61630930619b566e37d0":{"kind":"REWORK","_number":20,"created":"2020-06-03 07:38:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/20"}}},"commit":{"parents":[{"commit":"7e25305c7585845007d17299ec0bd4583b73e41c","subject":"Fix ubuntu focal to use correct image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/7e25305c7585845007d17299ec0bd4583b73e41c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-03 07:38:23.000000000","tz":60},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b3e6eb713f51a0cbedc61630930619b566e37d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6b3e6eb713f51a0cbedc61630930619b566e37d0"}]},"branch":"refs/heads/master"},"f0a83553eb6e5c3ac2a6137d0248fa39a82c1366":{"kind":"REWORK","_number":21,"created":"2020-07-06 16:16:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/21"}}},"commit":{"parents":[{"commit":"eaae389a4bf9708eb32bb820429da6343edeb7d5","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eaae389a4bf9708eb32bb820429da6343edeb7d5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-06 16:15:40.000000000","tz":60},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f0a83553eb6e5c3ac2a6137d0248fa39a82c1366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/f0a83553eb6e5c3ac2a6137d0248fa39a82c1366"}]},"branch":"refs/heads/master"},"ab6f6324ae79a9030d2f6439a932f030abaf149e":{"kind":"REWORK","_number":22,"created":"2020-09-24 09:08:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/22"}}},"commit":{"parents":[{"commit":"eaae389a4bf9708eb32bb820429da6343edeb7d5","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/eaae389a4bf9708eb32bb820429da6343edeb7d5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-24 09:08:43.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ab6f6324ae79a9030d2f6439a932f030abaf149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ab6f6324ae79a9030d2f6439a932f030abaf149e"}]},"branch":"refs/heads/master"},"479e36ce80fc55aeae3d0681451766b61a46e9da":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2020-09-24 09:08:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/23"}}},"commit":{"parents":[{"commit":"a4f917b909fd3fa0fa5d1f40b6681f56cfa76820","subject":"Wait for aria2c to finish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a4f917b909fd3fa0fa5d1f40b6681f56cfa76820"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-24 09:08:52.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/479e36ce80fc55aeae3d0681451766b61a46e9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/479e36ce80fc55aeae3d0681451766b61a46e9da"}]},"branch":"refs/heads/master"},"6f7132342a7b37f9d4059fcf5335818ee9714142":{"kind":"REWORK","_number":24,"created":"2020-09-24 13:45:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/24"}}},"commit":{"parents":[{"commit":"a4f917b909fd3fa0fa5d1f40b6681f56cfa76820","subject":"Wait for aria2c to finish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a4f917b909fd3fa0fa5d1f40b6681f56cfa76820"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-24 13:45:22.000000000","tz":0},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6f7132342a7b37f9d4059fcf5335818ee9714142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6f7132342a7b37f9d4059fcf5335818ee9714142"}]},"branch":"refs/heads/master"},"cdd6315a8b486d13e13fa7b9783000f32aa22294":{"kind":"REWORK","_number":25,"created":"2020-10-06 06:35:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/613078/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/613078/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/613078/25"}}},"commit":{"parents":[{"commit":"01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/01eecae8d3e728f44ac7cbadf23f2f5f2c3c07f3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2018-10-24 15:26:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-06 06:34:52.000000000","tz":60},"subject":"Minimise distro packages installed into the lxc image","message":"Minimise distro packages installed into the lxc image\n\nThis patch removes all packages from the lxc image cache preparation\nwhich are not required to run the \u0027setup-hosts\u0027 playbook, or pass the\nlxc_hosts role tests.\n\nAdditional packages should be installed by the ansible roles which\nfurther customise the containers.\n\nDepends-On: https://review.openstack.org/619040\nChange-Id: I9bdc698e4c712a6fb3d7a3b76413b478e1e7e91a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cdd6315a8b486d13e13fa7b9783000f32aa22294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cdd6315a8b486d13e13fa7b9783000f32aa22294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
