)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~614127","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ieb5004737ebbf565bc26b39b63cdb289aaf1e6f3","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"Ieb5004737ebbf565bc26b39b63cdb289aaf1e6f3","subject":"missing document start \"---\"","status":"MERGED","created":"2018-10-30 07:59:40.000000000","updated":"2019-04-24 19:04:51.000000000","submitted":"2019-04-24 19:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614127-1556132691319-6db76d6a","meta_rev_id":"e7b00ebcfc9d38a809eea98b8b8ad4a6a7315e7c","_number":614127,"virtual_id_number":614127,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 19:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:35:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69057229ad76422b85c66be3999e54852e55fe78","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-30 07:59:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db86cda96be77ef50003da7d34f1e744997f848c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:26:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/614127/1/check/openstack-ansible-linters/967f433/ : SUCCESS in 6m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/27/614127/1/check/openstack-ansible-functional-centos-7/ffdffef/ : SUCCESS in 13m 06s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/27/614127/1/check/openstack-ansible-functional-opensuse-150/88888ed/ : SUCCESS in 10m 38s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/27/614127/1/check/openstack-ansible-functional-opensuse-423/327ec46/ : SUCCESS in 9m 38s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-functional-ubuntu-bionic/3494517/ : SUCCESS in 8m 46s\n- openstack-tox-docs http://logs.openstack.org/27/614127/1/check/openstack-tox-docs/6c93a26/html/ : SUCCESS in 3m 09s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-dir-centos-7/7094159/ : SUCCESS in 16m 45s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-dir-opensuse-423/2e26e14/ : SUCCESS in 13m 02s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-dir-opensuse-150/64b6895/ : SUCCESS in 15m 50s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/1d1776e/ : SUCCESS in 12m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/6deb8b6/ : FAILURE in 12m 25s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-btrfs-centos-7/f635752/ : SUCCESS in 13m 32s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-btrfs-opensuse-423/57b5255/ : SUCCESS in 13m 02s\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-btrfs-opensuse-150/86e80c2/ : SUCCESS in 18m 20s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/b536de5/ : SUCCESS in 14m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-machinectl-centos-7/51ed1c2/ : SUCCESS in 16m 50s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-machinectl-opensuse-423/62174e3/ : SUCCESS in 12m 14s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-machinectl-opensuse-150/7ff59e3/ : SUCCESS in 16m 55s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/2bfce08/ : SUCCESS in 13m 55s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-overlayfs-centos-7/64bc5c2/ : SUCCESS in 16m 45s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-overlayfs-opensuse-423/2d39e5f/ : SUCCESS in 17m 06s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/d3bcdf9/ : SUCCESS in 18m 43s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/27/614127/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/b998965/ : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ffc2fc326e79c8fac9f929c0ea3944d8e01b9bfe","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:35:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cde462118355989fd5726a9d2ffd6c580026a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:35:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9801123d1001f51915e561e25f275512c27096e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:04:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e7b00ebcfc9d38a809eea98b8b8ad4a6a7315e7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:04:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/27/614127/1/gate/openstack-ansible-linters/9b269eb/ : SUCCESS in 5m 07s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/27/614127/1/gate/openstack-ansible-functional-debian-stable/d21a88b/ : SUCCESS in 7m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/27/614127/1/gate/openstack-ansible-functional-centos-7/a4f57fc/ : SUCCESS in 16m 01s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/27/614127/1/gate/openstack-ansible-functional-ubuntu-bionic/38014aa/ : SUCCESS in 8m 07s\n- openstack-tox-docs http://logs.openstack.org/27/614127/1/gate/openstack-tox-docs/830670d/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc032bc69f6fd2f34439926f763ccde0ca22580f","revisions":{"bc032bc69f6fd2f34439926f763ccde0ca22580f":{"kind":"REWORK","_number":1,"created":"2018-10-30 07:59:40.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/27/614127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/27/614127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/614127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/614127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/614127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/27/614127/1"}}},"commit":{"parents":[{"commit":"3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd","subject":"apparmor: Allow cgroup v2 mounts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3cdfd8c5319b32ab155a4e4cbfd4c8bf805c5cdd"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:59:23.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:59:38.000000000","tz":0},"subject":"missing document start \"---\"","message":"missing document start \"---\"\n\nChange-Id: Ieb5004737ebbf565bc26b39b63cdb289aaf1e6f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bc032bc69f6fd2f34439926f763ccde0ca22580f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/bc032bc69f6fd2f34439926f763ccde0ca22580f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
