)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~652120","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fqueens~I4fa838706f1163fd68ff68258bfc66cbf13bad94","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/queens","topic":"dnsmasq_pkill_queens","hashtags":[],"change_id":"I4fa838706f1163fd68ff68258bfc66cbf13bad94","subject":"Use pkill for lxc-dnsmasq systemd unit file","status":"MERGED","created":"2019-04-12 17:49:49.000000000","updated":"2019-06-05 19:53:39.000000000","submitted":"2019-06-05 19:53:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652120-1559764419906-4b762db2","meta_rev_id":"0c8a7a1e60b2d5a9e5e4d557b2ce939ce7b80fb6","_number":652120,"virtual_id_number":652120,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-05 19:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2019-05-31 02:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 19:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 19:25:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 02:39:32.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-05 19:25:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-05 19:53:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4efa420f4a2c145e25a3ba872fef98525f2ca22d","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-12 17:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"586c1d21ccb31a6b95d8909643b27ff518b7b949","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 18:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/1/check/build-openstack-sphinx-docs/27c2ade/html/ : SUCCESS in 2m 44s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/1/check/openstack-ansible-linters/13d401b/ : SUCCESS in 4m 40s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-centos-7/8d68a4c/ : SUCCESS in 9m 09s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-opensuse-423/db204a3/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-ubuntu-xenial/459007e/ : SUCCESS in 7m 09s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-centos-7/2a8c13c/ : SUCCESS in 20m 31s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-opensuse-423/fef3eec/ : FAILURE in 8m 20s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/5a9f211/ : SUCCESS in 10m 44s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/9cb6a0b/ : SUCCESS in 11m 34s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/c8d2b10/ : SUCCESS in 13m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/a547ccd/ : SUCCESS in 11m 51s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/573114a/ : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9b1bcd89d23808894f8368fce674ccf0334c60bc","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-12 18:31:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d5e733e979fa8704c3ff7119f53a603cb7715431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 18:59:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/1/check/build-openstack-sphinx-docs/0ba6854/html/ : SUCCESS in 3m 03s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/1/check/openstack-ansible-linters/cc7fb99/ : SUCCESS in 4m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-centos-7/453219e/ : SUCCESS in 9m 45s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-opensuse-423/18a31b6/ : SUCCESS in 6m 38s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-ubuntu-xenial/a4b775b/ : SUCCESS in 6m 35s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-centos-7/9402b41/ : SUCCESS in 18m 46s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-opensuse-423/f15043d/ : FAILURE in 7m 46s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/8cba4d6/ : SUCCESS in 9m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/43348e1/ : SUCCESS in 10m 40s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/bd6ba49/ : SUCCESS in 10m 47s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/0ac281c/ : SUCCESS in 9m 36s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/681ba88/ : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aada87ddc8f3cf4171be208b6567c0b7b444815c","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-15 18:53:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d08285d98ec38a97d1c86f9cacbc2ab6fa926c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 19:17:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/1/check/build-openstack-sphinx-docs/2035c62/html/ : SUCCESS in 3m 11s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/1/check/openstack-ansible-linters/16cd82e/ : SUCCESS in 6m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-centos-7/87e70f6/ : SUCCESS in 8m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-opensuse-423/a9e379e/ : SUCCESS in 8m 33s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-ubuntu-xenial/b5278f8/ : SUCCESS in 6m 41s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-centos-7/75f2faf/ : SUCCESS in 19m 26s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-opensuse-423/6188c2c/ : FAILURE in 10m 02s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/dcdd8ee/ : SUCCESS in 11m 54s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/f17f7f4/ : SUCCESS in 13m 11s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/70684d4/ : SUCCESS in 12m 05s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/b966206/ : SUCCESS in 11m 46s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/b2c9ea7/ : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3bb76f1f403ebd9351bfe87e05117e8b2357c530","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-23 21:25:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db95227c73cacbe8a8e7e19c12a53789a367695a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:49:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/1/check/build-openstack-sphinx-docs/6c4a1e6/html/ : SUCCESS in 3m 09s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/1/check/openstack-ansible-linters/74c6a88/ : FAILURE in 4m 50s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-centos-7/78fb501/ : FAILURE in 5m 12s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-opensuse-423/13e9836/ : FAILURE in 4m 23s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-functional-ubuntu-xenial/9c5b3be/ : FAILURE in 3m 18s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-centos-7/740b92e/ : FAILURE in 5m 05s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-opensuse-423/11c5b57/ : FAILURE in 4m 35s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/00b0310/ : FAILURE in 3m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/cfb4c1f/ : FAILURE in 4m 07s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/eab5e3a/ : FAILURE in 4m 12s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/e295ae1/ : FAILURE in 4m 26s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/b8aeca1/ : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"103a72cfaa6c070acb1158bcb8e4b07f863d6a4e","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-23 22:26:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9d84dec3701ec23b133424dd1122f4812628d80c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:57:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/7c5bfde/html/ : SUCCESS in 2m 33s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/609432a/ : FAILURE in 4m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/8feda2e/ : FAILURE in 5m 35s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/eb64ee2/ : FAILURE in 5m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/582e097/ : FAILURE in 4m 55s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/3e12b8e/ : FAILURE in 5m 16s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/fcc4f87/ : FAILURE in 4m 37s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/d4d1f6e/ : FAILURE in 4m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/93efc06/ : FAILURE in 4m 48s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/41df264/ : FAILURE in 4m 56s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/083d842/ : POST_FAILURE in 5m 56s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/6c93fd7/ : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ecf4d90b355344c67472a15978b50d4e77f64ca4","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-25 14:31:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"316716cfa2601c95f2cc66b3ef6d5682b8960645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:47:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/96f2446/html/ : SUCCESS in 2m 56s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/e129bb6/ : FAILURE in 4m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/e23273e/ : FAILURE in 5m 13s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/ae35885/ : FAILURE in 4m 36s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/3aebe51/ : FAILURE in 4m 22s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/cd4b0e5/ : FAILURE in 8m 02s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/8afa905/ : FAILURE in 4m 22s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/c5eaee2/ : FAILURE in 6m 47s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/272c38c/ : FAILURE in 4m 23s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/52bde6e/ : FAILURE in 4m 19s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/606bdf0/ : FAILURE in 4m 19s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/9ec509f/ : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"cd2373e8096b2fdb50535876538679ba69221219","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-10 14:50:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"78d7a5a9ccb72b6b04ae90853d43d733234dff3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:04:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/d54de01/html/ : SUCCESS in 3m 18s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/ee74070/ : FAILURE in 4m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/65a8c19/ : FAILURE in 6m 00s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/93f8c5a/ : FAILURE in 4m 34s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/8c58197/ : FAILURE in 6m 09s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/e737803/ : FAILURE in 4m 38s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/4085048/ : FAILURE in 4m 05s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/1dfa785/ : FAILURE in 7m 32s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/a84ace8/ : FAILURE in 3m 27s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/48f72f0/ : FAILURE in 7m 13s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/fe1d7ef/ : FAILURE in 5m 50s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/db4b504/ : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c39c942ba86f530fb29f40afbcb821215ea6b6bd","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-21 18:48:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9917eb706e8e8eb1da08d15d50bfe9742a7a67d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:29:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/c0abb97/html/ : SUCCESS in 2m 51s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/ab51e00/ : SUCCESS in 5m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/2e5a69b/ : SUCCESS in 10m 57s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/8885c98/ : SUCCESS in 8m 37s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/6ff1657/ : SUCCESS in 7m 20s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/81d2ba2/ : SUCCESS in 17m 45s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/1b3aad1/ : FAILURE in 9m 22s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/d8992eb/ : SUCCESS in 13m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/e0d9526/ : SUCCESS in 9m 56s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/24f825b/ : SUCCESS in 12m 44s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/5b5615f/ : SUCCESS in 9m 27s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/59e11a6/ : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"29cdf658cdeb8e80be836575c26c725a493b8be4","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-29 14:17:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b77393474b54f20ce56d8d29464579db2ab8c9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 14:54:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/a4eabbd/html/ : SUCCESS in 3m 21s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/a23b5a7/ : SUCCESS in 6m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/c7b332c/ : SUCCESS in 10m 06s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/8d87adc/ : SUCCESS in 7m 25s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/0b5e34a/ : SUCCESS in 8m 46s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/f7f2d1e/ : SUCCESS in 18m 01s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/8174a56/ : FAILURE in 7m 57s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/6519fbe/ : SUCCESS in 10m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/ae1af0f/ : SUCCESS in 11m 01s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/8bd1894/ : SUCCESS in 10m 59s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/2a19af7/ : SUCCESS in 17m 49s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/7661b56/ : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9a16d1ceba7a9762910a51623595f2c375d5fd13","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-30 14:11:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cecbd9cc250994c659f534dd8fcd9fb7bb166567","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 14:42:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/2/check/build-openstack-sphinx-docs/ce82848/html/ : SUCCESS in 3m 04s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/2/check/openstack-ansible-linters/2c8edde/ : SUCCESS in 4m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-centos-7/6dfd4a3/ : SUCCESS in 9m 36s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-opensuse-423/fd18a75/ : SUCCESS in 7m 26s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-functional-ubuntu-xenial/7dfa5ab/ : SUCCESS in 7m 00s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-centos-7/ec51dd8/ : SUCCESS in 22m 12s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-opensuse-423/87bfd5d/ : FAILURE in 7m 18s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/1e82c41/ : SUCCESS in 11m 52s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/9601fcc/ : SUCCESS in 11m 20s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/036339a/ : SUCCESS in 12m 11s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/b7413e2/ : SUCCESS in 10m 22s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/c7ca145/ : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"96f2bafee88decb903d1ca06256744ea9f7b0f16","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-30 14:57:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e92e900c417efd6eb84aeaeff1f681fcebeedbdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:27:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/3/check/build-openstack-sphinx-docs/993e161/html/ : SUCCESS in 3m 24s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/3/check/openstack-ansible-linters/df3b9ad/ : SUCCESS in 5m 19s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/3/check/openstack-ansible-functional-centos-7/d43362c/ : SUCCESS in 10m 02s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/3/check/openstack-ansible-functional-opensuse-423/cb0edf9/ : SUCCESS in 7m 28s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-functional-ubuntu-xenial/42cf293/ : SUCCESS in 7m 34s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-dir-centos-7/bdd7e3c/ : SUCCESS in 22m 12s\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-dir-ubuntu-xenial/fcd8a4f/ : SUCCESS in 18m 18s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/88988d7/ : SUCCESS in 12m 55s\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/d896adf/ : SUCCESS in 25m 23s\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/2e8b5ba/ : SUCCESS in 11m 07s\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/20/652120/3/check/openstack-ansible-lxc-zfs-ubuntu-xenial/efa988d/ : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9f2f7ea4189aa03879e3b76319fa9f67e05c3a9b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-31 02:39:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"244fd167dcca010ab6f19fcc865f16310be81085","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-05 19:25:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9ec23b5606daccf2838ce18234217d73ec26dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:26:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85d697d9bcff3e38802b23d075fda77661e2a356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:53:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/652120/3/gate/build-openstack-sphinx-docs/3866240/html/ : SUCCESS in 3m 17s\n- openstack-ansible-linters http://logs.openstack.org/20/652120/3/gate/openstack-ansible-linters/b55bfa2/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/652120/3/gate/openstack-ansible-functional-centos-7/98a20db/ : SUCCESS in 9m 20s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/20/652120/3/gate/openstack-ansible-functional-opensuse-423/316e3c0/ : SUCCESS in 7m 46s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/20/652120/3/gate/openstack-ansible-functional-ubuntu-xenial/57e078b/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":3},{"id":"a43483a60cd1f97edd2d4e2ff25ad628d656ba59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:53:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cffdaf8a795f00fea686ee2debff50675539ff40","revisions":{"b089421f759d2999d420191a3d3bd07997165271":{"kind":"REWORK","_number":1,"created":"2019-04-12 17:49:49.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/20/652120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/652120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/1"}}},"commit":{"parents":[{"commit":"696ad08c7133656fda5c76a1d586cf10c5b499ab","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/696ad08c7133656fda5c76a1d586cf10c5b499ab"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-10 17:37:06.000000000","tz":-300},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-12 17:49:20.000000000","tz":-300},"subject":"Use pkill for lxc-dnsmasq systemd unit file","message":"Use pkill for lxc-dnsmasq systemd unit file\n\nPreviously psmisc was not installed by default, causing\nthe killall in the lxc-dnsmasq systemd unit file to\nsilently fail if killall wasn\u0027t found.  This prevented\nrunning instances of dnsmasq to be torn down when\nrestarting the service.\n\nThis would exhibit as a problem during upgrades as\nthe older dnsmasq service would be running and the\nnew service would be unable to take over the port.\nThis would cause lxc containers to lose connectivity.\n\nThis commit switches to using pkill and ensures procps\nis always installed by default.\n\nChange-Id: I4fa838706f1163fd68ff68258bfc66cbf13bad94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b089421f759d2999d420191a3d3bd07997165271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b089421f759d2999d420191a3d3bd07997165271"}]},"branch":"refs/heads/stable/queens"},"9af7487f75dd6e0cadae827cfb719b5d4c4296f8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-23 22:26:16.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/20/652120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/652120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/2"}}},"commit":{"parents":[{"commit":"294f21c6dc5b5ac4eb1f8a9f5dae37326b2b0f48","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/294f21c6dc5b5ac4eb1f8a9f5dae37326b2b0f48"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-10 17:37:06.000000000","tz":-300},"committer":{"name":"Antony Messerli","email":"antony@mes.ser.li","date":"2019-04-23 22:26:16.000000000","tz":0},"subject":"Use pkill for lxc-dnsmasq systemd unit file","message":"Use pkill for lxc-dnsmasq systemd unit file\n\nPreviously psmisc was not installed by default, causing\nthe killall in the lxc-dnsmasq systemd unit file to\nsilently fail if killall wasn\u0027t found.  This prevented\nrunning instances of dnsmasq to be torn down when\nrestarting the service.\n\nThis would exhibit as a problem during upgrades as\nthe older dnsmasq service would be running and the\nnew service would be unable to take over the port.\nThis would cause lxc containers to lose connectivity.\n\nThis commit switches to using pkill and ensures procps\nis always installed by default.\n\nChange-Id: I4fa838706f1163fd68ff68258bfc66cbf13bad94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9af7487f75dd6e0cadae827cfb719b5d4c4296f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/9af7487f75dd6e0cadae827cfb719b5d4c4296f8"}]},"branch":"refs/heads/stable/queens"},"cffdaf8a795f00fea686ee2debff50675539ff40":{"kind":"REWORK","_number":3,"created":"2019-05-30 14:57:28.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/20/652120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/652120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/652120/3"}}},"commit":{"parents":[{"commit":"696ad08c7133656fda5c76a1d586cf10c5b499ab","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/696ad08c7133656fda5c76a1d586cf10c5b499ab"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-10 17:37:06.000000000","tz":-300},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-05-30 14:56:32.000000000","tz":-300},"subject":"Use pkill for lxc-dnsmasq systemd unit file","message":"Use pkill for lxc-dnsmasq systemd unit file\n\nPreviously psmisc was not installed by default, causing\nthe killall in the lxc-dnsmasq systemd unit file to\nsilently fail if killall wasn\u0027t found.  This prevented\nrunning instances of dnsmasq to be torn down when\nrestarting the service.\n\nThis would exhibit as a problem during upgrades as\nthe older dnsmasq service would be running and the\nnew service would be unable to take over the port.\nThis would cause lxc containers to lose connectivity.\n\nThis commit switches to using pkill and ensures procps\nis always installed by default.\n\nAlso removes openSUSE 42.3 lxc job since 42.3 is EOL soon\nand the jobs are broken.\n\nChange-Id: I4fa838706f1163fd68ff68258bfc66cbf13bad94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cffdaf8a795f00fea686ee2debff50675539ff40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/cffdaf8a795f00fea686ee2debff50675539ff40"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
