)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~652123","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fpike~I4fa838706f1163fd68ff68258bfc66cbf13bad94","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/pike","topic":"dnsmasq_pkill_pike","hashtags":[],"change_id":"I4fa838706f1163fd68ff68258bfc66cbf13bad94","subject":"Use pkill for lxc-dnsmasq systemd unit file","status":"MERGED","created":"2019-04-12 17:55:23.000000000","updated":"2019-06-06 02:30:26.000000000","submitted":"2019-06-06 02:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652123-1559788226263-5574ccc7","meta_rev_id":"c0a65e731f22c1eae7412412e14481d93264b716","_number":652123,"virtual_id_number":652123,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-05 18:10:43.000000000","_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-06 02:30:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2019-06-06 01:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 20:03:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":1,"date":"2019-06-06 01:54:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 20:03:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-06 01:54:56.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-06 02:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3d6934d23b18e71b4e12455b6c948d3c84ed655","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-04-12 17:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3ded7a8045b884567cdf6bbc0bd4bad263271ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 18:10:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/23/652123/1/check/build-openstack-sphinx-docs/03e5000/html/ : SUCCESS in 3m 15s\n- openstack-ansible-linters http://logs.openstack.org/23/652123/1/check/openstack-ansible-linters/071de2d/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-centos-7/fe8a3b0/ : SUCCESS in 11m 21s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-opensuse-423/2b0503b/ : SUCCESS in 8m 36s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-ubuntu-xenial/95330f0/ : SUCCESS in 8m 55s\n- openstack-ansible-lxc-dir http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-dir/a7dd281/ : SUCCESS in 13m 26s (non-voting)\n- openstack-ansible-lxc-btrfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-btrfs/3930716/ : SUCCESS in 12m 55s (non-voting)\n- openstack-ansible-lxc-machinectl http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-machinectl/47def57/ : SUCCESS in 12m 54s (non-voting)\n- openstack-ansible-lxc-overlayfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-overlayfs/c162752/ : SUCCESS in 13m 09s (non-voting)\n- openstack-ansible-lxc-zfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-zfs/4d19e24/ : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6724ae27ae554c98ce122fdbaa286ca720c02781","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-10 14:50:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"768ddde1d79d066d2b5068d988b44e39dbef4271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:02:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/23/652123/1/check/build-openstack-sphinx-docs/69d9f6c/html/ : SUCCESS in 3m 21s\n- openstack-ansible-linters http://logs.openstack.org/23/652123/1/check/openstack-ansible-linters/ccc739a/ : FAILURE in 4m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-centos-7/2b62308/ : FAILURE in 4m 46s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-opensuse-423/18c07e5/ : FAILURE in 4m 11s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-ubuntu-xenial/a1b8283/ : FAILURE in 3m 21s\n- openstack-ansible-lxc-dir http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-dir/afaa82f/ : FAILURE in 3m 26s (non-voting)\n- openstack-ansible-lxc-btrfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-btrfs/045825f/ : FAILURE in 3m 25s (non-voting)\n- openstack-ansible-lxc-machinectl http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-machinectl/177f216/ : FAILURE in 3m 48s (non-voting)\n- openstack-ansible-lxc-overlayfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-overlayfs/d4f3e07/ : FAILURE in 4m 38s (non-voting)\n- openstack-ansible-lxc-zfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-zfs/60d7c5b/ : FAILURE in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1966ed10c850e4831a0bd0a965293eaa03dd6ef1","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-29 14:17:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94e84025a31d46d32310d54053564b445d78a089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 14:52:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/23/652123/1/check/build-openstack-sphinx-docs/29ae68d/html/ : SUCCESS in 3m 41s\n- openstack-ansible-linters http://logs.openstack.org/23/652123/1/check/openstack-ansible-linters/31ab82f/ : FAILURE in 4m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-centos-7/511b5f5/ : FAILURE in 6m 15s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-opensuse-423/54c92b1/ : FAILURE in 5m 15s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-ubuntu-xenial/cb75183/ : FAILURE in 3m 48s\n- openstack-ansible-lxc-dir http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-dir/9a9f59f/ : FAILURE in 4m 15s (non-voting)\n- openstack-ansible-lxc-btrfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-btrfs/c476be3/ : FAILURE in 3m 32s (non-voting)\n- openstack-ansible-lxc-machinectl http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-machinectl/b40fddf/ : FAILURE in 6m 58s (non-voting)\n- openstack-ansible-lxc-overlayfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-overlayfs/9a5af50/ : FAILURE in 3m 51s (non-voting)\n- openstack-ansible-lxc-zfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-zfs/0ff1dfa/ : FAILURE in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cbf2fdc967f83816ac4c40aef8cd1ad04e2d644","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-06-05 18:10:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff2f2568752453ac4ae6cf705931111b6e8356df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 18:53:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/23/652123/1/check/build-openstack-sphinx-docs/cc75905/html/ : SUCCESS in 3m 05s\n- openstack-ansible-linters http://logs.openstack.org/23/652123/1/check/openstack-ansible-linters/a9df2ba/ : SUCCESS in 4m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-centos-7/da866ff/ : SUCCESS in 10m 28s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-opensuse-423/5150bce/ : SUCCESS in 7m 56s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/652123/1/check/openstack-ansible-functional-ubuntu-xenial/6b2a9b8/ : SUCCESS in 8m 41s\n- openstack-ansible-lxc-dir http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-dir/b5ca4d8/ : SUCCESS in 12m 16s (non-voting)\n- openstack-ansible-lxc-btrfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-btrfs/5817f19/ : SUCCESS in 11m 44s (non-voting)\n- openstack-ansible-lxc-machinectl http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-machinectl/cb879f6/ : SUCCESS in 11m 21s (non-voting)\n- openstack-ansible-lxc-overlayfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-overlayfs/5fb9706/ : SUCCESS in 11m 35s (non-voting)\n- openstack-ansible-lxc-zfs http://logs.openstack.org/23/652123/1/check/openstack-ansible-lxc-zfs/b206734/ : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9744e944f354b26ef92cb6a0c56beca6be6f58fa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-05 20:03:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2634452b1c62e587967b4f25c7bd8dc4a4a94314","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-06 01:54:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e63c16a31cfda839acd6752aee5f2cb819cd597","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:55:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb9f394ca7cd007e0f3b2704d7b90a6faa5f641b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 02:30:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"806ab34f22f66ff3ec05659e0313900a05fecc58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 02:30:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/23/652123/1/gate/build-openstack-sphinx-docs/9befbbe/html/ : SUCCESS in 3m 17s\n- openstack-ansible-linters http://logs.openstack.org/23/652123/1/gate/openstack-ansible-linters/07da64a/ : SUCCESS in 5m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/652123/1/gate/openstack-ansible-functional-centos-7/b543699/ : SUCCESS in 11m 23s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/652123/1/gate/openstack-ansible-functional-opensuse-423/04077ba/ : SUCCESS in 7m 55s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/652123/1/gate/openstack-ansible-functional-ubuntu-xenial/1f84f36/ : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"107c55ca1c433be395392caf9f513f547f88f257","revisions":{"107c55ca1c433be395392caf9f513f547f88f257":{"kind":"REWORK","_number":1,"created":"2019-04-12 17:55:23.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/23/652123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/23/652123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/652123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/652123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/652123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/23/652123/1"}}},"commit":{"parents":[{"commit":"35867502a1be652206e98205e3725348f44dcc2b","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/35867502a1be652206e98205e3725348f44dcc2b"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-10 17:37:06.000000000","tz":-300},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-04-12 17:55:02.000000000","tz":-300},"subject":"Use pkill for lxc-dnsmasq systemd unit file","message":"Use pkill for lxc-dnsmasq systemd unit file\n\nPreviously psmisc was not installed by default, causing\nthe killall in the lxc-dnsmasq systemd unit file to\nsilently fail if killall wasn\u0027t found.  This prevented\nrunning instances of dnsmasq to be torn down when\nrestarting the service.\n\nThis would exhibit as a problem during upgrades as\nthe older dnsmasq service would be running and the\nnew service would be unable to take over the port.\nThis would cause lxc containers to lose connectivity.\n\nThis commit switches to using pkill and ensures procps\nis always installed by default.\n\nChange-Id: I4fa838706f1163fd68ff68258bfc66cbf13bad94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/107c55ca1c433be395392caf9f513f547f88f257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/107c55ca1c433be395392caf9f513f547f88f257"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
