)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~652268","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I45647062c5766f1f9d4dd098f107e91caf52a3d7","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"fix_150","hashtags":[],"change_id":"I45647062c5766f1f9d4dd098f107e91caf52a3d7","subject":"Fix image fetching process for leap 15","status":"MERGED","created":"2019-04-14 09:30:57.000000000","updated":"2019-04-23 20:35:02.000000000","submitted":"2019-04-23 20:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652268-1556051702317-76a6aaaf","meta_rev_id":"5d2c6302880f1eec7afff8e22ffb869295b56dba","_number":652268,"virtual_id_number":652268,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"date":"2019-04-23 12:58:50.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-04-23 20:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-04-19 10:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-23 11:21:00.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-23 20:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-04-23 20:19:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-15 19:06:18.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-16 18:21:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-17 18:09:07.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-19 10:44:35.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-04-23 11:21:00.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-23 20:19:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-23 20:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d2472628a45e7b30aedc5bb1e5f96eeabdf80c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-14 09:30:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbd7d3f856bbf319df80514fdc01ceaeb710a9bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-14 09:37:34.000000000","message":"Patch Set 1: Workflow-1\n\nWe depend on lxc_containers_create, which seems to be broken as well.","accounts_in_message":[],"_revision_number":1},{"id":"67631dddc548cd4a54b92954fb493bc37b6b7dcd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-14 09:56:32.000000000","message":"Removed Workflow-1 by Dmitriy Rabotjagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"bbceda536a9290898193d73059ecd829d8cb6860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 10:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/1/check/openstack-ansible-linters/fe3194f/ : SUCCESS in 8m 09s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-debian-stable/aacc1d8/ : SUCCESS in 8m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-centos-7/c391177/ : SUCCESS in 8m 54s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-ubuntu-bionic/4ab6a74/ : SUCCESS in 6m 38s\n- openstack-tox-docs http://logs.openstack.org/68/652268/1/check/openstack-tox-docs/75f9ef6/html/ : SUCCESS in 2m 19s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-centos-7/f8c280f/ : SUCCESS in 15m 11s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-opensuse-150/b9dfd79/ : SUCCESS in 20m 18s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/43ad909/ : SUCCESS in 11m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/085c8b4/ : SUCCESS in 10m 31s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-btrfs-opensuse-150/d6186ee/ : SUCCESS in 15m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/16e281e/ : SUCCESS in 10m 02s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-centos-7/918b485/ : SUCCESS in 17m 28s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-opensuse-150/cc52029/ : SUCCESS in 13m 58s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6308d62/ : SUCCESS in 20m 36s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/6cc4da5/ : SUCCESS in 20m 38s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/cdb3925/ : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b2c6c4d8c15eca7198c7f42dae4a2adb04286c7a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-15 19:06:18.000000000","message":"Patch Set 1: Code-Review+1\n\ngreat, thanks a lot!","accounts_in_message":[],"_revision_number":1},{"id":"88f461fab9d308efb7975a23c1e5bb171f226d1c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-16 18:21:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"323c0cacd1c37582fc56e080690093792f5aa581","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-16 18:54:16.000000000","message":"Patch Set 1:\n\nPlease consider https://review.openstack.org/#/c/653073/ as an alternative.","accounts_in_message":[],"_revision_number":1},{"id":"e5f703a9ae69c6570393f2f5c12b76b7c720a778","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-16 20:51:21.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.openstack.org/#/c/653073","accounts_in_message":[],"_revision_number":1},{"id":"65d3c40e889d05bed3acd6e67b818585d8a71b6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-17 10:00:28.000000000","message":"Restored\n\nJust as an alternative will restore patch for now, if we\u0027ll decide to use aria after all. Will abandon once https://review.openstack.org/#/c/653073 get merged.","accounts_in_message":[],"_revision_number":1},{"id":"435ebad2c932c7fa5e749221951f94e8e37bb146","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-17 10:02:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36042246a61d85c8404887af6cc3b78f3373368b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 11:09:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/1/check/openstack-ansible-linters/9e99a0b/ : SUCCESS in 7m 53s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-debian-stable/2348b8a/ : SUCCESS in 11m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-centos-7/8c72743/ : SUCCESS in 10m 17s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-functional-ubuntu-bionic/a8296ec/ : SUCCESS in 8m 42s\n- openstack-tox-docs http://logs.openstack.org/68/652268/1/check/openstack-tox-docs/c3d5082/html/ : SUCCESS in 3m 53s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-centos-7/34e2786/ : SUCCESS in 14m 46s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-opensuse-150/69dd3c8/ : SUCCESS in 17m 22s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/8f06e27/ : SUCCESS in 15m 24s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/f48a093/ : SUCCESS in 15m 12s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-btrfs-opensuse-150/4330638/ : SUCCESS in 20m 49s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/2c9075b/ : SUCCESS in 10m 57s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-centos-7/2effe7f/ : SUCCESS in 16m 11s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-machinectl-opensuse-150/bee1236/ : SUCCESS in 19m 32s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/1380316/ : SUCCESS in 12m 44s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/993a2dc/ : SUCCESS in 21m 17s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/68/652268/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/53a03d9/ : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5511cedbba73278ed4e9151437feaad4be4360e1","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-17 18:09:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"640804c944e8178db28262ebe412f5ca40cbed9e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-17 18:34:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2239751b43a9f693be4ec81b2052b4dae7afd0f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-18 13:57:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73676eeb3ba1ba2e19b2f2b3d3f9f5403185ba37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:29:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/2/check/openstack-ansible-linters/f7a5e09/ : SUCCESS in 7m 01s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-debian-stable/2e4f055/ : FAILURE in 10m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-centos-7/a071d9d/ : SUCCESS in 11m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-ubuntu-bionic/2089ba5/ : SUCCESS in 12m 04s\n- openstack-tox-docs http://logs.openstack.org/68/652268/2/check/openstack-tox-docs/4b69627/html/ : SUCCESS in 3m 39s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-centos-7/8714a4a/ : SUCCESS in 14m 32s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-opensuse-150/b07de8e/ : SUCCESS in 18m 22s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/06e1a56/ : SUCCESS in 11m 48s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/a67d0e0/ : SUCCESS in 12m 41s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-btrfs-opensuse-150/c30cfa2/ : SUCCESS in 20m 31s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/49752b9/ : SUCCESS in 12m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-centos-7/5420ac9/ : SUCCESS in 16m 51s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-opensuse-150/7d59807/ : SUCCESS in 18m 07s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/abbdf80/ : SUCCESS in 19m 55s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/62e6677/ : SUCCESS in 19m 30s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/4219455/ : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":2},{"id":"390a5224f26509f5ac76ed01b0da86fbe9c2c46c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-18 15:30:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8ecd021a5be5ce11434cce82c0af22f9a89581cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 16:00:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/2/check/openstack-ansible-linters/f8be05d/ : SUCCESS in 6m 04s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-debian-stable/9b20c19/ : SUCCESS in 8m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-centos-7/5d61e51/ : SUCCESS in 9m 26s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-functional-ubuntu-bionic/64a7a54/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/68/652268/2/check/openstack-tox-docs/b10cbbb/html/ : SUCCESS in 3m 10s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-centos-7/eadf328/ : SUCCESS in 14m 32s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-opensuse-150/7eb3acf/ : SUCCESS in 17m 23s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/2fa5af0/ : SUCCESS in 13m 56s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/6d11f6c/ : SUCCESS in 13m 51s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-btrfs-opensuse-150/fb9ebee/ : SUCCESS in 20m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/3531b11/ : SUCCESS in 16m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-centos-7/5acb806/ : SUCCESS in 14m 24s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-machinectl-opensuse-150/311fc4d/ : SUCCESS in 18m 00s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/4157b21/ : SUCCESS in 14m 25s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/c2eccdf/ : SUCCESS in 20m 55s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/68/652268/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/55cc739/ : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"aeb1ca13b8d188beb80182730276db96d24c23f6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-18 17:08:56.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI like it, but have a few comments.","accounts_in_message":[],"_revision_number":2},{"id":"8cdfd5922ff295ceffeb4db016033f680cb6874f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-19 09:46:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eba1f9d227eac3af883e76efc534b5ef064ba4cb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-19 09:49:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"435fdebc3f62faee332b8ef708b8bb6e7867b683","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-19 09:50:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c56964fb4153c17b4016811435c63a4b2e66246","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-19 10:44:09.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI think this looks good to go. My comment shouldn\u0027t block merging this.","accounts_in_message":[],"_revision_number":2},{"id":"7821bd3708f4aaf3ec7b2e3d1a3c20bb8fc4ad02","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-19 10:44:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c232d05db77f2e79e3c904239b84faad44b74c21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 13:17:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/4/check/openstack-ansible-linters/a9b56fb/ : SUCCESS in 5m 20s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/4/check/openstack-ansible-functional-debian-stable/23a42da/ : SUCCESS in 11m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/4/check/openstack-ansible-functional-centos-7/b5b46ae/ : SUCCESS in 10m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-functional-ubuntu-bionic/1b0a5b6/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/68/652268/4/check/openstack-tox-docs/69078aa/html/ : SUCCESS in 2m 52s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-dir-centos-7/a043f34/ : SUCCESS in 16m 38s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-dir-opensuse-150/13a6361/ : SUCCESS in 15m 31s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-dir-ubuntu-bionic/ff73885/ : SUCCESS in 11m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/0090c9b/ : SUCCESS in 13m 38s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-btrfs-opensuse-150/e1d0e2d/ : SUCCESS in 16m 46s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/adf7cea/ : SUCCESS in 16m 03s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-machinectl-centos-7/86acaf3/ : SUCCESS in 15m 51s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-machinectl-opensuse-150/947c606/ : SUCCESS in 18m 45s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/215488c/ : SUCCESS in 17m 50s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-overlayfs-opensuse-150/162ed54/ : SUCCESS in 20m 59s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/68/652268/4/check/openstack-ansible-lxc-zfs-ubuntu-bionic/b16c465/ : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":4},{"id":"cc3f1454fe5c77264238929fabbea275a349784e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-19 13:20:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82ce2e9795c3b4965785dc862dab17cb49c1ceeb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-19 14:35:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6046eeb40f004474552f7cbc879ddc9491f5ff33","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-23 11:21:00.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nI am good with this, but I think it\u0027s not great.","accounts_in_message":[],"_revision_number":4},{"id":"dc5813ff0dfe5089b2518a1758c6eb4aaa537261","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 12:58:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf299731b7a01856b97eddfa9a8488d7cba7ec20","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-23 17:05:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"433d9fea4c3e3c2b5face9739aaffc0018e350f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:19:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1b1f47b5b63177d543b6d19689eca8974df434f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:19:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0adec2bc8819f5e3d48f5cd5e86c41369477c1b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4929d66aacbfc96d1ed984d0829eb28a61b22eca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 20:35:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/68/652268/4/gate/openstack-ansible-linters/b39b2e6/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/68/652268/4/gate/openstack-ansible-functional-debian-stable/6981360/ : SUCCESS in 8m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/68/652268/4/gate/openstack-ansible-functional-centos-7/9a2040d/ : SUCCESS in 9m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/68/652268/4/gate/openstack-ansible-functional-ubuntu-bionic/62d2965/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/68/652268/4/gate/openstack-tox-docs/e18946c/html/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c245713675efef226ffb69e7efa165800503f0ac","revisions":{"29798afc96b26f22396aac7d63537db10cc90005":{"kind":"REWORK","_number":1,"created":"2019-04-14 09:30:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/652268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/68/652268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/1"}}},"commit":{"parents":[{"commit":"d1c45fb0cbed6aadbaa568c88c0fea468a02cf46","subject":"Use pkill for lxc-dnsmasq systemd unit file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d1c45fb0cbed6aadbaa568c88c0fea468a02cf46"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-14 09:30:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-14 09:30:45.000000000","tz":180},"subject":"Fix OpenSuse 15.0","message":"Fix OpenSuse 15.0\n\nThis patch fixes suse 15 container_image file retrievment process.\n\nChange-Id: I45647062c5766f1f9d4dd098f107e91caf52a3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/29798afc96b26f22396aac7d63537db10cc90005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/29798afc96b26f22396aac7d63537db10cc90005"}]},"branch":"refs/heads/master"},"e33cdd205f4a9d1a0da6de220fee8f5117127137":{"kind":"REWORK","_number":2,"created":"2019-04-18 13:57:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/652268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/68/652268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/2"}}},"commit":{"parents":[{"commit":"d1c45fb0cbed6aadbaa568c88c0fea468a02cf46","subject":"Use pkill for lxc-dnsmasq systemd unit file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d1c45fb0cbed6aadbaa568c88c0fea468a02cf46"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-14 09:30:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-04-18 13:51:53.000000000","tz":180},"subject":"Fix image fetching process for leap 15","message":"Fix image fetching process for leap 15\n\nThis patch fixes suse 15 container_image file retrievment process,\nas aria2c does not save image under the correct name as it\u0027s a\nmetafile.\n\nThis reverts commit 6c9e26e965b98b7c1d38a3003a5a2c018f896db6.\nChange-Id: I45647062c5766f1f9d4dd098f107e91caf52a3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e33cdd205f4a9d1a0da6de220fee8f5117127137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/e33cdd205f4a9d1a0da6de220fee8f5117127137"}]},"branch":"refs/heads/master"},"24cc94884b533aac297532a8dd82d1855b6bc5bf":{"kind":"REWORK","_number":3,"created":"2019-04-19 09:46:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/652268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/68/652268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/3"}}},"commit":{"parents":[{"commit":"d1c45fb0cbed6aadbaa568c88c0fea468a02cf46","subject":"Use pkill for lxc-dnsmasq systemd unit file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d1c45fb0cbed6aadbaa568c88c0fea468a02cf46"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-14 09:30:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-04-19 09:46:12.000000000","tz":180},"subject":"Fix image fetching process for leap 15","message":"Fix image fetching process for leap 15\n\nThis patch fixes suse 15 container_image file retrievment process,\nas aria2c does not save image under the correct name as it\u0027s a\nmetafile.\n\nThis reverts commit 6c9e26e965b98b7c1d38a3003a5a2c018f896db6.\nChange-Id: I45647062c5766f1f9d4dd098f107e91caf52a3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/24cc94884b533aac297532a8dd82d1855b6bc5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/24cc94884b533aac297532a8dd82d1855b6bc5bf"}]},"branch":"refs/heads/master"},"c245713675efef226ffb69e7efa165800503f0ac":{"kind":"REWORK","_number":4,"created":"2019-04-19 09:50:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/652268/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/68/652268/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/68/652268/4"}}},"commit":{"parents":[{"commit":"d1c45fb0cbed6aadbaa568c88c0fea468a02cf46","subject":"Use pkill for lxc-dnsmasq systemd unit file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d1c45fb0cbed6aadbaa568c88c0fea468a02cf46"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-14 09:30:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-04-19 09:50:25.000000000","tz":180},"subject":"Fix image fetching process for leap 15","message":"Fix image fetching process for leap 15\n\nThis patch fixes suse 15 container_image file retrievment process,\nas aria2c does not save image under the correct name as it\u0027s a\nmetafile.\n\nThis reverts commit 6c9e26e965b98b7c1d38a3003a5a2c018f896db6.\nChange-Id: I45647062c5766f1f9d4dd098f107e91caf52a3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c245713675efef226ffb69e7efa165800503f0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/c245713675efef226ffb69e7efa165800503f0ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
