)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~654341","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I0c877da55db4491ef027c21cae65c11ca8a705dc","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I0c877da55db4491ef027c21cae65c11ca8a705dc","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:12:41.000000000","updated":"2019-04-22 11:23:39.000000000","submitted":"2019-04-22 11:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654341-1555932219061-332a2ac9","meta_rev_id":"a6cc76d830ed957c0f0c5da2b4d34fab955503fa","_number":654341,"virtual_id_number":654341,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 11:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 11:02:57.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 11:02:57.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 11:02:57.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 11:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a315623c23396555ea8f92e0b6652ce05394442","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7955ee2dcd3e6ca96cc21c464522c7cadb440e6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/654341/1/check/openstack-ansible-linters/87d3787/ : SUCCESS in 5m 48s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/41/654341/1/check/openstack-ansible-functional-debian-stable/bde1ce0/ : SUCCESS in 8m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/654341/1/check/openstack-ansible-functional-centos-7/0ce4206/ : SUCCESS in 9m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-functional-ubuntu-bionic/de44856/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/41/654341/1/check/openstack-tox-docs/afa2833/html/ : SUCCESS in 2m 52s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-dir-centos-7/ebf1d97/ : SUCCESS in 14m 37s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-dir-opensuse-150/131400d/ : FAILURE in 9m 07s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/36f817a/ : SUCCESS in 11m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/f86696f/ : SUCCESS in 11m 52s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-btrfs-opensuse-150/de4a36b/ : FAILURE in 13m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/7089711/ : SUCCESS in 11m 48s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-machinectl-centos-7/927c526/ : SUCCESS in 14m 44s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-machinectl-opensuse-150/9cdb798/ : FAILURE in 13m 27s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/13a8ad6/ : SUCCESS in 17m 15s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/d95da8f/ : FAILURE in 9m 27s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/41/654341/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/ef12884/ : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b0ccb06fedd562c5a1ba71b06746e79bab8ae66d","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 11:02:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c72bc295b2c9f3f7cd0b8a49ba781fce9cc8f736","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:03:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"557f0cc11e146dff3225b607c21dabb91cb040b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:23:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/654341/1/gate/openstack-ansible-linters/24334d3/ : SUCCESS in 5m 55s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/41/654341/1/gate/openstack-ansible-functional-debian-stable/122a261/ : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/654341/1/gate/openstack-ansible-functional-centos-7/dfb64c8/ : SUCCESS in 9m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/41/654341/1/gate/openstack-ansible-functional-ubuntu-bionic/2974b67/ : SUCCESS in 8m 17s\n- openstack-tox-docs http://logs.openstack.org/41/654341/1/gate/openstack-tox-docs/5928ce8/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"20c83a71ee092a2cfd98d680af30913b13c7d8bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:23:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77a23b3e4f983f04676e4e25a754bb67896f2f2f","revisions":{"77a23b3e4f983f04676e4e25a754bb67896f2f2f":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:12:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/654341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/41/654341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/654341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/654341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/654341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/41/654341/1"}}},"commit":{"parents":[{"commit":"47b7d4d39b6e4684ae290fa1270ab4610dc20431","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/47b7d4d39b6e4684ae290fa1270ab4610dc20431"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:39.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I0c877da55db4491ef027c21cae65c11ca8a705dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/77a23b3e4f983f04676e4e25a754bb67896f2f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/77a23b3e4f983f04676e4e25a754bb67896f2f2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
