)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~655246","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I9a2bc7ded20f772af82956a81a9864c7ee17039c","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"fix_opensuse","hashtags":[],"change_id":"I9a2bc7ded20f772af82956a81a9864c7ee17039c","subject":"Use the \u0027real\u0027 name for the LXC image","status":"MERGED","created":"2019-04-23 20:39:56.000000000","updated":"2019-04-26 08:21:46.000000000","submitted":"2019-04-26 08:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655246-1556266906929-3740e7af","meta_rev_id":"19ec5f95a363146d05e8a0dd41421cb0ca307128","_number":655246,"virtual_id_number":655246,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"date":"2019-04-25 17:54:53.000000000","_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-26 08:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-04-25 13:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-26 08:03:17.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 08:03:17.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27379,"name":"jrosser","inactive":true},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 08:49:32.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-24 08:49:52.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-04-24 08:50:33.000000000","updated_by":{"_account_id":27379,"name":"jrosser","inactive":true},"reviewer":{"_account_id":27379,"name":"jrosser","inactive":true},"state":"REVIEWER"},{"updated":"2019-04-24 20:55:38.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-25 13:21:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-26 08:03:17.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-26 08:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9186ba3948cc38bcc60cbeeab9baa0708cef7799","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-23 20:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dc0a1591b9c17a52796cd0aa319a59b4712a737","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-23 20:43:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f869d83cdce493a5800aa95502ebc9c4d4bac671","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-23 20:44:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0df29e6788b6cbd43d3e35c7832e5abab74c79a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:47:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/46/655246/2/check/openstack-ansible-linters/4e32423/ : SUCCESS in 6m 11s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/46/655246/2/check/openstack-ansible-functional-debian-stable/0a78c75/ : SUCCESS in 8m 24s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/655246/2/check/openstack-ansible-functional-centos-7/2a8a99b/ : SUCCESS in 8m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-functional-ubuntu-bionic/33bdcdd/ : SUCCESS in 10m 46s\n- openstack-tox-docs http://logs.openstack.org/46/655246/2/check/openstack-tox-docs/72549a0/html/ : SUCCESS in 2m 51s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-dir-centos-7/c3d0f1e/ : SUCCESS in 14m 06s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-dir-opensuse-150/940e158/ : SUCCESS in 18m 16s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/ee4dda9/ : SUCCESS in 12m 23s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/6a730de/ : SUCCESS in 13m 01s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-btrfs-opensuse-150/878cb90/ : SUCCESS in 16m 22s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/a86c3eb/ : SUCCESS in 10m 08s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-machinectl-centos-7/134091b/ : SUCCESS in 14m 36s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-machinectl-opensuse-150/23ad066/ : SUCCESS in 19m 16s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/902449a/ : SUCCESS in 12m 26s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/01f5a4f/ : SUCCESS in 20m 34s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/46/655246/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/592c983/ : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6d354e88f3c01b0f07b68561bfb829cfe0eeadac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 08:49:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nIt\u0027s pretty arguable change for me, which have both pros and cons. So I\u0027ll set +1 for this and would like to listen to other cores opinion on this.","accounts_in_message":[],"_revision_number":2},{"id":"4dfc8cbfea9155c78b70260a8b704f21795605d0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 11:17:43.000000000","message":"Patch Set 2: Code-Review+2\n\nOk for me.","accounts_in_message":[],"_revision_number":2},{"id":"7400440e0a1e305369d37f2ffd1c8c04fc77c7f9","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-24 16:36:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"038eae18de703aca4abc08a9c960a22693400a9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 17:07:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"307bc274d49a6f61b2d8a6a4893f13e2cc63957f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 17:11:09.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nIt\u0027s a fair comment, we don\u0027t know how stable the aria interface is. But I suppose, as it is a package from distro (not a ppa), it will not get large modifications inside a LTS distro.","accounts_in_message":[],"_revision_number":2},{"id":"5c94b44957cb23cd8af6e0a4477239cbe2ba8dcf","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-24 19:01:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02887a64b1aa51166c3c83fb2153afffdf3cf24d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 20:25:00.000000000","message":"Patch Set 2:\n\nProbably we should check, whether file for download is actually a file (but not metalink). And if it\u0027s metalink - use some appropriate method?\nI\u0027m wondering if we may set --follow-metalink\u003dmem - then we should have only one file downloaded, which should be easy to get from local filesystem. Don\u0027t have a laptop on my hands until tomorrow to check that...","accounts_in_message":[],"_revision_number":2},{"id":"1f22519a893749e87ecd6d2670bf80f73cb47d5e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 20:34:08.000000000","message":"Patch Set 2:\n\nOr even try to use aria --show-files","accounts_in_message":[],"_revision_number":2},{"id":"501cbcfabb35c1353fdafc45034651fb53ceafec","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 20:55:38.000000000","message":"Patch Set 2:\n\nthis feels eh.\n\nwhy can\u0027t suse just provide an actual tar.gz ?","accounts_in_message":[],"_revision_number":2},{"id":"bb1b8fb4d0be3b9ee0ab37942a4fb9a17a241402","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 03:58:14.000000000","message":"Patch Set 2:\n\nI couldn\u0027t agree more mnaser, but oh well ;)\n\nI\u0027ll play with some of the other aria2c options tomorrow noonedeadpunk and see whether I can simplify this change.\n\nThanks for the hints!","accounts_in_message":[],"_revision_number":2},{"id":"8eb10a976fcc2c0a8295a11426fc7f1a5e7076e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 06:56:15.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSo I\u0027ve played a bit with options, and seems that your\u0027s really good one. Otherwise we need to place some script, which will be doing almost the same - saving filename to some file (or maybe directly to ansible facts?), which we will be reading afterwards. But probably this might be a bit more reliable, as now we just hope, that nothing will be in stdout after --on-download-complete execution.","accounts_in_message":[],"_revision_number":2},{"id":"d41565e47bfd3b2cd6fced040e895d18434be761","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 12:59:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4bd2872702f7a733a47ec4c876f958504dd511fa","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 13:01:00.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for your helpful comments Dmitriy. I added a helper script so that we don\u0027t rely anymore simply on stdout.","accounts_in_message":[],"_revision_number":2},{"id":"93a4bd897e6311b0b58cf4cfa83f5d11995d8daa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 13:09:42.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis looks to me really good. Just one thing I don\u0027t like.","accounts_in_message":[],"_revision_number":3},{"id":"775cd1881379f71101bb6bd24064d827e3c1f593","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 13:12:11.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"4024fc7808c61ad2a331949f170d5eeb1a92016b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 13:12:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"54705551fa7545b339ae0873701c6a1d5ecc7b74","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-25 13:21:41.000000000","message":"Patch Set 4: Code-Review+2\n\nI like this change, thanks for your work Nicolas!","accounts_in_message":[],"_revision_number":4},{"id":"d98dc273121b248099afa41340e86002274cd0c6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 13:32:32.000000000","message":"Patch Set 4:\n\nThanks Dmitriy!","accounts_in_message":[],"_revision_number":4},{"id":"ef5ff2be75c30dbd44b3899fa6b56a988de2db7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 14:52:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/46/655246/4/check/openstack-ansible-linters/f03cb7c/ : SUCCESS in 4m 38s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-debian-stable/470ca39/ : SUCCESS in 7m 46s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-centos-7/efabf2b/ : SUCCESS in 8m 10s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-ubuntu-bionic/8d188bf/ : SUCCESS in 7m 05s\n- openstack-tox-docs http://logs.openstack.org/46/655246/4/check/openstack-tox-docs/2359c29/html/ : SUCCESS in 2m 38s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-centos-7/fb285f6/ : SUCCESS in 14m 52s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-opensuse-150/a942488/ : SUCCESS in 18m 53s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-ubuntu-bionic/55354ce/ : SUCCESS in 11m 26s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/7ce3fb4/ : SUCCESS in 12m 49s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-btrfs-opensuse-150/209a1ce/ : SUCCESS in 18m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/f01bac8/ : TIMED_OUT in 1h 32m 18s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-centos-7/bb842f9/ : SUCCESS in 19m 41s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-opensuse-150/0e98ab9/ : SUCCESS in 23m 17s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/cb1ebd6/ : SUCCESS in 14m 04s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-overlayfs-opensuse-150/46c6a45/ : SUCCESS in 19m 58s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-zfs-ubuntu-bionic/dbdf0db/ : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":4},{"id":"598422b84fb7582a5654f9b22745b762a1295a06","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-04-25 17:54:53.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"69167b8a0fbdcb924bfd143fad128d0a3414e006","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 18:23:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/46/655246/4/check/openstack-ansible-linters/a46494d/ : SUCCESS in 8m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-debian-stable/950f7c5/ : SUCCESS in 8m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-centos-7/4f2f05e/ : SUCCESS in 10m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-functional-ubuntu-bionic/546d2a6/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/46/655246/4/check/openstack-tox-docs/6d0ac8a/html/ : SUCCESS in 2m 31s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-centos-7/006602a/ : SUCCESS in 13m 42s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-opensuse-150/1505ea5/ : SUCCESS in 18m 14s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-dir-ubuntu-bionic/a97809b/ : SUCCESS in 12m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/a0a3e2e/ : SUCCESS in 12m 24s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-btrfs-opensuse-150/17d5425/ : SUCCESS in 19m 47s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/453d125/ : SUCCESS in 10m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-centos-7/e20ba9e/ : SUCCESS in 16m 02s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-machinectl-opensuse-150/b6f16b9/ : SUCCESS in 15m 04s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/0676cfb/ : SUCCESS in 11m 48s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-overlayfs-opensuse-150/a4d7f73/ : SUCCESS in 17m 25s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/46/655246/4/check/openstack-ansible-lxc-zfs-ubuntu-bionic/6b753c6/ : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":4},{"id":"9e9ed92f535cae3664fba643e79cb4e7d4bf06dd","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 07:50:33.000000000","message":"Patch Set 4: Code-Review-1\n\nGod damnit, this thing has changed again.","accounts_in_message":[],"_revision_number":4},{"id":"a66f468b50690b228aa9eb04e49a4623539e6a71","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 08:03:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nOk link was changed too in this patch. Good for me.","accounts_in_message":[],"_revision_number":4},{"id":"e09463da235db3caf04df7b16a3d188ae5d37942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:03:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"47ff5947c29090d9c0dd66d8175995b2d428e16f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"aae5810818dd8c4f2ff0edccffb7682457645ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 08:21:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/46/655246/4/gate/openstack-ansible-linters/fbd1756/ : SUCCESS in 4m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/46/655246/4/gate/openstack-ansible-functional-debian-stable/f920180/ : SUCCESS in 7m 20s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/46/655246/4/gate/openstack-ansible-functional-centos-7/9edf629/ : SUCCESS in 13m 44s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/46/655246/4/gate/openstack-ansible-functional-ubuntu-bionic/a8b57cf/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/46/655246/4/gate/openstack-tox-docs/336c508/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ba8a339377ecbf04f82007bb75b5b2f816d67cdf","revisions":{"ca5fcf5d49e8c0dd9548f0276865cad14f6a1200":{"kind":"REWORK","_number":1,"created":"2019-04-23 20:39:56.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/46/655246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/655246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/1"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:30:46.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:37:34.000000000","tz":-360},"subject":"Use the \u0027real\u0027 name for the LXC image","message":"Use the \u0027real\u0027 name for the LXC image\n\nThe Open Build Service repository for the Leap 15.0 LXC image returns\na metalink which causes aria2c to download 2 files. The first is an\nempty file (the metalink) and the second is the actual LXC image. The\nname of the second file changes frequently.\n\nThis change uses the `--on-download-complete` callback of aria2c to\nprint out the name of the actual file.\n\nChange-Id: I9a2bc7ded20f772af82956a81a9864c7ee17039c\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca5fcf5d49e8c0dd9548f0276865cad14f6a1200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ca5fcf5d49e8c0dd9548f0276865cad14f6a1200"}]},"branch":"refs/heads/master"},"6221397665d47277dc66a6c1ffa666be3bc04396":{"kind":"REWORK","_number":2,"created":"2019-04-23 20:43:17.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/46/655246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/655246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/2"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:30:46.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:42:50.000000000","tz":-360},"subject":"Use the \u0027real\u0027 name for the LXC image","message":"Use the \u0027real\u0027 name for the LXC image\n\nThe Open Build Service repository for the Leap 15.0 LXC image returns\na metalink which causes aria2c to download 2 files. The first is an\nempty file (the metalink) and the second is the actual LXC image. The\nname of the second file changes frequently.\n\nThis change uses the `--on-download-complete` callback of aria2c to\nprint out the name of the actual file.\n\nChange-Id: I9a2bc7ded20f772af82956a81a9864c7ee17039c\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6221397665d47277dc66a6c1ffa666be3bc04396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6221397665d47277dc66a6c1ffa666be3bc04396"}]},"branch":"refs/heads/master"},"581befd95174daf8fc868e266beb97b5bc562f7b":{"kind":"REWORK","_number":3,"created":"2019-04-25 12:59:39.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/46/655246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/655246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/3"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:30:46.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-25 12:59:17.000000000","tz":-360},"subject":"Use the \u0027real\u0027 name for the LXC image","message":"Use the \u0027real\u0027 name for the LXC image\n\nThe Open Build Service repository for the Leap 15.0 LXC image returns\na metalink which causes aria2c to download 2 files. The first is an\nempty file (the metalink) and the second is the actual LXC image. The\nname of the second file changes frequently.\n\nThis change uses the `--on-download-complete` callback of aria2c to\ncall a helper script which links the expected filename to the actual\nfilename.\n\nChange-Id: I9a2bc7ded20f772af82956a81a9864c7ee17039c\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/581befd95174daf8fc868e266beb97b5bc562f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/581befd95174daf8fc868e266beb97b5bc562f7b"}]},"branch":"refs/heads/master"},"ba8a339377ecbf04f82007bb75b5b2f816d67cdf":{"kind":"REWORK","_number":4,"created":"2019-04-25 13:12:37.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/46/655246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/46/655246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/46/655246/4"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-23 20:30:46.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-04-25 13:12:30.000000000","tz":-360},"subject":"Use the \u0027real\u0027 name for the LXC image","message":"Use the \u0027real\u0027 name for the LXC image\n\nThe Open Build Service repository for the Leap 15.0 LXC image returns\na metalink which causes aria2c to download 2 files. The first is an\nempty file (the metalink) and the second is the actual LXC image. The\nname of the second file changes frequently.\n\nThis change uses the `--on-download-complete` callback of aria2c to\ncall a helper script which links the expected filename to the actual\nfilename.\n\nChange-Id: I9a2bc7ded20f772af82956a81a9864c7ee17039c\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ba8a339377ecbf04f82007bb75b5b2f816d67cdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ba8a339377ecbf04f82007bb75b5b2f816d67cdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
