)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~655337","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I087b1045722bb6246544e42a1a0b60c88adc4c3f","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I087b1045722bb6246544e42a1a0b60c88adc4c3f","subject":"update git.openstack.org to opendev","status":"ABANDONED","created":"2019-04-24 07:27:39.000000000","updated":"2021-01-13 17:23:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47e0d8232ff920e13b71aa349bb6bdc4f85d230e","_number":655337,"virtual_id_number":655337,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-24 15:16:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 07:40:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 15:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f5fcc2f2e508f3c726681b663e348623356786e","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 07:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d62b5d69a5a94e54703aa7f858706308865f7f56","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-24 07:40:22.000000000","message":"Patch Set 1: Code-Review-1\n\nSee comments in https://review.opendev.org/#/c/654744","accounts_in_message":[],"_revision_number":1},{"id":"43c98f8e54762c7893dec053fab8212220cbcee6","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 08:09:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"72bdd0c5f7cf0968ce0816b253950cc75099bc6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:48:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/37/655337/2/check/openstack-ansible-linters/f8efa6e/ : FAILURE in 3m 12s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/37/655337/2/check/openstack-ansible-functional-debian-stable/c055460/ : FAILURE in 2m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/37/655337/2/check/openstack-ansible-functional-centos-7/1c45ef7/ : FAILURE in 3m 20s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-functional-ubuntu-bionic/d674c32/ : FAILURE in 2m 31s\n- openstack-tox-docs http://logs.openstack.org/37/655337/2/check/openstack-tox-docs/7086284/html/ : SUCCESS in 3m 58s\n- build-openstack-releasenotes http://logs.openstack.org/37/655337/2/check/build-openstack-releasenotes/18d019b/html/ : SUCCESS in 3m 39s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-dir-centos-7/c49e4af/ : SUCCESS in 14m 48s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-dir-opensuse-150/8f5e4fb/ : SUCCESS in 17m 49s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/812b2e2/ : SUCCESS in 10m 39s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/09a642d/ : SUCCESS in 10m 41s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-btrfs-opensuse-150/570dcad/ : SUCCESS in 15m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/e931501/ : SUCCESS in 17m 14s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-machinectl-centos-7/42f78f0/ : SUCCESS in 13m 46s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-machinectl-opensuse-150/f6c5b4b/ : SUCCESS in 18m 38s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/5efb05f/ : SUCCESS in 11m 48s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/4970feb/ : SUCCESS in 18m 57s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/37/655337/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/229cafc/ : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5bfd8e8ab046f839323132f4c9eaae7634aea697","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 14:35:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6ff4f4eef835189f6f389376ae4bada14dabad8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:16:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/655337/3/check/openstack-ansible-linters/2a4a6f2/ : SUCCESS in 5m 50s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/37/655337/3/check/openstack-ansible-functional-debian-stable/32cf6fe/ : SUCCESS in 8m 45s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/37/655337/3/check/openstack-ansible-functional-centos-7/99de3e1/ : SUCCESS in 8m 11s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-functional-ubuntu-bionic/e427833/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/37/655337/3/check/openstack-tox-docs/518dc86/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/37/655337/3/check/build-openstack-releasenotes/eb4652a/html/ : SUCCESS in 2m 39s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-dir-centos-7/4299041/ : SUCCESS in 14m 35s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-dir-opensuse-150/ec7fb4f/ : SUCCESS in 17m 04s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-dir-ubuntu-bionic/684e23e/ : SUCCESS in 11m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/fd99768/ : SUCCESS in 12m 20s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-btrfs-opensuse-150/23c9f6f/ : SUCCESS in 19m 54s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/5257228/ : SUCCESS in 11m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-machinectl-centos-7/c7577bf/ : SUCCESS in 21m 54s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-machinectl-opensuse-150/6fbc74d/ : SUCCESS in 17m 32s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/1052a08/ : SUCCESS in 20m 03s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-overlayfs-opensuse-150/92f1e7b/ : SUCCESS in 16m 55s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/37/655337/3/check/openstack-ansible-lxc-zfs-ubuntu-bionic/f90d685/ : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":3},{"id":"47e0d8232ff920e13b71aa349bb6bdc4f85d230e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-13 17:23:35.000000000","message":"Abandoned\n\nthe changes in this patch are merged in master through other patches already","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d925d38e958c697467d5ac3f3698bafdb7e04a34","revisions":{"23d23805d7a0ea93ebcae9378375a74bb13a3c2f":{"kind":"REWORK","_number":1,"created":"2019-04-24 07:27:39.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/37/655337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/655337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/1"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 07:27:03.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 07:27:31.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I087b1045722bb6246544e42a1a0b60c88adc4c3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23d23805d7a0ea93ebcae9378375a74bb13a3c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/23d23805d7a0ea93ebcae9378375a74bb13a3c2f"}]},"branch":"refs/heads/master"},"8b3d6cb005c21434ce11694964a40dee5c1d710f":{"kind":"REWORK","_number":2,"created":"2019-04-24 08:09:33.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/37/655337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/655337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/2"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 07:27:03.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 08:09:25.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I087b1045722bb6246544e42a1a0b60c88adc4c3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8b3d6cb005c21434ce11694964a40dee5c1d710f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8b3d6cb005c21434ce11694964a40dee5c1d710f"}]},"branch":"refs/heads/master"},"d925d38e958c697467d5ac3f3698bafdb7e04a34":{"kind":"REWORK","_number":3,"created":"2019-04-24 14:35:20.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/37/655337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/37/655337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/37/655337/3"}}},"commit":{"parents":[{"commit":"b5ec1c7eb82f0516ceee19b82529f2158c1f27f3","subject":"Merge \"Fix image fetching process for leap 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b5ec1c7eb82f0516ceee19b82529f2158c1f27f3"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 07:27:03.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 14:34:45.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I087b1045722bb6246544e42a1a0b60c88adc4c3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d925d38e958c697467d5ac3f3698bafdb7e04a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/d925d38e958c697467d5ac3f3698bafdb7e04a34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
