)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~656625","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I7770c5417feacfc43a3ad8801125603a55ba9ea8","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"fix_aria_centos","hashtags":[],"change_id":"I7770c5417feacfc43a3ad8801125603a55ba9ea8","subject":"Temporary fix of failing aria2 on centos","status":"MERGED","created":"2019-05-01 10:00:35.000000000","updated":"2019-05-13 14:46:31.000000000","submitted":"2019-05-03 18:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656625-1556906462904-11ac1cc1","meta_rev_id":"c2a7f1b40ca85f60c526ee499e1ea379f9d39137","_number":656625,"virtual_id_number":656625,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-03 18:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-03 17:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-03 17:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-05-03 17:41:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 17:20:33.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-03 17:41:34.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-03 18:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9d82b0b28f115cd3f0a3560e81a3cd3076f94fa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-01 10:00:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08a15b09a5e40bc2c0e40ca5a5f6a3e11f454b5a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-01 10:04:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b2226424bdc425254695eeda99bfce7f8df83b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 10:22:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/656625/2/check/openstack-ansible-linters/6457747/ : SUCCESS in 5m 00s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/25/656625/2/check/openstack-ansible-functional-debian-stable/0b9c068/ : SUCCESS in 9m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/25/656625/2/check/openstack-ansible-functional-centos-7/3d89512/ : SUCCESS in 10m 03s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-functional-ubuntu-bionic/66e4b24/ : SUCCESS in 7m 27s\n- openstack-tox-docs http://logs.openstack.org/25/656625/2/check/openstack-tox-docs/ff8112d/html/ : SUCCESS in 2m 35s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-dir-centos-7/f8c3240/ : SUCCESS in 13m 43s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-dir-opensuse-150/ae180fe/ : FAILURE in 10m 01s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/ca5a882/ : SUCCESS in 14m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/5cf8b04/ : SUCCESS in 13m 35s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-btrfs-opensuse-150/afe4a25/ : FAILURE in 9m 32s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/1a96190/ : SUCCESS in 10m 44s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-machinectl-centos-7/c685c22/ : SUCCESS in 15m 26s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-machinectl-opensuse-150/618147a/ : FAILURE in 9m 13s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/4f37394/ : SUCCESS in 11m 11s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/e00978b/ : FAILURE in 8m 59s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/25/656625/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/a5b1694/ : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"aeadac827d204707c7541d58e13e58c8b8deae80","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-01 11:07:03.000000000","message":"Topic changed from fix_araia_centos to fix_aria_centos","accounts_in_message":[],"_revision_number":2},{"id":"8d82d189c16b83a9da443439866ae81a888399d7","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-01 15:46:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fc278b47580f1ca3012801be106438e76335a52","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-01 16:07:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"62f7993b1e0bee94839b6037e570468b02100eb7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-01 17:11:06.000000000","message":"Patch Set 2:\n\nok, seems that redhat released a solution for this, so probably we won\u0027t need this at all https://bugzilla.redhat.com/show_bug.cgi?id\u003d1704413","accounts_in_message":[],"_revision_number":2},{"id":"08ff22b48a0ad000b0cf03cd9c10bb58e91a33ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-01 20:12:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4c3ed6197e6a50d65d9db55dee578f209e87b665","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 20:45:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/25/656625/3/check/openstack-ansible-linters/1eeb6ae/ : SUCCESS in 6m 22s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/25/656625/3/check/openstack-ansible-functional-debian-stable/914a235/ : SUCCESS in 9m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/25/656625/3/check/openstack-ansible-functional-centos-7/cb32563/ : FAILURE in 10m 22s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-functional-ubuntu-bionic/c20c302/ : SUCCESS in 12m 50s\n- openstack-tox-docs http://logs.openstack.org/25/656625/3/check/openstack-tox-docs/1d38442/html/ : SUCCESS in 3m 34s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-dir-centos-7/b1dfcaf/ : FAILURE in 9m 04s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-dir-opensuse-150/3f4ad46/ : FAILURE in 7m 58s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-dir-ubuntu-bionic/6d22e93/ : SUCCESS in 11m 22s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/97f55c9/ : SUCCESS in 10m 34s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-btrfs-opensuse-150/2deaafe/ : FAILURE in 8m 12s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/b1df4f8/ : SUCCESS in 13m 34s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-machinectl-centos-7/8e3c425/ : FAILURE in 14m 42s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-machinectl-opensuse-150/6abcf7a/ : FAILURE in 10m 10s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/46a5a91/ : SUCCESS in 11m 53s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-overlayfs-opensuse-150/42ecf81/ : FAILURE in 8m 50s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/25/656625/3/check/openstack-ansible-lxc-zfs-ubuntu-bionic/69b0287/ : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":3},{"id":"b5a26c0e93cc5bf6e04f37b5594fa38393d106f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-02 03:11:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c7afc317c10c9b0726d806c53c9e52e08b4ac42b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 03:45:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/656625/4/check/openstack-ansible-linters/35fb7d9/ : SUCCESS in 6m 08s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/25/656625/4/check/openstack-ansible-functional-debian-stable/4efb082/ : SUCCESS in 8m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/25/656625/4/check/openstack-ansible-functional-centos-7/a11f26b/ : SUCCESS in 10m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-functional-ubuntu-bionic/ab58f6c/ : SUCCESS in 8m 29s\n- openstack-tox-docs http://logs.openstack.org/25/656625/4/check/openstack-tox-docs/5d1a5a9/html/ : SUCCESS in 3m 21s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-dir-centos-7/092e660/ : SUCCESS in 15m 17s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-dir-opensuse-150/2d013e0/ : FAILURE in 9m 26s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-dir-ubuntu-bionic/c82d770/ : SUCCESS in 11m 28s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/e038ed4/ : SUCCESS in 13m 37s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-btrfs-opensuse-150/bb4b3c9/ : FAILURE in 9m 44s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/a079aad/ : SUCCESS in 12m 43s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-machinectl-centos-7/7378671/ : SUCCESS in 15m 07s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-machinectl-opensuse-150/f181193/ : FAILURE in 10m 05s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/3edd272/ : SUCCESS in 11m 10s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-overlayfs-opensuse-150/00b7078/ : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/25/656625/4/check/openstack-ansible-lxc-zfs-ubuntu-bionic/86231ea/ : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":4},{"id":"75c586ffe28a3c63565a0bc6e870f94db8276efc","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-03 17:20:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0db9aabd28b498ffa701d0bd4bf8690a8d75f6a8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-03 17:41:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"caf3564b4915420c091bd1d5642d37cc24ccea61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:41:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2df6ff0290e71f09823a74cbeff331f0d6e51621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:01:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/25/656625/4/gate/openstack-ansible-linters/b36997a/ : SUCCESS in 4m 38s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/25/656625/4/gate/openstack-ansible-functional-debian-stable/d9ca1cd/ : SUCCESS in 9m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/25/656625/4/gate/openstack-ansible-functional-centos-7/562ea68/ : SUCCESS in 9m 51s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/25/656625/4/gate/openstack-ansible-functional-ubuntu-bionic/4704f20/ : SUCCESS in 15m 01s\n- openstack-tox-docs http://logs.openstack.org/25/656625/4/gate/openstack-tox-docs/f9286c8/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"852fe5f64821bcb4c3f321b5c3bef6b6db8b38f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:01:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d3276243d48ffdff261e55249bd8dc0e1d32b67d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-06 00:38:47.000000000","message":"Created a revert of this change as Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","accounts_in_message":[],"_revision_number":4},{"id":"edcd43eeffae80a2d698282fde194a84ef8da988","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-06 14:11:59.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 5fb9fe50359e21dcc3a0d0dce7436b9f8e233831","accounts_in_message":[],"_revision_number":4},{"id":"2b95e07bc084bf19067bb4d2d622851c8d0d9165","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 10:10:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2a7f1b40ca85f60c526ee499e1ea379f9d39137","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-13 14:46:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"43e61820f42d88f36ec0ab4865800cbb4b1076ae","revisions":{"57cc278f705b55783816bc0cae6bb2553974ed15":{"kind":"REWORK","_number":1,"created":"2019-05-01 10:00:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/25/656625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/656625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/1"}}},"commit":{"parents":[{"commit":"864e6b851342d364b3313d72c9b2d6fb31d86a96","subject":"Merge \"Use the \u0027real\u0027 name for the LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/864e6b851342d364b3313d72c9b2d6fb31d86a96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:00:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:00:03.000000000","tz":180},"subject":"Temporary fix of failing aria2 on centos","message":"Temporary fix of failing aria2 on centos\n\nAdded get_url as an alternate downloader, when aria2 is failing\ndue to SSL error. Now it happens for CentOS and is reproducable.\nWe probably won\u0027t need this patch once aria2 package will be fixed.\n\nChange-Id: I7770c5417feacfc43a3ad8801125603a55ba9ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57cc278f705b55783816bc0cae6bb2553974ed15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/57cc278f705b55783816bc0cae6bb2553974ed15"}]},"branch":"refs/heads/master"},"ec812270e4d1b024f1a7e67ddd33f1041b2e864b":{"kind":"REWORK","_number":2,"created":"2019-05-01 10:04:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/25/656625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/656625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/2"}}},"commit":{"parents":[{"commit":"864e6b851342d364b3313d72c9b2d6fb31d86a96","subject":"Merge \"Use the \u0027real\u0027 name for the LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/864e6b851342d364b3313d72c9b2d6fb31d86a96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:00:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:03:51.000000000","tz":180},"subject":"Temporary fix of failing aria2 on centos","message":"Temporary fix of failing aria2 on centos\n\nAdded get_url as an alternate downloader, when aria2 is failing\ndue to SSL error. Now it happens for CentOS and is reproducable.\nWe probably won\u0027t need this patch once aria2 package will be fixed.\n\nChange-Id: I7770c5417feacfc43a3ad8801125603a55ba9ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec812270e4d1b024f1a7e67ddd33f1041b2e864b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ec812270e4d1b024f1a7e67ddd33f1041b2e864b"}]},"branch":"refs/heads/master"},"19baa2a9228b0f5029e14539f1b896570a2a34e4":{"kind":"REWORK","_number":3,"created":"2019-05-01 20:12:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/25/656625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/656625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/3"}}},"commit":{"parents":[{"commit":"864e6b851342d364b3313d72c9b2d6fb31d86a96","subject":"Merge \"Use the \u0027real\u0027 name for the LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/864e6b851342d364b3313d72c9b2d6fb31d86a96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:00:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-01 20:11:50.000000000","tz":0},"subject":"Temporary fix of failing aria2 on centos","message":"Temporary fix of failing aria2 on centos\n\nAdded get_url as an alternate downloader, when aria2 is failing\ndue to SSL error. Now it happens for CentOS and is reproducable.\nWe probably won\u0027t need this patch once aria2 package will be fixed.\n\nChange-Id: I7770c5417feacfc43a3ad8801125603a55ba9ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/19baa2a9228b0f5029e14539f1b896570a2a34e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/19baa2a9228b0f5029e14539f1b896570a2a34e4"}]},"branch":"refs/heads/master"},"43e61820f42d88f36ec0ab4865800cbb4b1076ae":{"kind":"REWORK","_number":4,"created":"2019-05-02 03:11:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/25/656625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/25/656625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/25/656625/4"}}},"commit":{"parents":[{"commit":"864e6b851342d364b3313d72c9b2d6fb31d86a96","subject":"Merge \"Use the \u0027real\u0027 name for the LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/864e6b851342d364b3313d72c9b2d6fb31d86a96"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-01 10:00:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-02 03:11:17.000000000","tz":0},"subject":"Temporary fix of failing aria2 on centos","message":"Temporary fix of failing aria2 on centos\n\nAdded get_url as an alternate downloader, when aria2 is failing\ndue to SSL error. Now it happens for CentOS and is reproducable.\nWe probably won\u0027t need this patch once aria2 package will be fixed.\n\nChange-Id: I7770c5417feacfc43a3ad8801125603a55ba9ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
