)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~657163","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I66e6f2dc83938f05b6913a6500b292416b592793","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I66e6f2dc83938f05b6913a6500b292416b592793","subject":"Do not use \u0027null\u0027 when trying to locate vars files","status":"MERGED","created":"2019-05-04 20:44:51.000000000","updated":"2019-05-12 21:33:55.000000000","submitted":"2019-05-12 21:33:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657163-1557696835657-96febb6b","meta_rev_id":"adfb602972b3853b60ef5552a98b8efb25d3e99e","_number":657163,"virtual_id_number":657163,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-12 21:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 20:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 17:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 20:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 17:59:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-12 20:37:51.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 21:33:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ecf0541664e5389e5dc5093c98870b5d3044b65","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-04 20:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2dc8a09b023b48f453e0cb97c8f04843d394702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 21:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/63/657163/1/check/openstack-ansible-linters/d810ddb/ : SUCCESS in 4m 21s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/63/657163/1/check/openstack-ansible-functional-debian-stable/209111d/ : SUCCESS in 7m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/63/657163/1/check/openstack-ansible-functional-centos-7/38717e7/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-functional-ubuntu-bionic/92f318c/ : SUCCESS in 6m 42s\n- openstack-tox-docs http://logs.openstack.org/63/657163/1/check/openstack-tox-docs/83d81f9/html/ : SUCCESS in 2m 11s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-dir-centos-7/1e53076/ : SUCCESS in 16m 35s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-dir-opensuse-150/ee23951/ : FAILURE in 6m 56s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/883fd3d/ : SUCCESS in 14m 11s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/ac301bb/ : SUCCESS in 12m 48s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-btrfs-opensuse-150/b8181d8/ : FAILURE in 10m 26s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/0ba987a/ : SUCCESS in 13m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-machinectl-centos-7/a887d4c/ : SUCCESS in 14m 59s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-machinectl-opensuse-150/1645eba/ : FAILURE in 6m 58s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/5810743/ : SUCCESS in 10m 58s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/b118d3d/ : FAILURE in 9m 32s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/63/657163/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/8b56de1/ : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"449e77acf2baf4edc1728f4c0e4bcfcad0498860","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-12 17:09:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"38fde7c84cb3052a44bd89780a6f968060dae6f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:55:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/63/657163/2/check/openstack-ansible-linters/41d28fb/ : SUCCESS in 7m 50s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-debian-stable/30015ef/ : SUCCESS in 6m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-centos-7/7de6086/ : SUCCESS in 12m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-ubuntu-bionic/5d8a1e1/ : SUCCESS in 7m 46s\n- openstack-tox-docs http://logs.openstack.org/63/657163/2/check/openstack-tox-docs/7bde5ab/html/ : SUCCESS in 2m 25s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-centos-7/61c3d87/ : SUCCESS in 17m 04s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-opensuse-150/7e18163/ : FAILURE in 15m 53s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/44afeb2/ : SUCCESS in 10m 59s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/3edffb8/ : SUCCESS in 10m 00s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-opensuse-150/251675f/ : FAILURE in 18m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/b7b06f0/ : SUCCESS in 11m 51s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-centos-7/cb34542/ : FAILURE in 8m 27s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-opensuse-150/3f6e169/ : FAILURE in 10m 40s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/cbd90f7/ : SUCCESS in 11m 25s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/ad6cfe1/ : FAILURE in 16m 49s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/e400527/ : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3b29f67f7775cbed69435b523f76ad157672dc5d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 17:56:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a30d3b25411bd382e8eaac4e8a9b4f5b620f230c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 17:59:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0890597521c235fd29321a7fcfa0a0f3b64c6ed4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 18:36:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/63/657163/2/check/openstack-ansible-linters/dcb9472/ : SUCCESS in 7m 31s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-debian-stable/25683ca/ : SUCCESS in 9m 27s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-centos-7/c9839d4/ : FAILURE in 7m 58s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-ubuntu-bionic/7bb28a6/ : SUCCESS in 13m 20s\n- openstack-tox-docs http://logs.openstack.org/63/657163/2/check/openstack-tox-docs/febf053/html/ : SUCCESS in 3m 57s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-centos-7/e5af2b8/ : SUCCESS in 18m 21s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-opensuse-150/b90f238/ : FAILURE in 17m 04s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/41f63c5/ : SUCCESS in 17m 01s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/2296697/ : SUCCESS in 18m 26s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-opensuse-150/90f19fc/ : FAILURE in 20m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/485b1d9/ : SUCCESS in 17m 56s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-centos-7/45d55a8/ : SUCCESS in 18m 38s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-opensuse-150/1020ab4/ : FAILURE in 15m 42s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/4d14c38/ : SUCCESS in 18m 23s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/21f77f9/ : FAILURE in 20m 22s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/ba5aec2/ : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5c15832cf1388f5d2cf7225c6583ca13e0cc4fec","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 20:37:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c61e8ee4e6f42e7cc7c74af1a02ed622fbecbd99","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 20:57:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2d2d7563750f3e6f53adae813ff50b75a62930d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:13:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/63/657163/2/check/openstack-ansible-linters/3490798/ : SUCCESS in 5m 13s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-debian-stable/09b4aea/ : SUCCESS in 6m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-centos-7/b8226fb/ : SUCCESS in 12m 16s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-functional-ubuntu-bionic/838732a/ : SUCCESS in 7m 45s\n- openstack-tox-docs http://logs.openstack.org/63/657163/2/check/openstack-tox-docs/5474897/html/ : SUCCESS in 2m 07s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-centos-7/1eec75d/ : SUCCESS in 15m 51s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-opensuse-150/0ffbabc/ : FAILURE in 13m 42s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/7527236/ : SUCCESS in 10m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/c07fcc9/ : SUCCESS in 11m 15s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-btrfs-opensuse-150/825e990/ : FAILURE in 17m 01s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/2e7c706/ : SUCCESS in 11m 30s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-centos-7/0cc7b64/ : SUCCESS in 12m 58s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-machinectl-opensuse-150/a4ec6f6/ : FAILURE in 12m 27s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/f5190a0/ : SUCCESS in 11m 14s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/ed42f02/ : FAILURE in 15m 45s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/63/657163/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/e48d6ff/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a6091d44fc3ed358e688040a3aff4c3d82affaa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:13:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68967a99a446e9ccf200dbc344fb48e213aea36d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:33:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"adfb602972b3853b60ef5552a98b8efb25d3e99e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:33:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/63/657163/2/gate/openstack-ansible-linters/62a644c/ : SUCCESS in 4m 15s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/63/657163/2/gate/openstack-ansible-functional-debian-stable/70e7e3b/ : SUCCESS in 7m 31s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/63/657163/2/gate/openstack-ansible-functional-centos-7/ff00a37/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/63/657163/2/gate/openstack-ansible-functional-ubuntu-bionic/51ea7db/ : SUCCESS in 7m 36s\n- openstack-tox-docs http://logs.openstack.org/63/657163/2/gate/openstack-tox-docs/5bfe328/html/ : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a39665eef157764eb731bef8fc6246e404a98914","revisions":{"6cbe29411fb696ac4777da8d213c6566e3867f8a":{"kind":"REWORK","_number":1,"created":"2019-05-04 20:44:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/63/657163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/63/657163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/1"}}},"commit":{"parents":[{"commit":"43e61820f42d88f36ec0ab4865800cbb4b1076ae","subject":"Temporary fix of failing aria2 on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-05-04 20:43:07.000000000","tz":-360},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-05-04 20:43:07.000000000","tz":-360},"subject":"Do not use \u0027null\u0027 when trying to locate vars files","message":"Do not use \u0027null\u0027 when trying to locate vars files\n\nThis generates warnings in the ansible log\n\nChange-Id: I66e6f2dc83938f05b6913a6500b292416b592793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6cbe29411fb696ac4777da8d213c6566e3867f8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6cbe29411fb696ac4777da8d213c6566e3867f8a"}]},"branch":"refs/heads/master"},"a39665eef157764eb731bef8fc6246e404a98914":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-12 17:09:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/63/657163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/63/657163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/63/657163/2"}}},"commit":{"parents":[{"commit":"af76d499056be4dca5b21d7b1340e17d413b257b","subject":"Merge \"Returning of the Suse LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/af76d499056be4dca5b21d7b1340e17d413b257b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-05-04 20:43:07.000000000","tz":-360},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-05-12 17:09:00.000000000","tz":0},"subject":"Do not use \u0027null\u0027 when trying to locate vars files","message":"Do not use \u0027null\u0027 when trying to locate vars files\n\nThis generates warnings in the ansible log\n\nChange-Id: I66e6f2dc83938f05b6913a6500b292416b592793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a39665eef157764eb731bef8fc6246e404a98914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a39665eef157764eb731bef8fc6246e404a98914"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
