)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~657292","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"fix_aria_centos","hashtags":[],"change_id":"Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","subject":"Revert \"Temporary fix of failing aria2 on centos\"","status":"MERGED","created":"2019-05-06 00:38:47.000000000","updated":"2019-05-20 10:37:18.000000000","submitted":"2019-05-20 10:37:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657292-1558348638912-a663f06a","meta_rev_id":"f32248104138fe121cbf25cad794ff101ee41cae","_number":657292,"virtual_id_number":657292,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-20 10:37:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-20 10:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-20 08:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-20 10:12:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 00:38:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-20 08:29:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-20 10:12:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-20 10:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cdd3d8f4c2f015872f1b9afe39506a6c3967714","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-06 00:38:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1ae9cca70b1c3e96e8400af8b81dbccee6d0e31","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-06 00:39:03.000000000","message":"Patch Set 1: Workflow-1\n\nThe fixed package is not actually promoted yet.","accounts_in_message":[],"_revision_number":1},{"id":"0fbe50e708ba3508de877d9e8f346ac3837265da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 00:59:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/92/657292/1/check/openstack-ansible-linters/cedfb76/ : SUCCESS in 4m 05s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/92/657292/1/check/openstack-ansible-functional-debian-stable/610f570/ : SUCCESS in 6m 19s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/657292/1/check/openstack-ansible-functional-centos-7/4f37905/ : FAILURE in 8m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-functional-ubuntu-bionic/d1917ce/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/92/657292/1/check/openstack-tox-docs/c614cee/html/ : SUCCESS in 2m 08s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-dir-centos-7/cf58765/ : FAILURE in 11m 09s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-dir-opensuse-150/f4c2180/ : FAILURE in 11m 53s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/98009c8/ : SUCCESS in 12m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/e12531e/ : SUCCESS in 11m 30s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-btrfs-opensuse-150/e380e7b/ : FAILURE in 11m 50s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/ca2d468/ : SUCCESS in 9m 30s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-machinectl-centos-7/6c566c0/ : FAILURE in 7m 51s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-machinectl-opensuse-150/fecb85d/ : FAILURE in 7m 19s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/6e6309a/ : SUCCESS in 11m 25s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/a55901f/ : FAILURE in 11m 34s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/92/657292/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/81793a7/ : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c87585032b200e5a976b1c8692242292d581884c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-06 05:52:35.000000000","message":"Patch Set 1:\n\nThere is stated, that fixed package is already in epel-testing, but not in the regular one, which we are using","accounts_in_message":[],"_revision_number":1},{"id":"0819735afefb18ca02c897aea255c74beea591bf","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 17:52:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2f5bb04875451ce0ad38720729881fa8453af35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 18:24:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/657292/2/check/openstack-ansible-linters/70fd025/ : SUCCESS in 3m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/92/657292/2/check/openstack-ansible-functional-debian-stable/219a658/ : SUCCESS in 9m 15s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/657292/2/check/openstack-ansible-functional-centos-7/07f905c/ : SUCCESS in 12m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-functional-ubuntu-bionic/94145d9/ : SUCCESS in 7m 05s\n- openstack-tox-docs http://logs.openstack.org/92/657292/2/check/openstack-tox-docs/f8690c8/html/ : SUCCESS in 2m 03s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-dir-centos-7/ef5cfe6/ : SUCCESS in 15m 46s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-dir-opensuse-150/686f0ec/ : FAILURE in 13m 28s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/dc25a39/ : SUCCESS in 10m 08s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/b728a25/ : SUCCESS in 9m 42s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-btrfs-opensuse-150/1fabbb7/ : FAILURE in 14m 26s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/3d42869/ : SUCCESS in 9m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-machinectl-centos-7/95cb67f/ : SUCCESS in 14m 51s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-machinectl-opensuse-150/4b606b8/ : FAILURE in 11m 17s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/35eb9d5/ : SUCCESS in 10m 38s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/c0ea483/ : FAILURE in 17m 58s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/92/657292/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/1d518d0/ : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8949868b1c614a5f95d6d28710632b0eced63200","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:01.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1b402c9b1b36c084fceb1a63a8a12acd73ade4d3","accounts_in_message":[],"_revision_number":2},{"id":"17d39d9bd5e1c3269496ac2099443a81161b2f6e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:43.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3a4bfb5d8823260cb809f788cd8d1756e0e08cb9","accounts_in_message":[],"_revision_number":2},{"id":"c5be35188786adfbf3a447503af271db83ead59e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 08:29:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"55831ea8e2639ae9a8ee641ccc37e05bdfac766a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-20 10:12:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d07a6cca45bc856a1b0c1ba3a11ffa407a04f2d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:12:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"511998ec13831fbb248990a589849c6ade0ef33a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:37:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c6c2ae97e9488fc3decaa6bf24c99ea7576eb25d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:37:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/657292/2/gate/openstack-ansible-linters/6fc76ae/ : SUCCESS in 5m 53s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/92/657292/2/gate/openstack-ansible-functional-debian-stable/8395509/ : SUCCESS in 8m 39s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/92/657292/2/gate/openstack-ansible-functional-centos-7/348d282/ : SUCCESS in 9m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/92/657292/2/gate/openstack-ansible-functional-ubuntu-bionic/bfeffb1/ : SUCCESS in 7m 59s\n- openstack-tox-docs http://logs.openstack.org/92/657292/2/gate/openstack-tox-docs/0f91938/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8cf5a6ad56231c26be4b367785f4d377749f8db2","revisions":{"446f5144fdea1dbb8096bcafe48280027a89bc23":{"kind":"REWORK","_number":1,"created":"2019-05-06 00:38:47.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/92/657292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/657292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/1"}}},"commit":{"parents":[{"commit":"43e61820f42d88f36ec0ab4865800cbb4b1076ae","subject":"Temporary fix of failing aria2 on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-06 00:38:47.000000000","tz":0},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-06 00:38:47.000000000","tz":0},"subject":"Revert \"Temporary fix of failing aria2 on centos\"","message":"Revert \"Temporary fix of failing aria2 on centos\"\n\nThe fix is no longer needed since a fixed aria2 package has been promoted in epel.\n\nSee https://bugzilla.redhat.com/show_bug.cgi?id\u003d1704413\n\nThis reverts commit 43e61820f42d88f36ec0ab4865800cbb4b1076ae.\n\nChange-Id: Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/446f5144fdea1dbb8096bcafe48280027a89bc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/446f5144fdea1dbb8096bcafe48280027a89bc23"}]},"branch":"refs/heads/master"},"8cf5a6ad56231c26be4b367785f4d377749f8db2":{"kind":"REWORK","_number":2,"created":"2019-05-19 17:52:00.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/92/657292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/92/657292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/92/657292/2"}}},"commit":{"parents":[{"commit":"b3939158743661019b697235fefeab16ca19300e","subject":"Use block/rescue for image pre-stage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b3939158743661019b697235fefeab16ca19300e"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 17:50:23.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 17:50:23.000000000","tz":-300},"subject":"Revert \"Temporary fix of failing aria2 on centos\"","message":"Revert \"Temporary fix of failing aria2 on centos\"\n\nThe fix is no longer needed since a fixed aria2 package has been promoted in epel.\n\nSee https://bugzilla.redhat.com/show_bug.cgi?id\u003d1704413\n\nThis is a revert of the following changes:\nI7770c5417feacfc43a3ad8801125603a55ba9ea8\nI26d3c9338247ffc789bf4a5291e65e954151cc51\nI4ec91f541e6d2d4cca4d899f4c7d519b169e2f50\n\nChange-Id: Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8cf5a6ad56231c26be4b367785f4d377749f8db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/8cf5a6ad56231c26be4b367785f4d377749f8db2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
