)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~657378","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I26e544c0d2a9eb740174a1d4c486b2fb583461dd","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","hashtags":[],"change_id":"I26e544c0d2a9eb740174a1d4c486b2fb583461dd","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-06 12:06:31.000000000","updated":"2019-08-15 00:52:40.000000000","submitted":"2019-08-15 00:52:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657378-1565830360026-3f7c88c0","meta_rev_id":"6bd6150daa2c184df9c2884875172eafa6139dd1","_number":657378,"virtual_id_number":657378,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-15 00:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-08-15 00:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-08-15 00:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 00:34:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-15 00:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49d54d4fdad9372f5db8941f919033f661b4e2c6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 12:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a5312738c1da7299b383edb5113755ebdd1256c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:26:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/657378/1/check/openstack-ansible-linters/6f74b4a/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/78/657378/1/check/openstack-ansible-functional-debian-stable/d4e2318/ : SUCCESS in 7m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/657378/1/check/openstack-ansible-functional-centos-7/cce3acb/ : SUCCESS in 8m 49s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-functional-ubuntu-bionic/f06f5a8/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/78/657378/1/check/openstack-tox-docs/f503134/html/ : SUCCESS in 2m 45s\n- build-openstack-releasenotes http://logs.openstack.org/78/657378/1/check/build-openstack-releasenotes/2451166/html/ : SUCCESS in 2m 26s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-dir-centos-7/655cd5f/ : SUCCESS in 14m 11s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-dir-opensuse-150/62d9389/ : FAILURE in 9m 17s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/8aebad5/ : SUCCESS in 11m 07s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/bfb1eed/ : SUCCESS in 11m 31s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-btrfs-opensuse-150/d59ad2c/ : FAILURE in 9m 59s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/1f93841/ : SUCCESS in 9m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-machinectl-centos-7/689ac0f/ : SUCCESS in 13m 53s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-machinectl-opensuse-150/1a4d816/ : FAILURE in 8m 31s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/2c09302/ : SUCCESS in 11m 42s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/76bed76/ : FAILURE in 8m 42s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/657378/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/33a25a5/ : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"280709e2b231cadd9d1985de9fbdd7de38e33db5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-15 00:34:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5db203547091111bbb6483f4ee7822055bdb573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:34:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60d4eb0562908e0d58f2c588df34c83866a570c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:52:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bab9bc3050dc4933828e93686ce357fc : SUCCESS in 5m 39s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/06f908a839564ea9898887227a51b3b5 : SUCCESS in 9m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/bcdf0b6172bf4eb9ada4f9deb4a1a5fd : SUCCESS in 10m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71b06f9312e54b7fa17fd4a623b58a3d : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50f6b7bc1da45e89cbcf11013949acc : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b016e8ffc6646bfb03591667d46bfc8 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6bd6150daa2c184df9c2884875172eafa6139dd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:52:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a02e4a9a51c10015e1a7d34feba28b35fabeb35f","revisions":{"a02e4a9a51c10015e1a7d34feba28b35fabeb35f":{"kind":"REWORK","_number":1,"created":"2019-05-06 12:06:31.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/657378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/657378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/657378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/657378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/657378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/657378/1"}}},"commit":{"parents":[{"commit":"43e61820f42d88f36ec0ab4865800cbb4b1076ae","subject":"Temporary fix of failing aria2 on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 12:06:03.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 12:06:29.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I26e544c0d2a9eb740174a1d4c486b2fb583461dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a02e4a9a51c10015e1a7d34feba28b35fabeb35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/a02e4a9a51c10015e1a7d34feba28b35fabeb35f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
