)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~657720","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I2bd032f0d5dc17f61e380b44a761fe110ccb42f5","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"suse_name","hashtags":[],"change_id":"I2bd032f0d5dc17f61e380b44a761fe110ccb42f5","subject":"Update (yet again) SUSE lxc container name","status":"MERGED","created":"2019-05-08 02:00:41.000000000","updated":"2019-05-08 10:46:37.000000000","submitted":"2019-05-08 06:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657720-1557296956547-92f2513b","meta_rev_id":"f9a9c2abcda43d70e9014f92d0c12ee8744778be","_number":657720,"virtual_id_number":657720,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-08 06:29:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-08 03:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-08 05:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-08 05:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 03:00:21.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-08 05:38:36.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-08 06:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a842e524181c69912099a0a33f33054cbc14feec","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-08 02:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20876a4177db05369829c879a9365d3a679778c0","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-08 03:00:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3ce8c0a2e6122c20d566bb69b914fd309259271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 03:36:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/20/657720/1/check/openstack-ansible-linters/f6e6055/ : SUCCESS in 4m 20s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-debian-stable/e094c9e/ : SUCCESS in 8m 12s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-centos-7/2e66de8/ : FAILURE in 11m 15s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-ubuntu-bionic/deced15/ : SUCCESS in 12m 11s\n- openstack-tox-docs http://logs.openstack.org/20/657720/1/check/openstack-tox-docs/fece657/html/ : SUCCESS in 2m 19s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-centos-7/a1390ec/ : SUCCESS in 25m 40s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-opensuse-150/da80015/ : FAILURE in 12m 15s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/abc4002/ : SUCCESS in 14m 55s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/812fe11/ : SUCCESS in 12m 24s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-btrfs-opensuse-150/eb69b38/ : FAILURE in 13m 10s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/fe1406a/ : TIMED_OUT in 1h 32m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-centos-7/7965482/ : SUCCESS in 16m 54s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-opensuse-150/ae6f68b/ : FAILURE in 13m 33s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/b55fdc7/ : SUCCESS in 16m 01s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/8448768/ : FAILURE in 14m 46s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/799f2af/ : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d8bdfa8372a15eb0ff2b069b60d1ad26a03c6dfb","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-08 05:38:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"268788e0d154c24c3ca124df55f38fd734f1217d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-08 05:38:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f03eb9c1c7a2d8a855a9630485b4956c42217590","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 06:14:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/657720/1/check/openstack-ansible-linters/7833ea8/ : SUCCESS in 4m 42s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-debian-stable/a687295/ : SUCCESS in 9m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-centos-7/9fdf9b6/ : SUCCESS in 12m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-functional-ubuntu-bionic/578d829/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/20/657720/1/check/openstack-tox-docs/b5bbb26/html/ : SUCCESS in 2m 26s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-centos-7/e21496b/ : SUCCESS in 17m 44s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-opensuse-150/27c9e98/ : FAILURE in 13m 04s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/c334d7a/ : SUCCESS in 20m 12s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/ae9bc65/ : SUCCESS in 10m 41s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-btrfs-opensuse-150/bff171a/ : FAILURE in 13m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/f245b00/ : SUCCESS in 10m 58s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-centos-7/3fccd9d/ : SUCCESS in 13m 33s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-machinectl-opensuse-150/d4c375b/ : FAILURE in 21m 43s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/9c7745e/ : SUCCESS in 12m 13s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/b776b1f/ : FAILURE in 14m 32s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/20/657720/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/88801a4/ : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"241c40346e942e06ec25a9d05c34db07181da9ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 06:15:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c02b7bdef377c475d5ba780537eaa487710b9e80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 06:29:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d413e37658919a71f2391449d1aadc3497ba660","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 06:29:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/657720/1/gate/openstack-ansible-linters/5207449/ : SUCCESS in 5m 34s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/20/657720/1/gate/openstack-ansible-functional-debian-stable/aae3787/ : SUCCESS in 8m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/20/657720/1/gate/openstack-ansible-functional-centos-7/8991051/ : SUCCESS in 8m 50s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/20/657720/1/gate/openstack-ansible-functional-ubuntu-bionic/4939ef6/ : SUCCESS in 7m 58s\n- openstack-tox-docs http://logs.openstack.org/20/657720/1/gate/openstack-tox-docs/d1ba2dc/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d552f4c44659ba7947af55c9e34ba4c34c6d3424","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-08 10:43:31.000000000","message":"Patch Set 1:\n\nHm, why do we returned to hardcoding image build?","accounts_in_message":[],"_revision_number":1},{"id":"815e9f796e29f07e9dc6e496761d30739dc5cbfc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-08 10:46:37.000000000","message":"Patch Set 1:\n\nSo had to rebase https://review.opendev.org/#/c/655851/3 on top of this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c7623d4edf97e57107924181809aa47f4cf0525","revisions":{"6c7623d4edf97e57107924181809aa47f4cf0525":{"kind":"REWORK","_number":1,"created":"2019-05-08 02:00:41.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/20/657720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/20/657720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/657720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/657720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/657720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/20/657720/1"}}},"commit":{"parents":[{"commit":"43e61820f42d88f36ec0ab4865800cbb4b1076ae","subject":"Temporary fix of failing aria2 on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/43e61820f42d88f36ec0ab4865800cbb4b1076ae"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-05-08 02:00:14.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-05-08 02:00:14.000000000","tz":-360},"subject":"Update (yet again) SUSE lxc container name","message":"Update (yet again) SUSE lxc container name\n\nChange-Id: I2bd032f0d5dc17f61e380b44a761fe110ccb42f5\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6c7623d4edf97e57107924181809aa47f4cf0525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/6c7623d4edf97e57107924181809aa47f4cf0525"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
