)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~658578","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~Ia2ddc09f7915367d1200156db92188300a7cc764","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia2ddc09f7915367d1200156db92188300a7cc764","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:57:49.000000000","updated":"2019-05-12 02:04:10.000000000","submitted":"2019-05-12 02:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658578-1557626650622-71816a1a","meta_rev_id":"cd1c73ea8b9c4748c3b83406ad15fe6f9b4e6d31","_number":658578,"virtual_id_number":658578,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 02:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:53:52.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 02:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"602fe0d2b0cc40fc2156e714542e1641eda6c91c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:57:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2b9c5b3ce4503c9987730de94f786ce8e72713c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 00:32:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/658578/1/check/openstack-ansible-linters/d0199c4/ : SUCCESS in 5m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/78/658578/1/check/openstack-ansible-functional-debian-stable/5019ba0/ : SUCCESS in 9m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/658578/1/check/openstack-ansible-functional-centos-7/76714dd/ : SUCCESS in 9m 53s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-functional-ubuntu-bionic/a8efe11/ : SUCCESS in 7m 18s\n- openstack-tox-docs http://logs.openstack.org/78/658578/1/check/openstack-tox-docs/cc7e29a/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/78/658578/1/check/build-openstack-releasenotes/1883da6/html/ : SUCCESS in 3m 29s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-dir-centos-7/ced2a2d/ : SUCCESS in 14m 09s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-dir-opensuse-150/da97184/ : FAILURE in 9m 15s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/7ddf53c/ : SUCCESS in 12m 45s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/3d744d4/ : SUCCESS in 10m 42s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-btrfs-opensuse-150/d406f7f/ : FAILURE in 10m 15s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/7ca284f/ : SUCCESS in 11m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-machinectl-centos-7/88d89e1/ : SUCCESS in 16m 42s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-machinectl-opensuse-150/e63e715/ : FAILURE in 9m 55s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/45150d2/ : SUCCESS in 12m 05s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/143b1f4/ : FAILURE in 8m 47s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/78/658578/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/54e5c92/ : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8a20bbf76f1eaca34ec038d28787dd6bdb82561d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:53:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0d160357046ca974a9b9b7c2845c50f6012d5d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:54:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ecf98c6349d0b9364ec34c35ba1512f94e8feb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:04:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6f6b51f71c185ca39a9dc31c0cd461e377592b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:04:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/658578/1/gate/openstack-ansible-linters/ed61294/ : SUCCESS in 5m 36s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/78/658578/1/gate/openstack-ansible-functional-debian-stable/b8504f5/ : SUCCESS in 8m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/78/658578/1/gate/openstack-ansible-functional-centos-7/9d24e64/ : SUCCESS in 9m 03s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/78/658578/1/gate/openstack-ansible-functional-ubuntu-bionic/0ea5457/ : SUCCESS in 8m 29s\n- openstack-tox-docs http://logs.openstack.org/78/658578/1/gate/openstack-tox-docs/3ba1245/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/78/658578/1/gate/build-openstack-releasenotes/b7fdd1c/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98089915cc552c63a9affbeedd81864510a01a33","revisions":{"98089915cc552c63a9affbeedd81864510a01a33":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:57:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/658578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/78/658578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/658578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/658578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/658578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/78/658578/1"}}},"commit":{"parents":[{"commit":"5fb9fe50359e21dcc3a0d0dce7436b9f8e233831","subject":"Temporary fix of failing aria2 on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/5fb9fe50359e21dcc3a0d0dce7436b9f8e233831"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:57:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:57:48.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia2ddc09f7915367d1200156db92188300a7cc764\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/98089915cc552c63a9affbeedd81864510a01a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/98089915cc552c63a9affbeedd81864510a01a33"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
