)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~658672","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~master~I26d3c9338247ffc789bf4a5291e65e954151cc51","project":"openstack/openstack-ansible-lxc_hosts","branch":"master","topic":"fix_aria_centos","hashtags":[],"change_id":"I26d3c9338247ffc789bf4a5291e65e954151cc51","subject":"Add retry for backup get_url task","status":"MERGED","created":"2019-05-12 16:25:05.000000000","updated":"2019-05-12 18:17:00.000000000","submitted":"2019-05-12 17:47:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658672-1557683246109-121c7fd4","meta_rev_id":"c17051f463e66f1a18656fb3bfc39607c1917330","_number":658672,"virtual_id_number":658672,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 17:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 16:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-12 17:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-12 17:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 16:27:41.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 17:03:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-12 17:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28e1c7c461027dbaad3f26b9641625131de8ac83","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 16:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7f12dfa5495c81f83ccbdef5090ec3813c14508","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 16:27:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9186d59645d91aecbecc8a0bc38307db79d9f09a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 16:51:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/658672/1/check/openstack-ansible-linters/660d0b7/ : SUCCESS in 5m 07s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/658672/1/check/openstack-ansible-functional-debian-stable/42e95c7/ : SUCCESS in 8m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/658672/1/check/openstack-ansible-functional-centos-7/4d49dcc/ : SUCCESS in 8m 48s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-functional-ubuntu-bionic/234e570/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/72/658672/1/check/openstack-tox-docs/6d322fb/html/ : SUCCESS in 3m 04s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-dir-centos-7/cc32779/ : SUCCESS in 13m 27s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-dir-opensuse-150/5fe7bdb/ : FAILURE in 19m 19s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/a3d31d0/ : SUCCESS in 10m 53s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/7320198/ : SUCCESS in 11m 13s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-btrfs-opensuse-150/6116ac3/ : FAILURE in 20m 45s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/6d91a37/ : SUCCESS in 9m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-machinectl-centos-7/4ff6785/ : SUCCESS in 13m 08s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-machinectl-opensuse-150/631d9e0/ : FAILURE in 16m 12s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/29d5b81/ : SUCCESS in 11m 12s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/5468cdd/ : FAILURE in 18m 15s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/72/658672/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/c27cfb4/ : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d9354a66d153f2f71745aa7cd8809101a73b21d9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-12 17:03:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35320b8807270af4d53bb99c1063a70e8a246ef0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:03:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8b3f23c17956e77547546704ad6e970d944381b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:47:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/658672/1/gate/openstack-ansible-linters/2eae4dc/ : SUCCESS in 7m 39s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/658672/1/gate/openstack-ansible-functional-debian-stable/0368828/ : SUCCESS in 9m 24s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/658672/1/gate/openstack-ansible-functional-centos-7/7fc06e5/ : SUCCESS in 12m 34s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/658672/1/gate/openstack-ansible-functional-ubuntu-bionic/097c7ef/ : SUCCESS in 6m 34s\n- openstack-tox-docs http://logs.openstack.org/72/658672/1/gate/openstack-tox-docs/fcceb34/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"57806b8366cd712a48d93035631ca7fa604265f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 17:47:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"904a7eb45b3737dec5407f463dbdc105ab9e9587","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 18:17:00.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b790fffc16b1822d78b8a4d571edefd2d4675dbd","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b1078aca96a1cc22ccaab335c0159b3bf5fa9a5","revisions":{"1b1078aca96a1cc22ccaab335c0159b3bf5fa9a5":{"kind":"REWORK","_number":1,"created":"2019-05-12 16:25:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/658672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/72/658672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/658672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/658672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/658672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/72/658672/1"}}},"commit":{"parents":[{"commit":"af76d499056be4dca5b21d7b1340e17d413b257b","subject":"Merge \"Returning of the Suse LXC image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/af76d499056be4dca5b21d7b1340e17d413b257b"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-12 16:24:54.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-12 16:24:54.000000000","tz":180},"subject":"Add retry for backup get_url task","message":"Add retry for backup get_url task\n\nAs CentOS aria2 package is still only in epel-testing repo,\nget_url tends to fail with timout errors, which forces to do rechecks\nand creates extra load on gates.\nSo we\u0027re adding retries for this task and extending timeout.\n\nChange-Id: I26d3c9338247ffc789bf4a5291e65e954151cc51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b1078aca96a1cc22ccaab335c0159b3bf5fa9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b1078aca96a1cc22ccaab335c0159b3bf5fa9a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
