)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~658683","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~I26d3c9338247ffc789bf4a5291e65e954151cc51","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","topic":"fix_aria_centos-stable/stein","hashtags":[],"change_id":"I26d3c9338247ffc789bf4a5291e65e954151cc51","subject":"Add retry for backup get_url task","status":"MERGED","created":"2019-05-12 18:17:00.000000000","updated":"2019-05-16 00:23:04.000000000","submitted":"2019-05-16 00:23:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658683-1557966184954-4a183efd","meta_rev_id":"1fa3e597f482e43357056588a8a33e339f774b2d","_number":658683,"virtual_id_number":658683,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"date":"2019-05-12 18:17:10.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-16 00:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":2,"date":"2019-05-16 00:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 21:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":1,"date":"2019-05-16 00:08:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 21:25:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-16 00:08:10.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-16 00:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b304ced650e33e0eaff4e2bbc0096120e69f5293","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 18:17:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed6d803ec1c45a5aaf22ff421fe1fb593fed70dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 18:17:10.000000000","message":"Patch Set 1:\n\nAs this method was backported, this patch is worth backporting as well, imo","accounts_in_message":[],"_revision_number":1},{"id":"6d00ae1c99e346228efa6fde8aaf6b78a0cd324b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 19:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/658683/1/check/openstack-ansible-linters/7256a13/ : SUCCESS in 4m 06s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/658683/1/check/openstack-ansible-functional-debian-stable/7a2dfc8/ : SUCCESS in 7m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/658683/1/check/openstack-ansible-functional-centos-7/f4769df/ : SUCCESS in 8m 55s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-functional-ubuntu-bionic/19b5848/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/83/658683/1/check/openstack-tox-docs/2f9e8d7/html/ : SUCCESS in 2m 14s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-dir-centos-7/a3a191f/ : SUCCESS in 14m 15s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-dir-opensuse-150/397dc7b/ : FAILURE in 7m 31s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/f6bdbdd/ : SUCCESS in 11m 06s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/74ef21e/ : SUCCESS in 11m 20s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-btrfs-opensuse-150/cda3993/ : FAILURE in 7m 42s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/6a42f6f/ : SUCCESS in 11m 18s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-machinectl-centos-7/618224f/ : SUCCESS in 14m 04s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-machinectl-opensuse-150/d87b7e3/ : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/937a589/ : SUCCESS in 10m 46s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/a8cdcd9/ : FAILURE in 9m 22s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/83/658683/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/df662db/ : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2d2110000dea74e6dd1c2991cf778ca467cbb67c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-14 21:25:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2bed2f1dc8cce1aed0a6f8a432e9758e38e37bc4","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-16 00:08:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"618b9018c07c7c4b197159ad3c7f1581b65ad828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 00:08:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71b9be47b2c2cd6c2ce1ee9b0e142044603867a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 00:23:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/658683/1/gate/openstack-ansible-linters/ba13af3/ : SUCCESS in 5m 15s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/658683/1/gate/openstack-ansible-functional-debian-stable/349a572/ : SUCCESS in 8m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/658683/1/gate/openstack-ansible-functional-centos-7/a4c165c/ : SUCCESS in 9m 07s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/658683/1/gate/openstack-ansible-functional-ubuntu-bionic/d2b34f3/ : SUCCESS in 8m 07s\n- openstack-tox-docs http://logs.openstack.org/83/658683/1/gate/openstack-tox-docs/5bdbe83/html/ : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"75efed0a458c7466de2508d1b0d809469d1ddcb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 00:23:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b790fffc16b1822d78b8a4d571edefd2d4675dbd","revisions":{"b790fffc16b1822d78b8a4d571edefd2d4675dbd":{"kind":"REWORK","_number":1,"created":"2019-05-12 18:17:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/658683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/83/658683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/658683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/658683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/658683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/83/658683/1"}}},"commit":{"parents":[{"commit":"98089915cc552c63a9affbeedd81864510a01a33","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/98089915cc552c63a9affbeedd81864510a01a33"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-12 16:24:54.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-12 18:17:00.000000000","tz":0},"subject":"Add retry for backup get_url task","message":"Add retry for backup get_url task\n\nAs CentOS aria2 package is still only in epel-testing repo,\nget_url tends to fail with timout errors, which forces to do rechecks\nand creates extra load on gates.\nSo we\u0027re adding retries for this task and extending timeout.\n\nChange-Id: I26d3c9338247ffc789bf4a5291e65e954151cc51\n(cherry picked from commit 1b1078aca96a1cc22ccaab335c0159b3bf5fa9a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b790fffc16b1822d78b8a4d571edefd2d4675dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/b790fffc16b1822d78b8a4d571edefd2d4675dbd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
