)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~659360","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Frocky~I4ec91f541e6d2d4cca4d899f4c7d519b169e2f50","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/rocky","topic":"image_fixes_rocky","hashtags":[],"change_id":"I4ec91f541e6d2d4cca4d899f4c7d519b169e2f50","subject":"Use block/rescue for image pre-stage","status":"MERGED","created":"2019-05-15 18:39:52.000000000","updated":"2019-05-16 10:01:04.000000000","submitted":"2019-05-16 10:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659360-1558000864732-cf8f5779","meta_rev_id":"bf64f199cb87061e7eaf2bd15bcc7967fb52c52a","_number":659360,"virtual_id_number":659360,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-15 21:50:57.000000000","_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-16 10:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2019-05-16 00:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-16 09:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-16 09:39:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 18:39:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-16 00:07:28.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-16 09:39:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-16 10:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"850d24ad5a81c65c5accc27610534fd73ff99282","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-15 18:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2149c97095b4a5ec71f36b53cd4188f0013e1a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:25:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/60/659360/1/check/openstack-ansible-linters/989198c/ : SUCCESS in 5m 57s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/60/659360/1/check/openstack-ansible-functional-centos-7/b7ce383/ : SUCCESS in 11m 00s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/60/659360/1/check/openstack-ansible-functional-opensuse-423/9cd1e14/ : SUCCESS in 9m 16s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/60/659360/1/check/openstack-ansible-functional-opensuse-150/a270860/ : FAILURE in 9m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-functional-ubuntu-bionic/ae410fd/ : SUCCESS in 10m 02s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-functional-ubuntu-xenial/3760b82/ : SUCCESS in 11m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/659360/1/check/build-openstack-sphinx-docs/24e03c3/html/ : SUCCESS in 3m 27s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-dir-centos-7/0db3180/ : SUCCESS in 14m 10s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-dir-opensuse-423/68f3af0/ : FAILURE in 11m 39s (non-voting)\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-dir-opensuse-150/29400aa/ : FAILURE in 11m 28s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/ad54fc9/ : SUCCESS in 12m 43s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/bfa04c2/ : SUCCESS in 10m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/c8b391f/ : SUCCESS in 12m 05s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/7940e34/ : SUCCESS in 11m 19s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-btrfs-centos-7/e95bfae/ : SUCCESS in 15m 08s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-btrfs-opensuse-423/fe03683/ : FAILURE in 10m 06s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-btrfs-opensuse-150/b962756/ : FAILURE in 9m 24s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/1b7a9d8/ : SUCCESS in 10m 46s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/26dd30a/ : FAILURE in 14m 49s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-machinectl-centos-7/27cbb45/ : SUCCESS in 12m 32s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-machinectl-opensuse-423/464073e/ : FAILURE in 15m 35s (non-voting)\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-machinectl-opensuse-150/36505cf/ : FAILURE in 9m 20s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/d7ced9b/ : SUCCESS in 22m 22s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/b763f7d/ : SUCCESS in 10m 52s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-overlayfs-centos-7/df2f732/ : SUCCESS in 19m 23s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-overlayfs-opensuse-423/5dbfcbe/ : FAILURE in 9m 25s (non-voting)\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/498e63e/ : FAILURE in 15m 27s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/196648c/ : SUCCESS in 11m 49s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/60/659360/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/527604d/ : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dedec0b6f8dbf233eaedc95c0bd600dbad33bf5d","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-15 19:36:03.000000000","message":"Patch Set 1:\n\nFixes CentOS image retrieval issues in gating","accounts_in_message":[],"_revision_number":1},{"id":"1fc17364ed92bda2ad19d2a66c6fb7cca1a68ec3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-15 21:18:33.000000000","message":"Patch Set 1: Code-Review-1\n\nplease can you add a \"(cherry picked from commit \u003ccommit-id\u003e)\" to the commit message","accounts_in_message":[],"_revision_number":1},{"id":"eb89ea43c46bf34b7eb6205b3cbb309a77461bb3","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-15 21:50:09.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"552626b840428e4a35ef41e851209239515da613","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-15 21:50:57.000000000","message":"Patch Set 2:\n\nUpdated, cherry-picked it but had to resolve conflicts in the rocky branch.","accounts_in_message":[],"_revision_number":2},{"id":"6af06858ab58f60dfea7c2f79b692953a32bb2c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:19:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/60/659360/2/check/openstack-ansible-linters/01c01ad/ : SUCCESS in 6m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/60/659360/2/check/openstack-ansible-functional-centos-7/4b7f863/ : SUCCESS in 10m 37s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/60/659360/2/check/openstack-ansible-functional-opensuse-423/81a24e5/ : SUCCESS in 10m 13s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/60/659360/2/check/openstack-ansible-functional-opensuse-150/45e8f69/ : FAILURE in 8m 22s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-functional-ubuntu-bionic/df1dbf7/ : SUCCESS in 7m 48s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-functional-ubuntu-xenial/87635a1/ : SUCCESS in 7m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/659360/2/check/build-openstack-sphinx-docs/bac1435/html/ : SUCCESS in 3m 01s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-dir-centos-7/f42e763/ : SUCCESS in 13m 59s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-dir-opensuse-423/d887d4e/ : FAILURE in 10m 00s (non-voting)\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-dir-opensuse-150/cbf9244/ : FAILURE in 9m 38s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-dir-ubuntu-xenial/bf866d9/ : SUCCESS in 15m 44s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-dir-ubuntu-bionic/e2ddf96/ : SUCCESS in 13m 58s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/aac35fe/ : SUCCESS in 12m 37s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/7be6285/ : SUCCESS in 11m 32s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-btrfs-centos-7/0d2295a/ : SUCCESS in 14m 45s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-btrfs-opensuse-423/7c9c8a3/ : FAILURE in 10m 25s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-btrfs-opensuse-150/f892a0d/ : FAILURE in 9m 41s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/065b1f2/ : SUCCESS in 13m 10s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/81530e6/ : FAILURE in 9m 27s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-machinectl-centos-7/165839d/ : SUCCESS in 15m 20s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-machinectl-opensuse-423/29f3cef/ : FAILURE in 9m 58s (non-voting)\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-machinectl-opensuse-150/a991a40/ : FAILURE in 11m 41s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/31ec0d0/ : SUCCESS in 12m 49s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/103bf2a/ : SUCCESS in 11m 59s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-overlayfs-centos-7/e8eae25/ : SUCCESS in 15m 42s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-overlayfs-opensuse-423/9e8f432/ : FAILURE in 10m 05s (non-voting)\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-overlayfs-opensuse-150/19e1159/ : FAILURE in 9m 20s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-zfs-ubuntu-xenial/ec9467d/ : SUCCESS in 16m 47s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/60/659360/2/check/openstack-ansible-lxc-zfs-ubuntu-bionic/df1da13/ : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f8a1f56dae8791e31e33ddd699ffe9fca4837cb8","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-16 00:07:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8fadf3c22cdce3a4bb80d2a58ed64a1ca1e6709","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-16 09:39:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30c560f77192a2c2fe910d717c0a76ff12d3bfc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:39:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"73caceac7f4ef9284009a0f75a3b2385535e5997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:01:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/60/659360/2/gate/openstack-ansible-linters/a411fc9/ : SUCCESS in 4m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/60/659360/2/gate/openstack-ansible-functional-centos-7/09511cc/ : SUCCESS in 9m 32s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/60/659360/2/gate/openstack-ansible-functional-opensuse-423/5e216b3/ : SUCCESS in 7m 41s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/60/659360/2/gate/openstack-ansible-functional-ubuntu-bionic/394a5cb/ : SUCCESS in 6m 45s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/60/659360/2/gate/openstack-ansible-functional-ubuntu-xenial/c464ff1/ : SUCCESS in 14m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/659360/2/gate/build-openstack-sphinx-docs/328d984/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6a8b107f67e644aa2b95feb16be97749aa93f091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 10:01:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"623d5da82840f5c60605029cd3c2ccffa5d0edf3","revisions":{"579909373c50e6f99d3c77fd8e0a8891210130a8":{"kind":"REWORK","_number":1,"created":"2019-05-15 18:39:52.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/60/659360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/60/659360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/1"}}},"commit":{"parents":[{"commit":"2d79568f6088d7a25c6558aed6b8d837b497722b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2d79568f6088d7a25c6558aed6b8d837b497722b"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-13 14:45:00.000000000","tz":180},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-05-15 18:39:25.000000000","tz":-300},"subject":"Use block/rescue for image pre-stage","message":"Use block/rescue for image pre-stage\n\nMove to usage of block/rescue for image pre-stage step.\n\nChange-Id: I4ec91f541e6d2d4cca4d899f4c7d519b169e2f50\nCo-Authored-By: Jesse Pretorius \u003cjesse@odyssey4.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/579909373c50e6f99d3c77fd8e0a8891210130a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/579909373c50e6f99d3c77fd8e0a8891210130a8"}]},"branch":"refs/heads/stable/rocky"},"623d5da82840f5c60605029cd3c2ccffa5d0edf3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-15 21:50:09.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/60/659360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/60/659360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/60/659360/2"}}},"commit":{"parents":[{"commit":"2d79568f6088d7a25c6558aed6b8d837b497722b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/2d79568f6088d7a25c6558aed6b8d837b497722b"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-13 14:45:00.000000000","tz":180},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-05-15 21:50:01.000000000","tz":-300},"subject":"Use block/rescue for image pre-stage","message":"Use block/rescue for image pre-stage\n\nMove to usage of block/rescue for image pre-stage step.\n\nCherry-picked from b3939158743661019b697235fefeab16ca19300e\n\nChange-Id: I4ec91f541e6d2d4cca4d899f4c7d519b169e2f50\nCo-Authored-By: Jesse Pretorius \u003cjesse@odyssey4.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/623d5da82840f5c60605029cd3c2ccffa5d0edf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/623d5da82840f5c60605029cd3c2ccffa5d0edf3"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
