)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~659980","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Fstein~Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/stein","topic":"fix_aria_centos","hashtags":[],"change_id":"Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","subject":"Revert \"Temporary fix of failing aria2 on centos\"","status":"MERGED","created":"2019-05-19 19:38:01.000000000","updated":"2019-05-20 10:38:22.000000000","submitted":"2019-05-20 10:38:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659980-1558348702830-172dd170","meta_rev_id":"79964acca630bb0cd7a5a6a31e99f869fb91712d","_number":659980,"virtual_id_number":659980,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-20 10:38:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-05-20 10:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 08:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-05-20 10:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 08:28:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-20 10:13:03.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-20 10:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7c895d58e0939b1ccfbaf32c0fe45392d91751d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dd5b545e362044b6995153300b81991dd894333d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:17.000000000","message":"Topic changed from fix_aria_centos-stable/stein to fix_aria_centos","accounts_in_message":[],"_revision_number":1},{"id":"88848535b3576f6356d0864f1a586557bf0d34b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 20:27:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/659980/1/check/openstack-ansible-linters/8d09dd5/ : SUCCESS in 4m 14s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/80/659980/1/check/openstack-ansible-functional-debian-stable/344567f/ : SUCCESS in 7m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/659980/1/check/openstack-ansible-functional-centos-7/3d9a3d4/ : SUCCESS in 13m 15s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-functional-ubuntu-bionic/34e7db2/ : SUCCESS in 9m 06s\n- openstack-tox-docs http://logs.openstack.org/80/659980/1/check/openstack-tox-docs/27311be/html/ : SUCCESS in 2m 02s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-dir-centos-7/0890dc9/ : SUCCESS in 13m 19s\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-dir-opensuse-150/e735f9b/ : FAILURE in 13m 12s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/a15bb5b/ : SUCCESS in 16m 38s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/c20af91/ : SUCCESS in 10m 09s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-btrfs-opensuse-150/dd69a87/ : FAILURE in 18m 07s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/1bd571d/ : SUCCESS in 9m 39s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-machinectl-centos-7/75c35da/ : SUCCESS in 17m 02s\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-machinectl-opensuse-150/332b9fa/ : FAILURE in 11m 08s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/1688d65/ : SUCCESS in 12m 08s\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/8442c65/ : FAILURE in 14m 10s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/80/659980/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/0b03716/ : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6aaf333f1867e2bcc1db51107a1b248840f11ca3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 08:28:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49d12ab79d3c32b99449cfd6202d11bfd19f8c49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-20 10:13:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6fe35d3174d5c67a5847df514ea5deeccfeb7564","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:13:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd165ae48e8fc1cfa7417b2e551b76c3e10f855a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:38:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79964acca630bb0cd7a5a6a31e99f869fb91712d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:38:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/659980/1/gate/openstack-ansible-linters/49f8247/ : SUCCESS in 5m 49s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/80/659980/1/gate/openstack-ansible-functional-debian-stable/86b09da/ : SUCCESS in 10m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/659980/1/gate/openstack-ansible-functional-centos-7/ea32974/ : SUCCESS in 8m 43s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/659980/1/gate/openstack-ansible-functional-ubuntu-bionic/5199a87/ : SUCCESS in 7m 54s\n- openstack-tox-docs http://logs.openstack.org/80/659980/1/gate/openstack-tox-docs/c626f2e/html/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b402c9b1b36c084fceb1a63a8a12acd73ade4d3","revisions":{"1b402c9b1b36c084fceb1a63a8a12acd73ade4d3":{"kind":"REWORK","_number":1,"created":"2019-05-19 19:38:01.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/80/659980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/80/659980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/659980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/659980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/659980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/80/659980/1"}}},"commit":{"parents":[{"commit":"ed4bed91b76e04b5bc81c8b7b64390bd8b0fc743","subject":"Use block/rescue for image pre-stage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/ed4bed91b76e04b5bc81c8b7b64390bd8b0fc743"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 17:50:23.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 19:38:01.000000000","tz":0},"subject":"Revert \"Temporary fix of failing aria2 on centos\"","message":"Revert \"Temporary fix of failing aria2 on centos\"\n\nThe fix is no longer needed since a fixed aria2 package has been promoted in epel.\n\nSee https://bugzilla.redhat.com/show_bug.cgi?id\u003d1704413\n\nThis is a revert of the following changes:\nI7770c5417feacfc43a3ad8801125603a55ba9ea8\nI26d3c9338247ffc789bf4a5291e65e954151cc51\nI4ec91f541e6d2d4cca4d899f4c7d519b169e2f50\n\nChange-Id: Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b402c9b1b36c084fceb1a63a8a12acd73ade4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/1b402c9b1b36c084fceb1a63a8a12acd73ade4d3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
