)]}'
{"id":"openstack%2Fopenstack-ansible-lxc_hosts~659981","triplet_id":"openstack%2Fopenstack-ansible-lxc_hosts~stable%2Frocky~Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","project":"openstack/openstack-ansible-lxc_hosts","branch":"stable/rocky","topic":"fix_aria_centos","hashtags":[],"change_id":"Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb","subject":"Revert \"Temporary fix of failing aria2 on centos\"","status":"MERGED","created":"2019-05-19 19:38:43.000000000","updated":"2019-05-20 10:37:19.000000000","submitted":"2019-05-20 10:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659981-1558348639655-f3a90b91","meta_rev_id":"0fa50fc350596163f4d1dcd5b3628832e6cdb62f","_number":659981,"virtual_id_number":659981,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-20 10:37:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-05-20 10:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 08:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-05-20 10:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 08:29:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-20 10:12:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-20 10:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"832e0401907857665a849f42517e9db7392563b3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"187125482876d2ef6f3c01973ea379dc1ede7d0a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-19 19:38:49.000000000","message":"Topic changed from fix_aria_centos-stable/rocky to fix_aria_centos","accounts_in_message":[],"_revision_number":1},{"id":"623df688c74a90349f01c9850f4f5740f9bf54a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 20:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/81/659981/1/check/openstack-ansible-linters/7a0f40b/ : SUCCESS in 7m 53s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/81/659981/1/check/openstack-ansible-functional-centos-7/cae1f38/ : SUCCESS in 11m 55s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/81/659981/1/check/openstack-ansible-functional-opensuse-423/8177cf0/ : SUCCESS in 7m 18s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/81/659981/1/check/openstack-ansible-functional-opensuse-150/ab95509/ : FAILURE in 8m 36s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-functional-ubuntu-bionic/751aca9/ : SUCCESS in 6m 32s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-functional-ubuntu-xenial/fa625c2/ : SUCCESS in 11m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/81/659981/1/check/build-openstack-sphinx-docs/ed5e2f1/html/ : SUCCESS in 4m 01s\n- openstack-ansible-lxc-dir-centos-7 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-dir-centos-7/bd970cc/ : SUCCESS in 17m 43s\n- openstack-ansible-lxc-dir-opensuse-423 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-dir-opensuse-423/207a5ed/ : FAILURE in 9m 01s (non-voting)\n- openstack-ansible-lxc-dir-opensuse-150 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-dir-opensuse-150/5d15efd/ : FAILURE in 8m 54s (non-voting)\n- openstack-ansible-lxc-dir-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-dir-ubuntu-xenial/53f0b8d/ : SUCCESS in 12m 30s\n- openstack-ansible-lxc-dir-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-dir-ubuntu-bionic/ccad52d/ : SUCCESS in 10m 36s\n- openstack-ansible-lxc-btrfs-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-btrfs-ubuntu-xenial/31415ee/ : SUCCESS in 11m 49s\n- openstack-ansible-lxc-btrfs-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-btrfs-ubuntu-bionic/0d86cc9/ : SUCCESS in 12m 52s (non-voting)\n- openstack-ansible-lxc-btrfs-centos-7 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-btrfs-centos-7/e4a00cb/ : SUCCESS in 13m 07s\n- openstack-ansible-lxc-btrfs-opensuse-423 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-btrfs-opensuse-423/7fa9f8d/ : FAILURE in 9m 40s (non-voting)\n- openstack-ansible-lxc-btrfs-opensuse-150 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-btrfs-opensuse-150/d5bccd4/ : FAILURE in 8m 00s (non-voting)\n- openstack-ansible-lxc-machinectl-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-machinectl-ubuntu-xenial/e4e272a/ : SUCCESS in 10m 04s\n- openstack-ansible-lxc-machinectl-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-machinectl-ubuntu-bionic/635d196/ : FAILURE in 7m 36s (non-voting)\n- openstack-ansible-lxc-machinectl-centos-7 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-machinectl-centos-7/29562aa/ : SUCCESS in 15m 19s\n- openstack-ansible-lxc-machinectl-opensuse-423 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-machinectl-opensuse-423/48bd4f0/ : FAILURE in 8m 53s (non-voting)\n- openstack-ansible-lxc-machinectl-opensuse-150 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-machinectl-opensuse-150/4cc86b6/ : FAILURE in 8m 15s (non-voting)\n- openstack-ansible-lxc-overlayfs-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-overlayfs-ubuntu-xenial/902f35d/ : SUCCESS in 15m 46s\n- openstack-ansible-lxc-overlayfs-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-overlayfs-ubuntu-bionic/bba7788/ : SUCCESS in 10m 43s\n- openstack-ansible-lxc-overlayfs-centos-7 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-overlayfs-centos-7/7215d57/ : SUCCESS in 17m 35s\n- openstack-ansible-lxc-overlayfs-opensuse-423 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-overlayfs-opensuse-423/5a3e6ce/ : FAILURE in 9m 16s (non-voting)\n- openstack-ansible-lxc-overlayfs-opensuse-150 http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-overlayfs-opensuse-150/de77215/ : FAILURE in 9m 01s (non-voting)\n- openstack-ansible-lxc-zfs-ubuntu-xenial http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-zfs-ubuntu-xenial/b1317df/ : SUCCESS in 14m 12s\n- openstack-ansible-lxc-zfs-ubuntu-bionic http://logs.openstack.org/81/659981/1/check/openstack-ansible-lxc-zfs-ubuntu-bionic/9770c4d/ : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"620858e371aa7413fdd9711d0f79de16c4fddd4b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 08:29:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d69056e13f3e4873e819100b2fecfa9136ed224","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-20 10:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd3f9f303cf646c006e66e5e5fdd1a6233cf86f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:12:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5275a12caf0a6191ff0622d28b7c692008cfb68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:37:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fa50fc350596163f4d1dcd5b3628832e6cdb62f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:37:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/81/659981/1/gate/openstack-ansible-linters/b7d8cf8/ : SUCCESS in 4m 27s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/81/659981/1/gate/openstack-ansible-functional-centos-7/8c0a48e/ : SUCCESS in 9m 26s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/81/659981/1/gate/openstack-ansible-functional-opensuse-423/730f7ce/ : SUCCESS in 8m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/81/659981/1/gate/openstack-ansible-functional-ubuntu-bionic/d3bc71f/ : SUCCESS in 8m 20s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/81/659981/1/gate/openstack-ansible-functional-ubuntu-xenial/e492a49/ : SUCCESS in 8m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/81/659981/1/gate/build-openstack-sphinx-docs/1ab1fcd/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a4bfb5d8823260cb809f788cd8d1756e0e08cb9","revisions":{"3a4bfb5d8823260cb809f788cd8d1756e0e08cb9":{"kind":"REWORK","_number":1,"created":"2019-05-19 19:38:43.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/81/659981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-lxc_hosts","ref":"refs/changes/81/659981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/81/659981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/81/659981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/81/659981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-lxc_hosts refs/changes/81/659981/1"}}},"commit":{"parents":[{"commit":"623d5da82840f5c60605029cd3c2ccffa5d0edf3","subject":"Use block/rescue for image pre-stage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/623d5da82840f5c60605029cd3c2ccffa5d0edf3"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 17:50:23.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-19 19:38:43.000000000","tz":0},"subject":"Revert \"Temporary fix of failing aria2 on centos\"","message":"Revert \"Temporary fix of failing aria2 on centos\"\n\nThe fix is no longer needed since a fixed aria2 package has been promoted in epel.\n\nSee https://bugzilla.redhat.com/show_bug.cgi?id\u003d1704413\n\nThis is a revert of the following changes:\nI7770c5417feacfc43a3ad8801125603a55ba9ea8\nI26d3c9338247ffc789bf4a5291e65e954151cc51\nI4ec91f541e6d2d4cca4d899f4c7d519b169e2f50\n\nChange-Id: Ia181556e77dcc85bea726e47fc8ca1f097cc3bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3a4bfb5d8823260cb809f788cd8d1756e0e08cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-lxc_hosts/commit/3a4bfb5d8823260cb809f788cd8d1756e0e08cb9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
